)]}'
{"id":"openinfra%2Frefstack-client~815052","triplet_id":"openinfra%2Frefstack-client~master~Ic00064ed0769aaad3730f1db6889ca180860370b","project":"openinfra/refstack-client","branch":"master","attention_set":{},"removed_from_attention_set":{"22873":{"account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"last_update":"2021-10-21 19:45:22.000000000","reason":"Martin Kopec replied on the change"}},"hashtags":[],"change_id":"Ic00064ed0769aaad3730f1db6889ca180860370b","subject":"Temporarily disable most of the jobs","status":"MERGED","created":"2021-10-21 19:19:47.000000000","updated":"2021-10-21 21:40:40.000000000","submitted":"2021-10-21 19:52:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"815052","meta_rev_id":"e5078b9b2e00cad1ace27c6d70ba4811cc2a79ca","_number":815052,"virtual_id_number":815052,"owner":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-21 19:52:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":2,"date":"2021-10-21 19:45:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":1,"date":"2021-10-21 19:45:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-21 19:31:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6fd6271b9f597eaded7ce31a6540da57d578fea9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-10-21 19:19:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"35e19fe6216b01114aca9d601df1ca442d3fbbf2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-10-21 19:31:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0afbf1e5464a8fe56753759a033769b4d0cf56bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 19:31:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e987883e532b4d5288ce34396d95f6a1 : FAILURE in 5m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/034b07b900ea468c8c1057e21b86fa6e : FAILURE in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63a0bcc9aa2242b8af93b20511c6094c : SUCCESS in 3m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd10f4d6c72e41318fbba87a2566dcdd : FAILURE in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b831c5e79a64d6f8485cd75bbba5cf6 : FAILURE in 5m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/337dd01e1e124e7d82e0fb839a34b305 : FAILURE in 6m 42s (non-voting)\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/96805e6e766041989f84bd3665116151 : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":1},{"id":"e60971243f0f8266fccbc908b610bf35119412f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 19:43:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/5d142d2e35b24ff5847d2e25cd9f1eb5 : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":2},{"id":"646336fe1e914fb5fb6299a84930b2bdc8c72768","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-10-21 19:45:22.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"127cfc3be1f3d74ebf7d523a3abd0995561e79cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 19:47:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"422989ce843d64fff92aba7995969c2a8c9e950d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 19:52:19.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/aace40aee310472c88aa0facb10c758d : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":2},{"id":"7823be95767c704067350b770f4e2f1fa1343256","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 19:52:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"848035fe738ab8d001287873d1c48533f108f831","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 20:01:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/dffed96821fe460e87f36c28c5bc73b0 : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":2},{"id":"e5078b9b2e00cad1ace27c6d70ba4811cc2a79ca","tag":"autogenerated:gerrit:revert","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-10-21 21:40:40.000000000","message":"Created a revert of this change as Ic01e711bf5677950e782922a265a7702341565d1","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1c2f0761005729e687a76eca42c3913f6c6c78d4","revisions":{"211dd9ae660cb3de3cc8ddd902a0fe78086ef968":{"kind":"REWORK","_number":1,"created":"2021-10-21 19:19:47.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/52/815052/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack-client","ref":"refs/changes/52/815052/1","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/52/815052/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/52/815052/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/52/815052/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack-client refs/changes/52/815052/1"}}},"commit":{"parents":[{"commit":"9b8892c2dd2e73ff8bc6c16e81f7857df2ebcf56","subject":"Change documentation configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/9b8892c2dd2e73ff8bc6c16e81f7857df2ebcf56"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2021-10-21 19:15:11.000000000","tz":0},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2021-10-21 19:15:11.000000000","tz":0},"subject":"Temporarily remove refstack jobs","message":"Temporarily remove refstack jobs\n\nrefstack jobs are broken and require a fix (see follow up patches)\non refstack-client side which can\u0027t be merged without a fix\non ansible-role-refstack-client. This will be reverted as soon as\nthe issue is resolved.\n\nChange-Id: Ic00064ed0769aaad3730f1db6889ca180860370b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/211dd9ae660cb3de3cc8ddd902a0fe78086ef968"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/211dd9ae660cb3de3cc8ddd902a0fe78086ef968"}]},"branch":"refs/heads/master"},"1c2f0761005729e687a76eca42c3913f6c6c78d4":{"kind":"REWORK","_number":2,"created":"2021-10-21 19:31:34.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/52/815052/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack-client","ref":"refs/changes/52/815052/2","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/52/815052/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/52/815052/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/52/815052/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack-client refs/changes/52/815052/2"}}},"commit":{"parents":[{"commit":"9b8892c2dd2e73ff8bc6c16e81f7857df2ebcf56","subject":"Change documentation configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/9b8892c2dd2e73ff8bc6c16e81f7857df2ebcf56"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2021-10-21 19:15:11.000000000","tz":0},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2021-10-21 19:29:12.000000000","tz":0},"subject":"Temporarily disable most of the jobs","message":"Temporarily disable most of the jobs\n\nrefstack jobs are broken and require a fix (see follow up patches)\non refstack-client side which can\u0027t be merged without a fix\non ansible-role-refstack-client.\n\npython jobs are broken too and require a fix in refstack-client\n(already proposed) which can\u0027t be merged without other mentioned\nfixes.\n\nThis will be reverted as soon as the issue is resolved.\n\nChange-Id: Ic00064ed0769aaad3730f1db6889ca180860370b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/1c2f0761005729e687a76eca42c3913f6c6c78d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/1c2f0761005729e687a76eca42c3913f6c6c78d4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
