)]}'
{"id":"openinfra%2Frefstack-client~821325","triplet_id":"openinfra%2Frefstack-client~master~Ic0e590d54d76e2d5839ced3f513e8b2ded705e6f","project":"openinfra/refstack-client","branch":"master","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2022-01-18 15:01:42.000000000","reason":"Change was submitted"},"7160":{"account":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"last_update":"2021-12-15 19:47:38.000000000","reason":"arkady kanevsky replied on the change"},"22873":{"account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"last_update":"2022-01-18 12:06:59.000000000","reason":"Martin Kopec replied on the change"},"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2021-12-10 11:59:10.000000000","reason":"Vida Haririan replied on the change"}},"hashtags":[],"change_id":"Ic0e590d54d76e2d5839ced3f513e8b2ded705e6f","subject":"Increase verbose level for not found tempest tests","status":"MERGED","created":"2021-12-09 21:49:48.000000000","updated":"2022-01-18 15:04:22.000000000","submitted":"2022-01-18 15:01:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"821325","meta_rev_id":"a82594323aa6ecc794498f69446d0aa4ad378bbe","_number":821325,"virtual_id_number":821325,"owner":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-18 15:01:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"all":[{"value":2,"date":"2021-12-15 19:47:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"value":2,"date":"2021-12-10 11:59:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"all":[{"value":1,"date":"2021-12-15 19:47:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}],"reviewers":{"REVIEWER":[{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-10 00:48:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-10 07:18:11.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2021-12-10 07:18:11.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2021-12-10 07:18:11.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"state":"REVIEWER"}],"messages":[{"id":"803248d63fdc897e7639d2ae0e2097bf814bae8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-12-09 21:49:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"796f4632caa499cf1c2f8ddeba944b950902fce0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-10 00:48:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5e2db6c2fcd84370805644d9b31b33d4 : SUCCESS in 3m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/51f62cf1afa645969296649363bae175 : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd6e4e1b392845b4aefa186443a8e748 : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/41d6bf3228a14f5bbbc9254db803e80d : SUCCESS in 3m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/159596cfd22947e193315d9e7d0570b8 : SUCCESS in 3m 10s\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/c623538a2ef348a098fd1c0966837861 : SUCCESS in 1h 33m 36s\n- refstack-client-devstack-wallaby https://zuul.opendev.org/t/openstack/build/08f221c5eab64f56be05a7e041341a84 : SUCCESS in 1h 35m 14s\n- refstack-client-devstack-victoria https://zuul.opendev.org/t/openstack/build/f952acc4057646f2b3794bf826bee147 : SUCCESS in 1h 34m 48s\n- refstack-client-devstack-ussuri https://zuul.opendev.org/t/openstack/build/d10eaa068bda4838a4d88eabda5a02f4 : SUCCESS in 1h 35m 05s\n- refstack-client-devstack-train https://zuul.opendev.org/t/openstack/build/6ac4d24e52b34e02be824dfa3fb490d6 : SUCCESS in 1h 41m 16s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/143deac3635e4d438dba3b8514de54a6 : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":1},{"id":"bcae714e092686a4d0f1194518456866129b0c26","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2021-12-10 11:59:10.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8d0adcf97b295a6c3efb577784dc11e19bdf5a3f","author":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"date":"2021-12-15 19:47:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a0d671b70736f96de03c5876cf360656ee71cfac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 19:48:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"65188c74b012008e52be089d732d505a0e31c040","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 21:45:49.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3bfd89037c5b44c9a0e1413a508d7f2c : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96da279e90ac4831a35bd90ba127dc45 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2802f3315b214c6aafa9d1fa991362fb : SUCCESS in 3m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/484e9b4a30ec4720ab6bb0e4b5f3a9e3 : SUCCESS in 3m 58s\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/9999b010083f46ea99b90b35230f712a : FAILURE in 1h 26m 43s\n- refstack-client-devstack-wallaby https://zuul.opendev.org/t/openstack/build/c6ee570f55d844ebb66671b3dea47f4b : SUCCESS in 1h 28m 49s\n- refstack-client-devstack-victoria https://zuul.opendev.org/t/openstack/build/5d4de38ab8e6411f8b6a35221a7d87cd : SUCCESS in 1h 28m 08s\n- refstack-client-devstack-ussuri https://zuul.opendev.org/t/openstack/build/f871b52f03ce4ebd84ad83c14e2026c0 : SUCCESS in 1h 20m 08s\n- refstack-client-devstack-train https://zuul.opendev.org/t/openstack/build/74dd9a0216344b89bd18373b954f0ff4 : SUCCESS in 1h 24m 15s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/829930017d7c41168e1b449435073c7b : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":1},{"id":"b57392a7ca03653689c64905488d607f5be0051e","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-12-15 21:49:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0bb952a7f77f48b6115d91000db79c159d360f79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 23:29:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7841154d6e514396ac781ab06d63c009 : SUCCESS in 3m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f41c7cdcab8f4a0bbb52f87efc950329 : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf1778443adb4c9dac97dfec023a8fd1 : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/994717b9e0cb413797ed4d8a4bd6fe94 : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5dff040890294828a73539d3e5b6015c : SUCCESS in 3m 33s\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/d855c08244f44d338037e3550daa202a : FAILURE in 1h 05m 19s\n- refstack-client-devstack-wallaby https://zuul.opendev.org/t/openstack/build/b4cf748447884931bc2c7429f5c0b982 : SUCCESS in 1h 35m 04s\n- refstack-client-devstack-victoria https://zuul.opendev.org/t/openstack/build/b873ad4632cd453aafa6a85badde0398 : SUCCESS in 1h 30m 25s\n- refstack-client-devstack-ussuri https://zuul.opendev.org/t/openstack/build/e902790ebecb4a9db4bbd0233ab188e9 : SUCCESS in 1h 10m 15s\n- refstack-client-devstack-train https://zuul.opendev.org/t/openstack/build/e1b256d2738240a29dde1f4c28dee673 : SUCCESS in 1h 28m 39s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/b92d9c8fe326486a88fbf35d3203403b : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":1},{"id":"5d37b9ee4f9a53d12f626ae1b7b15f36aee4b4fb","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-01-18 12:06:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b423483089fdcc7514f05a642ee26c56ebd13f9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-18 13:41:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9d597ae3a5b841f4a18766fcff650855 : SUCCESS in 5m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7925746b9e364542942159831e50e3ee : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3ba8cf6d75e4f52aa55e7a0b82b4396 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b908f7c9f5124087bee227440b6547ae : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/762848af83e7448b81a7c5a95d0aa19b : SUCCESS in 3m 31s\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/a4434b2490c649b7ab5e44fe70d785d0 : SUCCESS in 1h 26m 53s\n- refstack-client-devstack-wallaby https://zuul.opendev.org/t/openstack/build/9b827e07dbc24f56a8bea61d37a43c11 : SUCCESS in 1h 19m 43s\n- refstack-client-devstack-victoria https://zuul.opendev.org/t/openstack/build/92903642d8764f209ff6192fb8605223 : SUCCESS in 1h 21m 29s\n- refstack-client-devstack-ussuri https://zuul.opendev.org/t/openstack/build/c95c186283dd47659f65c9e06327d620 : SUCCESS in 1h 19m 16s\n- refstack-client-devstack-train https://zuul.opendev.org/t/openstack/build/92521652313d4838ba3d79130f8f3c7b : SUCCESS in 1h 19m 48s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/9139a5dc16bd4c979cd89f093a65650c : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":1},{"id":"0b2450e87fb001afce0e4c3d30fe441cbeeebde3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-18 13:42:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a732551916fdfe1243cc024ac68fc670d86d3cd9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-18 15:01:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ec95684553c64531a1503608735b730e : SUCCESS in 6m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ed851725447482f9a9925bcf7d5af4d : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d4386bcf4480437bb1c9db761010dcc6 : SUCCESS in 3m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d8388644797945c594cf327bb0b387ec : SUCCESS in 4m 04s\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/6c384d5cc56b4451ae6a18e5cd0238f6 : SUCCESS in 1h 18m 22s\n- refstack-client-devstack-wallaby https://zuul.opendev.org/t/openstack/build/a6e2f2bb4a024e139cd9dcce1dfc5008 : SUCCESS in 1h 18m 18s\n- refstack-client-devstack-victoria https://zuul.opendev.org/t/openstack/build/109b6130803043a6a9343ba6e4ad8bea : SUCCESS in 1h 07m 49s\n- refstack-client-devstack-ussuri https://zuul.opendev.org/t/openstack/build/4e84891a18024e3a8d7084d9a49693c1 : SUCCESS in 1h 07m 56s\n- refstack-client-devstack-train https://zuul.opendev.org/t/openstack/build/85b615d0b5184dd8943d73f9c8fa9288 : SUCCESS in 1h 14m 19s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/7457123bfc5643b48cab7e21b02f9a89 : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":1},{"id":"124a2f809a3a7f27e95f4c740a3487daab870719","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-18 15:01:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a82594323aa6ecc794498f69446d0aa4ad378bbe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-18 15:04:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- opendev-promote-docs https://zuul.opendev.org/t/openstack/build/d382e967060a41e3b8a977a25a48b31c : SUCCESS in 1m 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e9af50764e216fa6c220f1e0513777ec83dc213f","revisions":{"e9af50764e216fa6c220f1e0513777ec83dc213f":{"kind":"REWORK","_number":1,"created":"2021-12-09 21:49:48.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/25/821325/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack-client","ref":"refs/changes/25/821325/1","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/25/821325/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/25/821325/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/25/821325/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack-client refs/changes/25/821325/1"}}},"commit":{"parents":[{"commit":"ffcdca681b484e85783df96118a77e5b5aad81a5","subject":"Add contributing page and some updates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/ffcdca681b484e85783df96118a77e5b5aad81a5"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2021-12-09 21:07:47.000000000","tz":0},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2021-12-09 21:19:23.000000000","tz":0},"subject":"Increase verbose level for not found tempest tests","message":"Increase verbose level for not found tempest tests\n\nThis warning prints old tempest tests (which are marked\nas aliases in the guidelines) most of the time.\nThe warning confuses users becuase it may suggest that\na relevant test is missing, however, that\u0027s not the case,\naliases aren\u0027t supposed to be found because they were\nrenamed.\n\nTherefore this commit increases the verbose level needed\nto trigger these warnings. From now on, the warnings won\u0027t\nbe printed by default, debug level of verbosity will be\nrequired.\n\nChange-Id: Ic0e590d54d76e2d5839ced3f513e8b2ded705e6f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/e9af50764e216fa6c220f1e0513777ec83dc213f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/e9af50764e216fa6c220f1e0513777ec83dc213f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
