)]}'
{"id":"openinfra%2Frefstack-client~836125","triplet_id":"openinfra%2Frefstack-client~master~If9a7be2c0751ffef47f67a0d6937284a38566f68","project":"openinfra/refstack-client","branch":"master","topic":"interop-jobs","attention_set":{},"removed_from_attention_set":{"22873":{"account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"last_update":"2022-04-28 17:55:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If9a7be2c0751ffef47f67a0d6937284a38566f68","subject":"Update jobs for the new development cycle","status":"MERGED","created":"2022-04-01 08:59:19.000000000","updated":"2022-04-28 17:58:48.000000000","submitted":"2022-04-28 17:55:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"836125-interop-jobs","meta_rev_id":"20b9e8d938df755decfcaf682b05a9b13675b9ff","_number":836125,"virtual_id_number":836125,"owner":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-28 17:55:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"all":[{"value":2,"date":"2022-04-28 16:09:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"value":2,"date":"2022-04-28 15:31:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"all":[{"value":1,"date":"2022-04-28 16:09:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-01 09:53:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-28 15:31:08.000000000","updated_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2022-04-28 16:09:23.000000000","updated_by":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"reviewer":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"state":"REVIEWER"}],"messages":[{"id":"34810d5a9d91361911402c7a375c2d563d607443","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-04-01 08:59:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"224c5f97ba80571c8785b8fccef2e36ee9c19a71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-01 09:53:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2a222f7909494dc09c329d729240dcea : FAILURE in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f125cf9a81e9400b860e8e3be7498f30 : FAILURE in 3m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fca9a9e6519c42468d525065ddd4fbad : FAILURE in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe7d8e525b784505a820648aa1d27d8b : FAILURE in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b701b919d1d44e948ad6a5af4fcc8315 : FAILURE in 3m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6d49c96692834fe08b3784e70987e646 : FAILURE in 3m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5aaa75715fd64c9ba709926492d091d6 : FAILURE in 4m 55s (non-voting)\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/46dbb36523e842ce9a104636b1f8f586 : FAILURE in 40m 32s\n- refstack-client-devstack-yoga https://zuul.opendev.org/t/openstack/build/c833c6a7968243009acf75d23a9e0f9a : FAILURE in 42m 00s\n- refstack-client-devstack-xena https://zuul.opendev.org/t/openstack/build/548b128bc2be49bab8efd8da989eb96d : FAILURE in 45m 34s\n- refstack-client-devstack-wallaby https://zuul.opendev.org/t/openstack/build/00c9f4e62c4e4645bae03217e129fb23 : FAILURE in 45m 21s\n- refstack-client-devstack-victoria https://zuul.opendev.org/t/openstack/build/f7a592d707b349c48fc0ee27bcaeb34c : FAILURE in 44m 41s\n- refstack-client-devstack-ussuri https://zuul.opendev.org/t/openstack/build/0ed141a1a69045cb982f596f27f556e7 : FAILURE in 45m 10s\n- refstack-client-devstack-train https://zuul.opendev.org/t/openstack/build/eb0b64370dfe4b9f9a301a7e5961c353 : FAILURE in 49m 11s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/a3b78dcc350b4a5897a383869cb0179b : FAILURE in 2m 56s","accounts_in_message":[],"_revision_number":1},{"id":"a43e2aab3c414ac2a3342fd1250a7743d2503e34","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-04-04 05:24:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4991907dce8766a0b55a7055dd2916970f19cf38","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-04-04 05:33:57.000000000","message":"Topic set to interop-jobs","accounts_in_message":[],"_revision_number":1},{"id":"97c6b54950568b1c3749dcf647c11f4b9d82a17d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 06:08:38.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f2c5885a291b485f96d9b5c96b7b8ca1 : FAILURE in 3m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f6e0202690a435ca191c6768c0dbe92 : FAILURE in 3m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7121b9c01e9245a995c011504836ff10 : FAILURE in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb05e7ce7aae4f28ab625dce2a8f6b60 : FAILURE in 3m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/edf2a5ce4f4b4b889acd92f48ad3fdde : FAILURE in 3m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/40d636da527b4050844bae28ae7f0630 : FAILURE in 3m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2eaa94fd8ca1472db979701dcb4ce6ef : FAILURE in 4m 28s (non-voting)\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/2451d6470fc1447bb201b098d117c8a2 : FAILURE in 37m 01s\n- refstack-client-devstack-yoga https://zuul.opendev.org/t/openstack/build/8dd1d5f5a1f04010ada375828ceb4b6c : FAILURE in 36m 52s\n- refstack-client-devstack-xena https://zuul.opendev.org/t/openstack/build/eb11762e78a44fc79fa79f199ccff7c8 : FAILURE in 35m 11s\n- refstack-client-devstack-wallaby https://zuul.opendev.org/t/openstack/build/c9c7242eeee94f10baab62faa66588ca : FAILURE in 42m 13s\n- refstack-client-devstack-victoria https://zuul.opendev.org/t/openstack/build/569e88bbbab44a7f89aea19ee6c768ba : FAILURE in 38m 20s\n- refstack-client-devstack-ussuri https://zuul.opendev.org/t/openstack/build/be13205cb3d04ae5a650b79a5f575439 : FAILURE in 35m 02s\n- refstack-client-devstack-train https://zuul.opendev.org/t/openstack/build/04dc65136ac04d458138057619c26497 : FAILURE in 40m 39s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/40c249c7ffba4b259831697c6a8ca4c9 : FAILURE in 2m 40s","accounts_in_message":[],"_revision_number":1},{"id":"4effe8491c0e32c1a120457f5edda9e3385c9141","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-04-20 08:27:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a6f7501f8303987e3333edf10b43d0b17cc995eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 09:25:58.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5c69ac43250946e48d8b36679c80a1fd : FAILURE in 3m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1d0530b0236444ef812bff367a4e09cb : FAILURE in 3m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/876c6f1672944d909c72ed41c9f14131 : FAILURE in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f389315ae60441ab9dcf32cfe80eaed : FAILURE in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7338e7837ea64b42921acea8065b7168 : FAILURE in 3m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d68fce7893b04dc0bd2d8e979ecd3545 : FAILURE in 3m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/20212738c94145c187f7de40d62a8028 : FAILURE in 5m 11s (non-voting)\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/44c85cbc4aef43b5abb6822708f6d3dd : FAILURE in 46m 50s\n- refstack-client-devstack-yoga https://zuul.opendev.org/t/openstack/build/3f573b743eda4bf2b7a9b7aecfeeabcb : FAILURE in 47m 54s\n- refstack-client-devstack-xena https://zuul.opendev.org/t/openstack/build/b4362ba2d47e484db0a2dec36476b36d : FAILURE in 42m 07s\n- refstack-client-devstack-wallaby https://zuul.opendev.org/t/openstack/build/c1cfba44802742adbffb2ad99d22eee5 : FAILURE in 50m 29s\n- refstack-client-devstack-victoria https://zuul.opendev.org/t/openstack/build/1f1e12d20f8b4467bc94629f6fe99592 : FAILURE in 47m 23s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/53db819a95fd4229b2216bf2a00ee18e : FAILURE in 2m 44s","accounts_in_message":[],"_revision_number":2},{"id":"bb644ef0d7d433b8a451674e3fdbf5c77fc13843","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-04-20 12:44:33.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3692c92ad8a2c981019631cf47f9ed3c0064a2ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 13:27:38.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7bee07f7ce3040d79450809790772a9b : FAILURE in 3m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4d9c56eb089940adb42e4c126e32318f : FAILURE in 3m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2b675f7364fd4bc49e7f686eb4699a92 : FAILURE in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf6f661b85e04fb8b8ee8d1204cee17f : FAILURE in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd4449549d89413ea38b9fc170290e27 : FAILURE in 3m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8469e2becfcf454f81611039496a7652 : FAILURE in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ebe33fabc0454992a2b1f9fb95fc7c7c : FAILURE in 5m 01s (non-voting)\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/920c0b14351140219a94f27c964b0a91 : FAILURE in 32m 53s\n- refstack-client-devstack-yoga https://zuul.opendev.org/t/openstack/build/fbefafa90e5b4986a8b88007d77c94bc : FAILURE in 34m 10s\n- refstack-client-devstack-xena https://zuul.opendev.org/t/openstack/build/2bcaecd3ffbe460793936e619d65e315 : FAILURE in 39m 21s\n- refstack-client-devstack-wallaby https://zuul.opendev.org/t/openstack/build/88c33a899edc478cb60af09b30e05fc7 : FAILURE in 36m 51s\n- refstack-client-devstack-victoria https://zuul.opendev.org/t/openstack/build/f05a1a479a214a63b09e1fb3254f0c2b : FAILURE in 40m 59s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/657cf774633b433a9062f4a6709b867c : FAILURE in 2m 41s","accounts_in_message":[],"_revision_number":3},{"id":"cf68ae35d478657feb4f524a7afb6d0f5320d673","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-04-21 13:56:50.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"9a350d60b92e3cfc4e6538cecd4d979153b44382","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 14:56:31.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d40543a35b6f41b48c342300fa575f2a : FAILURE in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/01c4e00ba7f64ece944f8cf0edc411c7 : FAILURE in 3m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d031d140dbc147ce9586d8369735612d : FAILURE in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e70e6159d56a46cbaf8d9fd1c4b0cce2 : FAILURE in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f4e9d1cd28c840b9a20e234590a25101 : FAILURE in 3m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f0fe09466fb440c4bf280466b47b4b56 : FAILURE in 4m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/47ab857e174b4c188a49018af42d7e6a : FAILURE in 5m 46s (non-voting)\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/71d0668b6f1245238ebd4d98e5fa3f18 : FAILURE in 41m 23s\n- refstack-client-devstack-yoga https://zuul.opendev.org/t/openstack/build/f1c876f76fea4d7194799ed3eac9b836 : FAILURE in 45m 14s\n- refstack-client-devstack-xena https://zuul.opendev.org/t/openstack/build/4638cbc3286340c8a313943187c449e6 : FAILURE in 41m 46s\n- refstack-client-devstack-wallaby https://zuul.opendev.org/t/openstack/build/dde8f34291b4443693e4ffcbbe1347f9 : FAILURE in 50m 41s\n- refstack-client-devstack-victoria https://zuul.opendev.org/t/openstack/build/81b359bedaf74c8f98bd888813e82c63 : FAILURE in 46m 47s\n- refstack-client-devstack-ussuri https://zuul.opendev.org/t/openstack/build/41d8401b570c4b3bb47d2d2e443ce4ee : FAILURE in 43m 04s\n- refstack-client-devstack-train https://zuul.opendev.org/t/openstack/build/b304922d6b904ae5a617d9c9ca431ad6 : FAILURE in 43m 57s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/091395b182664aa59ece2a407e8109b5 : FAILURE in 3m 18s","accounts_in_message":[],"_revision_number":4},{"id":"054bb05caf670f170a4f2e6af2fbacd63c8855cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-04-22 11:45:51.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"b3e73009cd9c2f68c9f4916d2ce17335a81fadeb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 13:36:09.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/88deb6e6c2234f4a8e8d55701bc90903 : SUCCESS in 3m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d9869a7f7f746ada4f2da5753b2d1ac : SUCCESS in 3m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/84746b025d1b454a8133cb931f0ce5e9 : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29255d40a88b471ebd707a9b48d92823 : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f466754a9c24c92864f8d551988f21b : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f2143a947bc544e887dff1e68cc8cd33 : SUCCESS in 4m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/497941137f33496d94aae532887f659e : SUCCESS in 5m 34s (non-voting)\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/1a3154efc3d3430b8f75a4ca42b83ada : SUCCESS in 1h 32m 42s\n- refstack-client-devstack-yoga https://zuul.opendev.org/t/openstack/build/05411607accf48f18c26b6a60044533a : SUCCESS in 1h 43m 25s\n- refstack-client-devstack-xena https://zuul.opendev.org/t/openstack/build/eea5a52558c84e778ae9b0206d9e50df : SUCCESS in 1h 28m 59s\n- refstack-client-devstack-wallaby https://zuul.opendev.org/t/openstack/build/bbdac352595e47bab73c3d20789c4432 : SUCCESS in 1h 42m 49s\n- refstack-client-devstack-victoria https://zuul.opendev.org/t/openstack/build/1709beb327bc438e8612e955139d69d3 : SUCCESS in 1h 23m 35s\n- refstack-client-devstack-ussuri https://zuul.opendev.org/t/openstack/build/9418b840dab6420ca662b17c01322df4 : SUCCESS in 1h 40m 55s\n- refstack-client-devstack-train https://zuul.opendev.org/t/openstack/build/7113a1636dc84274ab80bc065db45f29 : SUCCESS in 1h 27m 24s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/3b85c6ce910a4beeb2dbacacea33e97f : SUCCESS in 2m 48s","accounts_in_message":[],"_revision_number":5},{"id":"f1ca2921de82616f5f0fabd64e7bdb887930112a","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2022-04-28 15:31:08.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"3a689910a1abaec43c860b87f3d68dd076b66d95","author":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"date":"2022-04-28 16:09:23.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"58ca3d519231065ec62b60357af58c11ec37b1a5","author":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"date":"2022-04-28 16:09:28.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"62aa4e161813b026f36fa7fdc5152a77eda45fd8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-28 16:10:36.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"d55fb3cfaf2a80791b2f45b675707c8d375c793c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-28 17:55:57.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/84a3c407790b4ff5bfbe336efac5f338 : SUCCESS in 3m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5efebab84b044e8590983b01343c05b7 : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b44f53f28a864ad4868ec7b4c4a86a41 : SUCCESS in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fff153fb321c403a8967732c7a3a79e2 : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eeaa621343da466686459656709059c0 : SUCCESS in 3m 36s\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/e6201be6a7e443d396e723f57b4df925 : SUCCESS in 1h 27m 10s\n- refstack-client-devstack-yoga https://zuul.opendev.org/t/openstack/build/8f9fa0fa4c1c4a83ac10df7d4e5ad38e : SUCCESS in 1h 37m 48s\n- refstack-client-devstack-xena https://zuul.opendev.org/t/openstack/build/7f3205e1c014462fbd3b85ecbc04d585 : SUCCESS in 1h 33m 21s\n- refstack-client-devstack-wallaby https://zuul.opendev.org/t/openstack/build/811b1fadb2b0469cb1f0abab46ee2bb8 : SUCCESS in 1h 28m 40s\n- refstack-client-devstack-victoria https://zuul.opendev.org/t/openstack/build/5e616ff2b5a44166b92ab95c52feca12 : SUCCESS in 1h 39m 15s\n- refstack-client-devstack-ussuri https://zuul.opendev.org/t/openstack/build/11c59f5ca4fc4c19b3e9afc6fbced0bf : SUCCESS in 1h 26m 28s\n- refstack-client-devstack-train https://zuul.opendev.org/t/openstack/build/b5ab7044fafb401280978f2396086e70 : SUCCESS in 1h 33m 22s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/921c6ebe996e4867b6983ad4f7355d6f : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":5},{"id":"a00e62eec640d6387cba2f6078ab09a511556a2c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-28 17:55:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"20b9e8d938df755decfcaf682b05a9b13675b9ff","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-28 17:58:48.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- opendev-promote-docs https://zuul.opendev.org/t/openstack/build/2a358048ab0742ceb0f9ee470000a4d2 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"d320c6c60abc2cf2e78990952a9a747d1dbb545a","revisions":{"d844b35466730bd56db2b2a4895523f81d103c7c":{"kind":"REWORK","_number":1,"created":"2022-04-01 08:59:19.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/25/836125/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack-client","ref":"refs/changes/25/836125/1","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/25/836125/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/25/836125/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/25/836125/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack-client refs/changes/25/836125/1"}}},"commit":{"parents":[{"commit":"225c977a1231eecc112d6e1946d79a545c5050c7","subject":"Merge \"Run apt-get with DEBIAN_FRONTEND noninteractive\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/225c977a1231eecc112d6e1946d79a545c5050c7"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2022-04-01 08:56:07.000000000","tz":120},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2022-04-01 08:58:52.000000000","tz":120},"subject":"Update jobs for the new development cycle","message":"Update jobs for the new development cycle\n\nAdding jobs for the Zed development cycle.\n\nDepends-On: https://review.opendev.org/c/openinfra/ansible-role-refstack-client/+/836124\nChange-Id: If9a7be2c0751ffef47f67a0d6937284a38566f68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/d844b35466730bd56db2b2a4895523f81d103c7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/d844b35466730bd56db2b2a4895523f81d103c7c"}]},"branch":"refs/heads/master"},"9d7f624be02f3ce846ac81ba5ec28401cf565b3c":{"kind":"REWORK","_number":2,"created":"2022-04-20 08:27:53.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/25/836125/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack-client","ref":"refs/changes/25/836125/2","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/25/836125/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/25/836125/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/25/836125/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack-client refs/changes/25/836125/2"}}},"commit":{"parents":[{"commit":"225c977a1231eecc112d6e1946d79a545c5050c7","subject":"Merge \"Run apt-get with DEBIAN_FRONTEND noninteractive\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/225c977a1231eecc112d6e1946d79a545c5050c7"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2022-04-01 08:56:07.000000000","tz":120},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2022-04-20 08:26:08.000000000","tz":120},"subject":"Update jobs for the new development cycle","message":"Update jobs for the new development cycle\n\nThis updates jobs for the Zed development cycle:\n* a yoga job is added\n* ussuri and train jobs are removed\n\nDepends-On: https://review.opendev.org/c/openinfra/ansible-role-refstack-client/+/836124\nChange-Id: If9a7be2c0751ffef47f67a0d6937284a38566f68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/9d7f624be02f3ce846ac81ba5ec28401cf565b3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/9d7f624be02f3ce846ac81ba5ec28401cf565b3c"}]},"branch":"refs/heads/master"},"1ee725f3539f09bea60b8c81b55ff53b92597c52":{"kind":"REWORK","_number":3,"created":"2022-04-20 12:44:33.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/25/836125/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack-client","ref":"refs/changes/25/836125/3","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/25/836125/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/25/836125/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/25/836125/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack-client refs/changes/25/836125/3"}}},"commit":{"parents":[{"commit":"225c977a1231eecc112d6e1946d79a545c5050c7","subject":"Merge \"Run apt-get with DEBIAN_FRONTEND noninteractive\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/225c977a1231eecc112d6e1946d79a545c5050c7"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2022-04-01 08:56:07.000000000","tz":120},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2022-04-20 12:44:22.000000000","tz":120},"subject":"Update jobs for the new development cycle","message":"Update jobs for the new development cycle\n\nThis updates jobs for the Zed development cycle:\n* a yoga job is added\n* ussuri and train jobs are removed\n\nDepends-On: https://review.opendev.org/c/openinfra/ansible-role-refstack-client/+/836124\nChange-Id: If9a7be2c0751ffef47f67a0d6937284a38566f68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/1ee725f3539f09bea60b8c81b55ff53b92597c52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/1ee725f3539f09bea60b8c81b55ff53b92597c52"}]},"branch":"refs/heads/master"},"d927d5fef5ce364eadf08610737ace525775e43a":{"kind":"REWORK","_number":4,"created":"2022-04-21 13:56:50.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/25/836125/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack-client","ref":"refs/changes/25/836125/4","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/25/836125/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/25/836125/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/25/836125/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack-client refs/changes/25/836125/4"}}},"commit":{"parents":[{"commit":"225c977a1231eecc112d6e1946d79a545c5050c7","subject":"Merge \"Run apt-get with DEBIAN_FRONTEND noninteractive\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/225c977a1231eecc112d6e1946d79a545c5050c7"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2022-04-01 08:56:07.000000000","tz":120},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2022-04-21 13:56:09.000000000","tz":120},"subject":"Update jobs for the new development cycle","message":"Update jobs for the new development cycle\n\nThis updates jobs for the Zed development cycle by adding a yoga\nand a xena jobs.\nAlso apply a workaround for the git issue caused by:\nhttps://github.com/git/git/commit/8959555cee7ec045958f9b6dd62e541affb7e7d9\n\nDepends-On: https://review.opendev.org/c/openinfra/ansible-role-refstack-client/+/836124\nChange-Id: If9a7be2c0751ffef47f67a0d6937284a38566f68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/d927d5fef5ce364eadf08610737ace525775e43a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/d927d5fef5ce364eadf08610737ace525775e43a"}]},"branch":"refs/heads/master"},"d320c6c60abc2cf2e78990952a9a747d1dbb545a":{"kind":"REWORK","_number":5,"created":"2022-04-22 11:45:51.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/25/836125/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack-client","ref":"refs/changes/25/836125/5","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/25/836125/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/25/836125/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/25/836125/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack-client refs/changes/25/836125/5"}}},"commit":{"parents":[{"commit":"225c977a1231eecc112d6e1946d79a545c5050c7","subject":"Merge \"Run apt-get with DEBIAN_FRONTEND noninteractive\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/225c977a1231eecc112d6e1946d79a545c5050c7"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2022-04-01 08:56:07.000000000","tz":120},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2022-04-22 11:43:54.000000000","tz":120},"subject":"Update jobs for the new development cycle","message":"Update jobs for the new development cycle\n\nThis updates jobs for the Zed development cycle by adding a yoga\nand a xena jobs.\nThis also bumps refstack-client\u0027s version in setup.cfg as we created\n0.1.0 tag couple of weeks ago - the version there refers to a new\nversion (released in the future)\n\nDepends-On: https://review.opendev.org/c/openinfra/ansible-role-refstack-client/+/836124\nChange-Id: If9a7be2c0751ffef47f67a0d6937284a38566f68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/d320c6c60abc2cf2e78990952a9a747d1dbb545a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/d320c6c60abc2cf2e78990952a9a747d1dbb545a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
