)]}'
{"id":"openinfra%2Frefstack-client~841326","triplet_id":"openinfra%2Frefstack-client~master~I12cc6206ebfe06b7cf3a50ea59bf1b55a3ff7303","project":"openinfra/refstack-client","branch":"master","attention_set":{},"removed_from_attention_set":{"22873":{"account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"last_update":"2022-05-12 18:00:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I12cc6206ebfe06b7cf3a50ea59bf1b55a3ff7303","subject":"Update setup.cfg and tox.ini","status":"MERGED","created":"2022-05-10 18:57:44.000000000","updated":"2022-05-12 18:01:26.000000000","submitted":"2022-05-12 18:00:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"841326","meta_rev_id":"b1c07a9439f11ffe6df94fcc376c026cec624e4d","_number":841326,"virtual_id_number":841326,"owner":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-12 18:00:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2022-05-12 16:08:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2022-05-12 13:24:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":1,"date":"2022-05-12 16:02:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-10 20:20:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-12 13:24:39.000000000","updated_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2022-05-12 16:08:15.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"}],"messages":[{"id":"d6b4701abc65e57383c3303a767ffba814545755","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-05-10 18:57:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"784ef5b1ea9a6830862203fa432ba6577c4fa863","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 20:20:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6692ed3a8fba4d1799eb59caf5a8f149 : SUCCESS in 3m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3df6f4c54fd74429bfe36fb34a78138c : SUCCESS in 3m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d2e1ef491db6481388312b949c715d4d : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df8817d17cfa48eeaa78a4ddafe2db85 : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/69541858107d4c2a93693a73234dff35 : SUCCESS in 3m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e25f0e51a56342dc8895d2022cfad1b1 : SUCCESS in 3m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ba6abecf818f4a829f44ca7211b41242 : SUCCESS in 4m 55s (non-voting)\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/3a98e4c6edba48c49f57beb662f73d87 : SUCCESS in 1h 16m 49s\n- refstack-client-devstack-yoga https://zuul.opendev.org/t/openstack/build/731c55c3584f438488c7501b2a65d54e : SUCCESS in 1h 15m 29s\n- refstack-client-devstack-xena https://zuul.opendev.org/t/openstack/build/90ff0d48600849edab0df5d806516833 : SUCCESS in 1h 07m 33s\n- refstack-client-devstack-wallaby https://zuul.opendev.org/t/openstack/build/6af4e21a0e924ba9bea3641eb07b6157 : SUCCESS in 1h 16m 11s\n- refstack-client-devstack-victoria https://zuul.opendev.org/t/openstack/build/e061132ea355425dbc2708b4d5e9fab4 : SUCCESS in 1h 15m 19s\n- refstack-client-devstack-ussuri https://zuul.opendev.org/t/openstack/build/96047b1a82054658ad52a5043ae0605f : SUCCESS in 1h 18m 52s\n- refstack-client-devstack-train https://zuul.opendev.org/t/openstack/build/9ba01a210dbd4094b166c6d97070c4a6 : SUCCESS in 1h 00m 28s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/55b7d89da7434f0e98f2834351fb38dd : SUCCESS in 2m 52s","accounts_in_message":[],"_revision_number":1},{"id":"6d88f6f3417d8db14e7430adabf70653870b9ede","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2022-05-12 13:24:39.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9a2477558ac40a7ca604dfeb423d128b70ff4077","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-05-12 16:02:35.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"81c5ea6e095ec6466eda6f3f3eea359c169db28a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 16:03:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"902c76b1a151a3ebe7d20682d844526c5bfc1e7f","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-05-12 16:08:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"47250e87521c74e46a6e3ae4a4979193d71ce2a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 18:00:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3869f1e17e9c47aca74a2766e8ecee0e : SUCCESS in 3m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/97c0f8e8cb034ea6bf0f7962f987c9db : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eafc34cf27b14ea4b56a92ba46f73a8f : SUCCESS in 3m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f7bc14af24ec4439ae1e270fd78dfe03 : SUCCESS in 3m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/25fc69842567452b92557d9513bc0ec1 : SUCCESS in 3m 39s\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/ee371db9d81d42a5be2436b41b4c8168 : SUCCESS in 1h 40m 01s\n- refstack-client-devstack-yoga https://zuul.opendev.org/t/openstack/build/fc68f7d1617a43b894e9a80e9fe0e0f4 : SUCCESS in 1h 34m 20s\n- refstack-client-devstack-xena https://zuul.opendev.org/t/openstack/build/8fd9091c4de544b9985785eece3d4cc8 : SUCCESS in 1h 39m 42s\n- refstack-client-devstack-wallaby https://zuul.opendev.org/t/openstack/build/39efe32b8478499c88b061996223da4e : SUCCESS in 1h 48m 53s\n- refstack-client-devstack-victoria https://zuul.opendev.org/t/openstack/build/1deccf902cf64189837684cbc07e4f1a : SUCCESS in 1h 33m 17s\n- refstack-client-devstack-ussuri https://zuul.opendev.org/t/openstack/build/b326b1758c0b44ebab2bc092ccfcb86f : SUCCESS in 1h 30m 59s\n- refstack-client-devstack-train https://zuul.opendev.org/t/openstack/build/0c0338b44b004ce9a27ec439e158889b : SUCCESS in 1h 29m 28s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/53f08bf5041d466ca00e2485002ea85e : SUCCESS in 2m 49s","accounts_in_message":[],"_revision_number":1},{"id":"67e4b0cf1778ee644b5b238ded1564da81849a3d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 18:00:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b1c07a9439f11ffe6df94fcc376c026cec624e4d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 18:01:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- opendev-promote-docs https://zuul.opendev.org/t/openstack/build/d5d84255044d4808948d149c6183eba7 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eb3c6d4957253cee8488e227a85863f2e58f6b77","revisions":{"eb3c6d4957253cee8488e227a85863f2e58f6b77":{"kind":"REWORK","_number":1,"created":"2022-05-10 18:57:44.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/26/841326/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack-client","ref":"refs/changes/26/841326/1","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/26/841326/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/26/841326/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/26/841326/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack-client refs/changes/26/841326/1"}}},"commit":{"parents":[{"commit":"d320c6c60abc2cf2e78990952a9a747d1dbb545a","subject":"Update jobs for the new development cycle","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/d320c6c60abc2cf2e78990952a9a747d1dbb545a"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2022-05-10 18:47:32.000000000","tz":120},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2022-05-10 18:56:51.000000000","tz":120},"subject":"Update setup.cfg and tox.ini","message":"Update setup.cfg and tox.ini\n\nAdd Python 3.9 to the list of supported Pythons and replace\ndashes (which are deprecated) with underscores.\n\nSetuptools v54.1.0 introduces a warning that the use of dash-separated\noptions in \u0027setup.cfg\u0027 will not be supported in a future version [1].\nGet ahead of the issue by replacing the dashes with underscores. Without\nthis, we see \u0027UserWarning\u0027 messages like the following on new enough\nversions of setuptools:\n\n  UserWarning: Usage of dash-separated \u0027description-file\u0027 will not be\n  supported in future versions. Please use the underscore name\n  \u0027description_file\u0027 instead\n\n[1] https://github.com/pypa/setuptools/commit/a2e9ae4cb\n\nChange-Id: I12cc6206ebfe06b7cf3a50ea59bf1b55a3ff7303\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/eb3c6d4957253cee8488e227a85863f2e58f6b77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/eb3c6d4957253cee8488e227a85863f2e58f6b77"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
