)]}'
{"id":"openinfra%2Frefstack-client~846370","triplet_id":"openinfra%2Frefstack-client~master~Idf69f02ba844bdeea21ec744ec41ca78bbd1907d","project":"openinfra/refstack-client","branch":"master","attention_set":{},"removed_from_attention_set":{"22873":{"account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"last_update":"2022-08-23 19:16:15.000000000","reason":"\u003cGERRIT_ACCOUNT_22873\u003e replied on the change","reason_account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},"10459":{"account":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"last_update":"2022-08-23 20:54:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idf69f02ba844bdeea21ec744ec41ca78bbd1907d","subject":"Update tempest version and drop U and T jobs","status":"MERGED","created":"2022-06-17 13:43:58.000000000","updated":"2022-08-23 20:56:58.000000000","submitted":"2022-08-23 20:54:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"846370","meta_rev_id":"a8e247ebe28ea0f9489d3fa513e0d6ee3604e161","_number":846370,"virtual_id_number":846370,"owner":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-23 20:54:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":2,"date":"2022-08-23 19:16:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":1,"date":"2022-08-23 19:16:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}],"CC":[{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-17 15:56:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-19 12:13:16.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"CC"}],"messages":[{"id":"3f69a66fd6b323dea35736dcd09f3724e1998870","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-06-17 13:43:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f6f510481facd1d360a2a8d976d6ac3019904f12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 15:56:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/26d193f4271244b5a603126e7afaec56 : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/40f099cd41694d7581ce40e8de7721e0 : SUCCESS in 3m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/69ff69afa6d94ef78ed5177f115f5128 : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a382dd187b5420ead0f70ea9b8751bf : SUCCESS in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01bb22e1ed834098851db35028a9d8e5 : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/72b9a3eb8bac49048a5843304d0cc375 : SUCCESS in 7m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/84632cef612a458eb40047daf1f19f0e : SUCCESS in 4m 24s (non-voting)\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/b3723a507e4d4c3e851e2f9ed67f39e5 : TIMED_OUT in 2h 04m 35s\n- refstack-client-devstack-yoga https://zuul.opendev.org/t/openstack/build/a3a8a336e55a43f98a1912ab6cbc823a : TIMED_OUT in 2h 04m 44s\n- refstack-client-devstack-xena https://zuul.opendev.org/t/openstack/build/3589f8702eb2491db9a53f336cf3b074 : TIMED_OUT in 2h 05m 13s\n- refstack-client-devstack-wallaby https://zuul.opendev.org/t/openstack/build/091128bac2254c5490fbf12fe87420e0 : TIMED_OUT in 2h 04m 02s\n- refstack-client-devstack-victoria https://zuul.opendev.org/t/openstack/build/3bbb8ef80ad74f289dacb3bf90142a3e : TIMED_OUT in 2h 04m 31s\n- refstack-client-devstack-ussuri https://zuul.opendev.org/t/openstack/build/6a517ddbdbbd4bbaa17bac41d99c6ebf : FAILURE in 45m 04s\n- refstack-client-devstack-train https://zuul.opendev.org/t/openstack/build/f33a1a4491b24caab84cd7da48d410cb : FAILURE in 45m 30s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/b2ff37c7aab84056ac783a70190935f6 : SUCCESS in 3m 02s","accounts_in_message":[],"_revision_number":1},{"id":"3b189345490724bf89cb1bf424b172368c49105f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-06-20 13:49:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"400b06fdb242cfdd52798058bcf506b57943bc89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 16:01:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/18a4ad32e35a44d5b1c415fc6e9ced4a : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c4c4cca4339a4115b997b6dceebe2bbc : SUCCESS in 3m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0b605f981ef74c03a649d37c8c53f255 : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/921e9fbe087848f687e4996f4c06507a : SUCCESS in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e763c5bcd1b447a4a73ddd38c0fcf92c : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/01df9786d52c48e3bbf91dd1c542f4c9 : SUCCESS in 3m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7ad169a8efd44cfc9b6fe05b85451278 : SUCCESS in 3m 50s (non-voting)\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/3c5e9b19ecd6425f9149c62c78f28b67 : TIMED_OUT in 2h 03m 48s\n- refstack-client-devstack-yoga https://zuul.opendev.org/t/openstack/build/ed64d50708f4455a9a78fb3496b964c4 : TIMED_OUT in 2h 03m 55s\n- refstack-client-devstack-xena https://zuul.opendev.org/t/openstack/build/fe17f7696623435082e9494c84ac1ec2 : TIMED_OUT in 2h 04m 41s\n- refstack-client-devstack-wallaby https://zuul.opendev.org/t/openstack/build/2ff822929fc7470c9bf78b9d18ff7369 : TIMED_OUT in 2h 03m 52s\n- refstack-client-devstack-victoria https://zuul.opendev.org/t/openstack/build/6c4375e0460f4f57934a2ca055930805 : TIMED_OUT in 2h 04m 23s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/c8597e580c054c10874caea844838671 : SUCCESS in 2m 48s","accounts_in_message":[],"_revision_number":2},{"id":"e2506f9024e0cb3bd327830b26f3f0500cfc9f27","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-07-20 15:35:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"19ddb69530c19d707357e241dc133d245f0c5e8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 17:53:27.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/623e76d24da145c7b424fd2a14f88b4c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d440663228af4b82bd059495c3524954 : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c682d06d5ee3477a9fae0f91bf87c807 : SUCCESS in 4m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1306dd934886490ca86ca1adceb956e6 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cdac7a3b327d4a758cf10e0eb5c12d9b : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3bfe8c6de9dc40228e59ceb3bfc1af07 : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/301ac9af1849447baa7c2c0e19a73fc4 : SUCCESS in 4m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/21f0b166c0ac4fc4a868c0198aa45ca3 : SUCCESS in 6m 26s (non-voting)\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/93da30be7371490eab19f7194dd23e2c : TIMED_OUT in 2h 05m 55s\n- refstack-client-devstack-yoga https://zuul.opendev.org/t/openstack/build/d5f4c3c47fb242abbda27ab8b0904b65 : TIMED_OUT in 2h 13m 53s\n- refstack-client-devstack-xena https://zuul.opendev.org/t/openstack/build/77ee3476514747b8976b8fd94ac56e4e : TIMED_OUT in 2h 14m 26s\n- refstack-client-devstack-wallaby https://zuul.opendev.org/t/openstack/build/e06c6eff688241f08510270a15eea367 : FAILURE in 50m 25s\n- refstack-client-devstack-victoria https://zuul.opendev.org/t/openstack/build/612c2dca98a0400fa9e1f4c7c9849e17 : TIMED_OUT in 2h 14m 05s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/2eee4e1e11d84a9a8c2ef3a48b4bd94b : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":3},{"id":"8d1261684b24faaa8fb604dd9f2a06a92c243a3d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-07-21 08:26:38.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e184683389893261396e248bb3d3512f7f8afa7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 10:15:13.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f8fb2f05f2f74b098cccaec038da840d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8641e8ca6e9f49e086eb120e9ee8198e : SUCCESS in 6m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5844140352b44c4596d77f21ba162136 : SUCCESS in 2m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/589f8a1842c0478ead1c566d4856c17a : SUCCESS in 7m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffd27fed07c144caae9dcc7d7c467c43 : SUCCESS in 9m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fb5f2839b5c642c58e51c0f8b2478de2 : SUCCESS in 6m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b648040e56d740b398b4df5ddb6d0ab7 : SUCCESS in 7m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/abe9810b721548da87ab2ea20c904666 : SUCCESS in 3m 22s (non-voting)\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/cae3f2ddc1164e58a77f7acd9d4107e5 : SUCCESS in 1h 43m 24s\n- refstack-client-devstack-yoga https://zuul.opendev.org/t/openstack/build/1db24bc938c54f6bb845653c81d0c668 : FAILURE in 1h 45m 13s\n- refstack-client-devstack-xena https://zuul.opendev.org/t/openstack/build/b6e6504caf3641f3bdae2d2226a6baa2 : FAILURE in 1h 38m 36s\n- refstack-client-devstack-wallaby https://zuul.opendev.org/t/openstack/build/8978d3499fc04e6ba88cf975ee15dd45 : FAILURE in 53m 44s\n- refstack-client-devstack-victoria https://zuul.opendev.org/t/openstack/build/629af525fb84404383971edaaebbfde5 : FAILURE in 1h 44m 53s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/716a1feea03747a3a44865573bd45542 : SUCCESS in 5m 45s","accounts_in_message":[],"_revision_number":4},{"id":"10996fd422c663ff74405fefcf841f299e764ed0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-07-21 12:22:01.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"aa8d2d1d11567c9a999a1bca1adea6c43258a856","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 13:20:34.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4f066874982845aea690d9c6f6db5a3e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/58eaea05a62940988ea9e187139cac9d : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eaa6a2c337834e9e9ca08223e6de37e8 : SUCCESS in 4m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b6e3196c2e49471e96f626a1feb89133 : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f74afcf8f63446f3be3bd80825a9594a : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/14b8ae872fb64092ada617be8528590c : SUCCESS in 4m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/476626454876450c9c0411e7a9b72d79 : SUCCESS in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/34a4cdcbb77043789d9ce6d7b84a9547 : SUCCESS in 7m 04s (non-voting)\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/174c0454dba1477786f3d0192c8c8a1d : FAILURE in 39m 46s\n- refstack-client-devstack-yoga https://zuul.opendev.org/t/openstack/build/409de0b70dc84b609be14aec60bee4fe : FAILURE in 39m 39s\n- refstack-client-devstack-xena https://zuul.opendev.org/t/openstack/build/c8b537f6b6e54be6a38bbddf76b9201f : FAILURE in 14m 33s\n- refstack-client-devstack-wallaby https://zuul.opendev.org/t/openstack/build/cff9b55ce4f8440bbca228dbb7edaf27 : FAILURE in 22m 07s\n- refstack-client-devstack-victoria https://zuul.opendev.org/t/openstack/build/333aea2acca2401f9041797d78ee585d : FAILURE in 56m 13s\n- refstack-client-devstack-ussuri https://zuul.opendev.org/t/openstack/build/c5be7134b7de4675a7be95272fb2992b : FAILURE in 39m 30s\n- refstack-client-devstack-train https://zuul.opendev.org/t/openstack/build/545faa68352248888f8e8002fd8fe71c : FAILURE in 38m 16s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/ad9f5966fff94bd698988cf0d0ffc294 : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":5},{"id":"74dfd7a0d27687795b103ec26e147c4558c6f3db","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-07-21 13:24:35.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e57ec543ad08b3cc38248cc53c35938a565b47db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 14:25:19.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/78e68e31a7844c78bbed543111ab76d3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/00e392a08df7476d84f0cee339116e20 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e911b7d5f6d4715a31cc8b287ce8325 : SUCCESS in 6m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6614f6ba293a438ebfb5eed8d43e1b1a : SUCCESS in 7m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e4f57ab9e50473c8fffa3639daeb4a2 : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/82bf18034dfb44bf8251dfe440b7a3d2 : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/04a1cb5251344cc5a481ae364c8d06f0 : SUCCESS in 6m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ca922965973a4a7a824bf7b23d6228fe : SUCCESS in 7m 15s (non-voting)\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/7792217eac934fd6840131302c0c8793 : FAILURE in 54m 14s\n- refstack-client-devstack-yoga https://zuul.opendev.org/t/openstack/build/7a9d0d1c833f4c1b96740288692cdeb0 : FAILURE in 55m 05s\n- refstack-client-devstack-xena https://zuul.opendev.org/t/openstack/build/edd3124815d94707adbc8fd1008388b1 : FAILURE in 23m 38s\n- refstack-client-devstack-wallaby https://zuul.opendev.org/t/openstack/build/0f0ba2af4dce41a0b5c15dbb2171eb9a : FAILURE in 24m 01s\n- refstack-client-devstack-victoria https://zuul.opendev.org/t/openstack/build/6a8ecfd5ef934d1382149b08201deae9 : FAILURE in 57m 19s\n- refstack-client-devstack-ussuri https://zuul.opendev.org/t/openstack/build/a067e760dbfb45559297a16ab6b8884e : FAILURE in 52m 35s\n- refstack-client-devstack-train https://zuul.opendev.org/t/openstack/build/c14757158c9d46769c7f087ac05b6dbd : FAILURE in 41m 18s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/741898c1b65c409a913806fc058bd618 : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":5},{"id":"4325b5bee46aa8c1a0c6afa4cd0344f0863d7aa1","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-07-22 11:48:57.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ab8eb71fe4f87446eaee2b53e6bcfc7ff93872c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-22 12:36:28.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/63c3d3af35c54668b781b30dbdea0d05\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef380070fba24e23b6833795e2df3d1f : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/88f46f526f7049f68f9d4620cd265989 : SUCCESS in 4m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4d16e6ebf55148c4b2b410067e910664 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4352bbb0d38443548c7bd89297229463 : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/183f9d3636fc4bb7a0759491082cb540 : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/73b4d0cbe66a4152b5e8f882742c90e3 : SUCCESS in 4m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2ab4d356052c4d0ab49a614a14d4ba55 : SUCCESS in 4m 25s (non-voting)\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/a19c5d1aa58548bc9dd11abc0217dd61 : FAILURE in 41m 51s\n- refstack-client-devstack-yoga https://zuul.opendev.org/t/openstack/build/0f3caf905d524e26b107ae581392bd9b : FAILURE in 45m 35s\n- refstack-client-devstack-xena https://zuul.opendev.org/t/openstack/build/bbb87d240821435dae885f7fb2ed057a : FAILURE in 35m 02s\n- refstack-client-devstack-wallaby https://zuul.opendev.org/t/openstack/build/87df7afb167840dc9d34c1e28dd1abf5 : FAILURE in 41m 29s\n- refstack-client-devstack-victoria https://zuul.opendev.org/t/openstack/build/fdd9a0138b5f49f7b8fc3cf5d1ef07cd : FAILURE in 45m 56s\n- refstack-client-devstack-ussuri https://zuul.opendev.org/t/openstack/build/55e1abdc5ba0420b86b550d5706190eb : FAILURE in 44m 46s\n- refstack-client-devstack-train https://zuul.opendev.org/t/openstack/build/63de36fdfb6b45d79a471c92ff1e1d8f : FAILURE in 44m 13s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/2bcaeb746ab24a3eb48d0cb7daf0dc48 : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":5},{"id":"3b14ccdc59e86c5b790f23e8790a8e26e255f6e5","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-07-22 12:57:25.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"0d1dbca7cde32e21eb0418c36a6f96a2bbcc279a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-22 14:28:11.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/85088eca85f4494b920112b7268055ee\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8eeaeb412b3b49889697dad7dc56d92c : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a6c86762f1294daa910d2d65fc788a05 : SUCCESS in 4m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6eb22cd5dad84167917d74c7f4b984dd : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2826c5a012140dd8a7ae908873c7d2b : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/16652129e5ef4f30966c6f3167c81717 : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f0e49b994c9d46dc95e77ec7a47ddb42 : SUCCESS in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/648566f416ba491786e90a0f29d36717 : SUCCESS in 4m 17s (non-voting)\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/d6e08fdf2aba41ceaad4c668e2649f5c : FAILURE in 1h 27m 48s\n- refstack-client-devstack-yoga https://zuul.opendev.org/t/openstack/build/2e1117a8cfa046579909176fe8869678 : SUCCESS in 1h 24m 29s\n- refstack-client-devstack-xena https://zuul.opendev.org/t/openstack/build/c5fd89fb2aa44429b36c1005a15e0213 : FAILURE in 1h 21m 06s\n- refstack-client-devstack-wallaby https://zuul.opendev.org/t/openstack/build/4bc2ce9b86d74228bc3b13980b630de1 : FAILURE in 44m 42s\n- refstack-client-devstack-victoria https://zuul.opendev.org/t/openstack/build/68e0e5ac831244a395579c3bc3153af5 : FAILURE in 1h 29m 04s\n- refstack-client-devstack-ussuri https://zuul.opendev.org/t/openstack/build/d29ecf75d1eb49578609b5e74b75e766 : FAILURE in 33m 05s\n- refstack-client-devstack-train https://zuul.opendev.org/t/openstack/build/747a1e919c494ce3a4a20ad2b77cf018 : FAILURE in 41m 04s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/26076b0fbb7c42b9b560afe0f3bb7a84 : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":5},{"id":"827f2247d1c1036c6a35b750948875e05c810b7e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-07-25 09:02:37.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"2e8f70b07954625e3a77e6f85e2b92b10ee7db6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 09:02:50.000000000","message":"Patch Set 6: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":6},{"id":"1a6404da468f28e4251131eacdafa34658ad0adb","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-07-25 11:51:27.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d50b431a10bce024d9a089405f6bf71f545068f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 13:26:23.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2ac0df61a2924eaba1c828efee861f40\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/11bdb3368a0c46a5ac4e239bc1ba24df : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4ac5efb31e594033b645a7687107de72 : SUCCESS in 4m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c9001f575e834399aa21685a94b154e6 : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8d0880294e34985995bc83cc6705a49 : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0b4d16398fdd4900ace106d5e48b16f8 : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a10b045f74b74f1f86713b0ba58c3c63 : SUCCESS in 4m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a6bf61d991494d4ea82fb68866e153f7 : SUCCESS in 4m 32s (non-voting)\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/fd5a7c69581d459c9be26f59a925c7c0 : FAILURE in 1h 32m 49s\n- refstack-client-devstack-yoga https://zuul.opendev.org/t/openstack/build/ac626b5a86ab40dc85d636e89a9aab4e : FAILURE in 1h 32m 48s\n- refstack-client-devstack-xena https://zuul.opendev.org/t/openstack/build/03f391ebfabc46ceaabfb4b598b3d6b0 : FAILURE in 1h 28m 13s\n- refstack-client-devstack-wallaby https://zuul.opendev.org/t/openstack/build/a4ead36950db4c329ec84df53c24a2cd : FAILURE in 32m 09s\n- refstack-client-devstack-victoria https://zuul.opendev.org/t/openstack/build/034a07d3bca744a9848f0a3cabc03f95 : FAILURE in 1h 32m 14s\n- refstack-client-devstack-ussuri https://zuul.opendev.org/t/openstack/build/b79d6f5b9a4648ae9c3c4b2084276aa2 : FAILURE in 31m 12s\n- refstack-client-devstack-train https://zuul.opendev.org/t/openstack/build/23a1aaa0b2584622b39544cc9fd80930 : FAILURE in 48m 11s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/4a9afb93d8f6435ea7291f5f2b57e6e8 : SUCCESS in 3m 23s","accounts_in_message":[],"_revision_number":6},{"id":"ee1464389be3ecef2277cbe703798be71510ff05","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-08-15 12:44:29.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"ed1aa21c04bcffa67c5f0c096ce2f19be14811e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 12:44:42.000000000","message":"Patch Set 7: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":7},{"id":"f8e610dad8710fea4d227b44be86942c3a7b2fee","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-08-15 14:07:43.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"f3e720f7d58af2922576843965ce154f57a3b954","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 15:49:23.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8b13de5073114262b0ef031c7b45ed1b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9afa2eca93a7499480cb1160443c18d2 : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8b96dbeed8714a8f8d303f1668ec525b : SUCCESS in 3m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/73ffadc9cd234f579f757109978a4d2d : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e2d466faae74e15b554e512f0c231cb : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2169074c3d4c42e59bf4392e5a8e74a7 : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cc04305a2aca411ca915c7e3312481c7 : SUCCESS in 4m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ff613282c6b144caa3dc5b3c8e59facd : SUCCESS in 4m 13s (non-voting)\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/d736b6eae399443c911a523254f00d52 : FAILURE in 51m 08s\n- refstack-client-devstack-yoga https://zuul.opendev.org/t/openstack/build/27aa5c6ea9254a3999b474318397826b : FAILURE in 1h 37m 14s\n- refstack-client-devstack-xena https://zuul.opendev.org/t/openstack/build/2505a06508d542d9bbbaa7f4292f62b2 : FAILURE in 1h 30m 20s\n- refstack-client-devstack-wallaby https://zuul.opendev.org/t/openstack/build/46d049680aa84c2a8c07f30ccb47642c : FAILURE in 42m 56s\n- refstack-client-devstack-victoria https://zuul.opendev.org/t/openstack/build/dff75a0c5f1d4f5685cfdb59923574be : FAILURE in 1h 40m 06s\n- refstack-client-devstack-ussuri https://zuul.opendev.org/t/openstack/build/6bbe666c876140519cd210991cfe4d31 : FAILURE in 42m 40s\n- refstack-client-devstack-train https://zuul.opendev.org/t/openstack/build/7383cd07263f4673add0e62111e380d9 : FAILURE in 41m 57s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/7aa2cb49a5654cd698c7153743498926 : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":7},{"id":"701b07176d12f7e281bf6327f3a2dfdb5a0fb17f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-08-18 10:15:18.000000000","message":"Uploaded patch set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"80e2e47f1ced21bbe1c7251b72223d0af2114ce5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 11:27:53.000000000","message":"Patch Set 8: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":8},{"id":"2c34ca4a816fd0eaba446609d5f4a91402e6422f","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-08-18 11:28:30.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"6399543f8358e92a5feb5dd273aa098fbbad5804","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 12:54:25.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/83eae2d248084049ab8187c1673e8b53\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6257ee1597b44b1eb361efdc9ab44f77 : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cfc89cf0c03c4ec486fc0f745c847ea8 : SUCCESS in 4m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4faa78023b8f47088791a7b6cb7c72bc : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cfa8dbf124b46f2a03f37bf77cffa12 : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/079368c910d143b8bab9a48c20f5fe52 : SUCCESS in 3m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9f6e485ef46549c691c75a0204d71782 : SUCCESS in 3m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f03303fbbc2d43c3a776703e751d02dc : SUCCESS in 4m 01s (non-voting)\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/3344f420dead423bb51394c63bb3ceff : FAILURE in 1h 08m 25s\n- refstack-client-devstack-yoga https://zuul.opendev.org/t/openstack/build/5c3d71c03cb04fddb3b4035f353476ee : FAILURE in 1h 19m 09s\n- refstack-client-devstack-xena https://zuul.opendev.org/t/openstack/build/be9038322a494647be1560069ea33909 : FAILURE in 1h 05m 40s\n- refstack-client-devstack-wallaby https://zuul.opendev.org/t/openstack/build/91beb402cda44c0ab485039254f554c3 : FAILURE in 47m 24s\n- refstack-client-devstack-victoria https://zuul.opendev.org/t/openstack/build/ee7c0772f7df4607aa57d9aa36f49b8d : FAILURE in 38m 06s\n- refstack-client-devstack-ussuri https://zuul.opendev.org/t/openstack/build/e53099df47804f3295785688b36f13d0 : FAILURE in 42m 31s\n- refstack-client-devstack-train https://zuul.opendev.org/t/openstack/build/547e4ba8b6ec4591bd597e789af29b65 : FAILURE in 45m 52s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/0e872d4554084a9ea25066017ec54b50 : SUCCESS in 2m 42s","accounts_in_message":[],"_revision_number":8},{"id":"0ebdb74b6339f888090fb2ffcb01772c4cb85a42","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2022-08-19 12:13:16.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"55710301d46ddeb16a5246a4b061146e4f301d03","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-08-23 09:05:17.000000000","message":"Patch Set 8:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":8},{"id":"2959887fbe0d2c329e132acb64ca4d59113fc80f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-08-23 12:57:09.000000000","message":"Patch Set 9: Commit message was updated.","accounts_in_message":[],"_revision_number":9},{"id":"b1c9836fad9d66bdd74142bd1184e9d11458b8a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 14:37:52.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7fd9f10673514840945c95edb38e4308\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0da1c9ca7cfe40f5817f15217860f665 : SUCCESS in 3m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/460484d0e5c34e53af1cb6b3403fd289 : FAILURE in 2m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0e3ff04d7dbe4aaebbf979a2e964047a : FAILURE in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a50533ce153d443794b8585e5ba18d6d : SUCCESS in 3m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/15022e31eaeb4ee095fcf95b6707c0ac : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f7b56988f3954fdb98863a9e6a1eca89 : SUCCESS in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/179665d8a7b84e5b8cf99a329085b102 : SUCCESS in 4m 08s (non-voting)\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/7d1833cff2e8425dad76470189aa4e67 : SUCCESS in 1h 27m 54s\n- refstack-client-devstack-yoga https://zuul.opendev.org/t/openstack/build/fa193e52a58c435393137d4639b0b529 : SUCCESS in 1h 19m 05s\n- refstack-client-devstack-xena https://zuul.opendev.org/t/openstack/build/95becd04f36c46f7ad52808f12722085 : SUCCESS in 1h 25m 28s\n- refstack-client-devstack-wallaby https://zuul.opendev.org/t/openstack/build/203c5e5b0ae548b3957ac5852f7b3853 : FAILURE in 49m 36s\n- refstack-client-devstack-victoria https://zuul.opendev.org/t/openstack/build/ccf663a8856741428f222fe9d1182a81 : SUCCESS in 1h 32m 37s\n- refstack-client-devstack-ussuri https://zuul.opendev.org/t/openstack/build/1d9893d009ce4505ae3e921a2710f35f : FAILURE in 48m 56s\n- refstack-client-devstack-train https://zuul.opendev.org/t/openstack/build/c3eb689942004c24b0eb0858e3271641 : FAILURE in 43m 44s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/6b1b9496986e40fb9a637fc467b9d2a6 : SUCCESS in 3m 28s","accounts_in_message":[],"_revision_number":9},{"id":"1f3422c43d4a645660f98bf275e492a55f90550d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-08-23 14:38:21.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"a7b20b5c08359887c386782e5b95a41a33a3a1c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 16:04:56.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0cdfccddec98481b8fbef89c37141edd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/59f386f308ae4527b9a8718ffcb3eb11 : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d27105d1a6f4cc4ac4c0a914b96b012 : SUCCESS in 2m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f9b6bcbffb944ff941f1037040277fe : SUCCESS in 2m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dcee6f8e754c44d5baf6510b1393e9ff : SUCCESS in 2m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a814c85c4eac4915a5bf77579d6d3ba5 : SUCCESS in 2m 59s (non-voting)\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/c5c53f45ab4646fda377acdbf8154f8e : SUCCESS in 1h 08m 53s\n- refstack-client-devstack-yoga https://zuul.opendev.org/t/openstack/build/0eb61adee532420496004733a3660af6 : SUCCESS in 1h 10m 47s\n- refstack-client-devstack-xena https://zuul.opendev.org/t/openstack/build/04366187d56a4a189ad265def736945d : SUCCESS in 1h 05m 48s\n- refstack-client-devstack-wallaby https://zuul.opendev.org/t/openstack/build/bac50cfe3a1348238f14f411f5272a9b : SUCCESS in 1h 25m 05s\n- refstack-client-devstack-victoria https://zuul.opendev.org/t/openstack/build/2473d32201cf41de973884bc1ca564c4 : SUCCESS in 1h 12m 24s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/bd23bbb3790c4eaaaf2f62bfec99a9f7 : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":10},{"id":"47cc1126e6ae35ff077cbd902c46442f54979e99","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-08-23 19:16:15.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"41aaa9c2eb222c2802e0181604ebfd834d3c5598","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 19:17:04.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"125b1a71a306a7ff296eede6e899b8fed979531f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 20:54:59.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26edbccb20494cd18635d7b3d91f327b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/001eae693fff4bd1afcddcbcfaf8cb5a : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8b5b18bfea5349489e39d24823ed9b08 : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c934b36eda82490188ef0d93c87d6557 : SUCCESS in 4m 11s\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/27a96af428fa4d3ca7d15565e4ba8a3e : SUCCESS in 1h 19m 04s\n- refstack-client-devstack-yoga https://zuul.opendev.org/t/openstack/build/2bc186459c07467b830672766d91646b : SUCCESS in 1h 17m 52s\n- refstack-client-devstack-xena https://zuul.opendev.org/t/openstack/build/93b83b4cc0924d3cafaf8f62bd506e20 : SUCCESS in 1h 14m 15s\n- refstack-client-devstack-wallaby https://zuul.opendev.org/t/openstack/build/187d8b83018d492280bc554f14721434 : SUCCESS in 1h 33m 27s\n- refstack-client-devstack-victoria https://zuul.opendev.org/t/openstack/build/6f3082831f1a41038261a541c6fb9a47 : SUCCESS in 1h 19m 44s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/3662230b519249d0801050751a3501fc : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":10},{"id":"4629c582aa509cc41716afadc601eb71a88419e4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 20:54:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"a8e247ebe28ea0f9489d3fa513e0d6ee3604e161","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 20:56:58.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36b7ff542ed1416a8b6d1f750d521bb0\n\n- opendev-promote-docs https://zuul.opendev.org/t/openstack/build/bc2a72df513848c59e5125333af919df : SUCCESS in 43s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"df3ce1694721d23852392cc18a07bf7bede3eebe","revisions":{"11428f52a293fe5947121e23625a37cf628c54ff":{"kind":"REWORK","_number":1,"created":"2022-06-17 13:43:58.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/70/846370/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack-client","ref":"refs/changes/70/846370/1","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/70/846370/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/70/846370/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/70/846370/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack-client refs/changes/70/846370/1"}}},"commit":{"parents":[{"commit":"eb3c6d4957253cee8488e227a85863f2e58f6b77","subject":"Update setup.cfg and tox.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/eb3c6d4957253cee8488e227a85863f2e58f6b77"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2022-06-17 13:43:41.000000000","tz":120},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2022-06-17 13:43:41.000000000","tz":120},"subject":"DNM CI jobs health check","message":"DNM CI jobs health check\n\nChange-Id: Idf69f02ba844bdeea21ec744ec41ca78bbd1907d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/11428f52a293fe5947121e23625a37cf628c54ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/11428f52a293fe5947121e23625a37cf628c54ff"}]},"branch":"refs/heads/master"},"ebca780f8ce6f55f604998328b5f83b9e17fc64a":{"kind":"REWORK","_number":2,"created":"2022-06-20 13:49:58.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/70/846370/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack-client","ref":"refs/changes/70/846370/2","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/70/846370/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/70/846370/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/70/846370/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack-client refs/changes/70/846370/2"}}},"commit":{"parents":[{"commit":"eb3c6d4957253cee8488e227a85863f2e58f6b77","subject":"Update setup.cfg and tox.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/eb3c6d4957253cee8488e227a85863f2e58f6b77"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2022-06-17 13:43:41.000000000","tz":120},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2022-06-20 13:49:46.000000000","tz":120},"subject":"Update tempest version","message":"Update tempest version\n\nThe commit updates tempest version and removes Ussuri and Train jobs\nas those aren\u0027t supported by the community anymore.\n\nChange-Id: Idf69f02ba844bdeea21ec744ec41ca78bbd1907d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/ebca780f8ce6f55f604998328b5f83b9e17fc64a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/ebca780f8ce6f55f604998328b5f83b9e17fc64a"}]},"branch":"refs/heads/master"},"ca31ab06cb49b267e53d3b372c56eb0164b7e051":{"kind":"REWORK","_number":3,"created":"2022-07-20 15:35:40.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/70/846370/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack-client","ref":"refs/changes/70/846370/3","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/70/846370/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/70/846370/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/70/846370/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack-client refs/changes/70/846370/3"}}},"commit":{"parents":[{"commit":"eb3c6d4957253cee8488e227a85863f2e58f6b77","subject":"Update setup.cfg and tox.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/eb3c6d4957253cee8488e227a85863f2e58f6b77"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2022-06-17 13:43:41.000000000","tz":120},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2022-07-20 15:35:14.000000000","tz":120},"subject":"Update tempest version and drop U and T jobs","message":"Update tempest version and drop U and T jobs\n\nThe commit updates tempest version and removes Ussuri and Train jobs\nas those aren\u0027t supported by the community anymore.\n\nChange-Id: Idf69f02ba844bdeea21ec744ec41ca78bbd1907d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/ca31ab06cb49b267e53d3b372c56eb0164b7e051"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/ca31ab06cb49b267e53d3b372c56eb0164b7e051"}]},"branch":"refs/heads/master"},"f0068d7617eb0af397b856e96a46b150af22a576":{"kind":"REWORK","_number":4,"created":"2022-07-21 08:26:38.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/70/846370/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack-client","ref":"refs/changes/70/846370/4","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/70/846370/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/70/846370/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/70/846370/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack-client refs/changes/70/846370/4"}}},"commit":{"parents":[{"commit":"eb3c6d4957253cee8488e227a85863f2e58f6b77","subject":"Update setup.cfg and tox.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/eb3c6d4957253cee8488e227a85863f2e58f6b77"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2022-06-17 13:43:41.000000000","tz":120},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2022-07-21 08:23:34.000000000","tz":120},"subject":"Update tempest version and drop U and T jobs","message":"Update tempest version and drop U and T jobs\n\nThe commit updates tempest version and removes Ussuri and Train jobs\nas those aren\u0027t supported by the community anymore.\n\nChange-Id: Idf69f02ba844bdeea21ec744ec41ca78bbd1907d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/f0068d7617eb0af397b856e96a46b150af22a576"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/f0068d7617eb0af397b856e96a46b150af22a576"}]},"branch":"refs/heads/master"},"8b5411fe5c1af0ba3050b85d9f6e3a905bee8496":{"kind":"REWORK","_number":5,"created":"2022-07-21 12:22:01.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/70/846370/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack-client","ref":"refs/changes/70/846370/5","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/70/846370/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/70/846370/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/70/846370/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack-client refs/changes/70/846370/5"}}},"commit":{"parents":[{"commit":"eb3c6d4957253cee8488e227a85863f2e58f6b77","subject":"Update setup.cfg and tox.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/eb3c6d4957253cee8488e227a85863f2e58f6b77"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2022-06-17 13:43:41.000000000","tz":120},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2022-07-21 12:21:41.000000000","tz":120},"subject":"Update tempest version and drop U and T jobs","message":"Update tempest version and drop U and T jobs\n\nThe commit updates tempest version and removes Ussuri and Train jobs\nas those aren\u0027t supported by the community anymore.\n\nDepends-On: https://review.opendev.org/c/openinfra/ansible-role-refstack-client/+/846109\nChange-Id: Idf69f02ba844bdeea21ec744ec41ca78bbd1907d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/8b5411fe5c1af0ba3050b85d9f6e3a905bee8496"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/8b5411fe5c1af0ba3050b85d9f6e3a905bee8496"}]},"branch":"refs/heads/master"},"f60756a4b8513edea9cb19684465f27bda9215e6":{"kind":"REWORK","_number":6,"created":"2022-07-25 09:02:37.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/70/846370/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack-client","ref":"refs/changes/70/846370/6","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/70/846370/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/70/846370/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/70/846370/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack-client refs/changes/70/846370/6"}}},"commit":{"parents":[{"commit":"eb3c6d4957253cee8488e227a85863f2e58f6b77","subject":"Update setup.cfg and tox.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/eb3c6d4957253cee8488e227a85863f2e58f6b77"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2022-06-17 13:43:41.000000000","tz":120},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2022-07-25 09:01:35.000000000","tz":120},"subject":"Update tempest version and drop U and T jobs","message":"Update tempest version and drop U and T jobs\n\nThe commit updates tempest version and removes Ussuri and Train jobs\nas those aren\u0027t supported by the community anymore.\n\nDepends-On: https://review.opendev.org/c/openinfra/ansible-role-refstack-client/+/850762\nChange-Id: Idf69f02ba844bdeea21ec744ec41ca78bbd1907d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/f60756a4b8513edea9cb19684465f27bda9215e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/f60756a4b8513edea9cb19684465f27bda9215e6"}]},"branch":"refs/heads/master"},"23e29aa5e6cf29b193ea5fd4c8f4f8e1e75d2f80":{"kind":"REWORK","_number":7,"created":"2022-08-15 12:44:29.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/70/846370/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack-client","ref":"refs/changes/70/846370/7","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/70/846370/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/70/846370/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/70/846370/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack-client refs/changes/70/846370/7"}}},"commit":{"parents":[{"commit":"eb3c6d4957253cee8488e227a85863f2e58f6b77","subject":"Update setup.cfg and tox.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/eb3c6d4957253cee8488e227a85863f2e58f6b77"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2022-06-17 13:43:41.000000000","tz":120},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2022-08-15 12:32:02.000000000","tz":120},"subject":"Update tempest version and drop U and T jobs","message":"Update tempest version and drop U and T jobs\n\nThe commit updates tempest version and removes Ussuri and Train jobs\nas those aren\u0027t supported by the community anymore.\nThe patch also updates old openstack.org links.\n\nDepends-On: https://review.opendev.org/c/openinfra/ansible-role-refstack-client/+/850877\nChange-Id: Idf69f02ba844bdeea21ec744ec41ca78bbd1907d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/23e29aa5e6cf29b193ea5fd4c8f4f8e1e75d2f80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/23e29aa5e6cf29b193ea5fd4c8f4f8e1e75d2f80"}]},"branch":"refs/heads/master"},"b7653925e03c614c97d03d45d753f276eddbf368":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2022-08-18 10:15:18.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/70/846370/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack-client","ref":"refs/changes/70/846370/8","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/70/846370/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/70/846370/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/70/846370/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack-client refs/changes/70/846370/8"}}},"commit":{"parents":[{"commit":"eb3c6d4957253cee8488e227a85863f2e58f6b77","subject":"Update setup.cfg and tox.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/eb3c6d4957253cee8488e227a85863f2e58f6b77"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2022-06-17 13:43:41.000000000","tz":120},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2022-08-18 10:15:01.000000000","tz":120},"subject":"Update tempest version and drop U and T jobs","message":"Update tempest version and drop U and T jobs\n\nThe commit updates tempest version and removes Ussuri and Train jobs\nas those aren\u0027t supported by the community anymore.\nThe patch also updates old openstack.org links.\n\nDepends-On: https://review.opendev.org/c/openinfra/ansible-role-refstack-client/+/853619\nDepends-On: https://review.opendev.org/c/openinfra/python-tempestconf/+/853517\nChange-Id: Idf69f02ba844bdeea21ec744ec41ca78bbd1907d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/b7653925e03c614c97d03d45d753f276eddbf368"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/b7653925e03c614c97d03d45d753f276eddbf368"}]},"branch":"refs/heads/master"},"0867568f33100ee173835bef23891303415fc0c8":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2022-08-23 12:57:09.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/70/846370/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack-client","ref":"refs/changes/70/846370/9","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/70/846370/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/70/846370/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/70/846370/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack-client refs/changes/70/846370/9"}}},"commit":{"parents":[{"commit":"eb3c6d4957253cee8488e227a85863f2e58f6b77","subject":"Update setup.cfg and tox.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/eb3c6d4957253cee8488e227a85863f2e58f6b77"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2022-06-17 13:43:41.000000000","tz":120},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2022-08-23 12:57:09.000000000","tz":0},"subject":"Update tempest version and drop U and T jobs","message":"Update tempest version and drop U and T jobs\n\nThe commit updates tempest version and removes Ussuri and Train jobs\nas those aren\u0027t supported by the community anymore.\nThe patch also updates old openstack.org links.\n\nChange-Id: Idf69f02ba844bdeea21ec744ec41ca78bbd1907d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/0867568f33100ee173835bef23891303415fc0c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/0867568f33100ee173835bef23891303415fc0c8"}]},"branch":"refs/heads/master","description":"Edit commit message"},"df3ce1694721d23852392cc18a07bf7bede3eebe":{"kind":"REWORK","_number":10,"created":"2022-08-23 14:38:21.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/70/846370/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack-client","ref":"refs/changes/70/846370/10","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/70/846370/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/70/846370/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/70/846370/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack-client refs/changes/70/846370/10"}}},"commit":{"parents":[{"commit":"eb3c6d4957253cee8488e227a85863f2e58f6b77","subject":"Update setup.cfg and tox.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/eb3c6d4957253cee8488e227a85863f2e58f6b77"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2022-06-17 13:43:41.000000000","tz":120},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2022-08-23 14:18:31.000000000","tz":120},"subject":"Update tempest version and drop U and T jobs","message":"Update tempest version and drop U and T jobs\n\nThe commit updates tempest version and removes Ussuri and Train jobs\nas those aren\u0027t supported by the community anymore.\nThe patch also updates old openstack.org links.\n\nDepends-On: https://review.opendev.org/c/openinfra/ansible-role-refstack-client/+/854211\nChange-Id: Idf69f02ba844bdeea21ec744ec41ca78bbd1907d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/df3ce1694721d23852392cc18a07bf7bede3eebe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/df3ce1694721d23852392cc18a07bf7bede3eebe"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}}]}],"submit_requirements":[]}
