)]}'
{"id":"openinfra%2Frefstack-client~860578","triplet_id":"openinfra%2Frefstack-client~master~Ic4e32ca84f4c7bf63ae0a1deb11c1b4cb44166d8","project":"openinfra/refstack-client","branch":"master","topic":"fix/python-version-download","attention_set":{},"removed_from_attention_set":{"30674":{"account":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"last_update":"2022-10-11 09:51:02.000000000","reason":"Change was submitted"},"22873":{"account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"last_update":"2022-10-11 08:35:00.000000000","reason":"\u003cGERRIT_ACCOUNT_22873\u003e replied on the change","reason_account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}}},"hashtags":[],"change_id":"Ic4e32ca84f4c7bf63ae0a1deb11c1b4cb44166d8","subject":"Fix the default version of Python for \u0027-p 3\u0027 option","status":"MERGED","created":"2022-10-06 14:50:10.000000000","updated":"2022-10-11 09:52:18.000000000","submitted":"2022-10-11 09:51:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"860578-fix/python-version-download","meta_rev_id":"f26b86937a298a6ce2351167f764ef8f7586d03f","_number":860578,"virtual_id_number":860578,"owner":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-11 09:51:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":2,"date":"2022-10-11 08:35:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":1,"date":"2022-10-11 08:35:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-06 14:59:23.000000000","updated_by":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2022-10-06 16:26:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1b45be502a09163912e6da4438f45cfe63c16a12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2022-10-06 14:50:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cab177f4f037336de5c78313adafb046b77599b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 16:26:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0cf1001d46c246fb8c694f1e16d78951\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/798d280cb93b4674af53d06b99cf274f : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d89a595e9df4cc9837090a530bb9a1d : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/212084e125ea4a008ad6f1568d6d0f47 : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a4ec04c95e54443bb5348c356f2aabc5 : SUCCESS in 3m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ee8560b136c841a997816c946e3b0646 : SUCCESS in 3m 57s (non-voting)\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/f35658a6321744329298d8f45efb73a7 : FAILURE in 37m 35s\n- refstack-client-devstack-yoga https://zuul.opendev.org/t/openstack/build/c8922f4e2a91421ea5096351082dabdf : SUCCESS in 1h 17m 35s\n- refstack-client-devstack-xena https://zuul.opendev.org/t/openstack/build/739645226f884e74a6fe7773d900fddf : SUCCESS in 1h 11m 43s\n- refstack-client-devstack-wallaby https://zuul.opendev.org/t/openstack/build/19da1e3efafb4c999ed709d2f6a99f5b : SUCCESS in 1h 22m 09s\n- refstack-client-devstack-victoria https://zuul.opendev.org/t/openstack/build/1907d006f55b4d3db964e101779764fe : SUCCESS in 1h 29m 23s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/d14e1ab3f2ae4e99b22c701e02bf708d : SUCCESS in 2m 26s","accounts_in_message":[],"_revision_number":1},{"id":"7356f7d08415c1c884b248902ef193e72f14b61d","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2022-10-07 09:02:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4f393bbcaf5d875362f41a324199f7d344176f06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-07 10:29:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c2adea196f74262b467456791a8ce66\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/690cf4d91de44f81925e5ed06199a1b1 : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02386ad4b21f4c18b8367b5d4482374a : SUCCESS in 3m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cb4c817d87b24ad0a30b0c0f6975e133 : SUCCESS in 3m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7a606f6c1ed74c0489a9c60cbc5eedbe : SUCCESS in 3m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6c087f3af97d4e48a989a282985f8aae : SUCCESS in 3m 37s (non-voting)\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/8dd747c1ff654b018ef5d5e0ca863a93 : SUCCESS in 1h 21m 15s\n- refstack-client-devstack-yoga https://zuul.opendev.org/t/openstack/build/707e365e8e9f42939e488e2cce77154c : SUCCESS in 1h 20m 45s\n- refstack-client-devstack-xena https://zuul.opendev.org/t/openstack/build/9ab3617e2d2f4c4282a7d41d33ce3375 : SUCCESS in 1h 12m 00s\n- refstack-client-devstack-wallaby https://zuul.opendev.org/t/openstack/build/184336acde934a1da61a625fef852291 : SUCCESS in 1h 19m 23s\n- refstack-client-devstack-victoria https://zuul.opendev.org/t/openstack/build/ae3d24a00e2845a3b2ab9cf09fba71ef : SUCCESS in 1h 17m 12s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/6ca92c7763014e7595fa7e47dcb3bafb : SUCCESS in 3m 14s","accounts_in_message":[],"_revision_number":1},{"id":"9fa26f563a9c3ff0e551d03bb73f0c6c2c03efd2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2022-10-07 11:09:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d2507d335d476d8ac3fa77d3feccaf075e9c6189","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-07 12:11:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e9eccde161c44d095c80ff638dc27ba\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d285e6137f42469e92f7ba0a7b37dd41 : SUCCESS in 2m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7481ac1f5c1d49c9a460ce02aae29b3c : SUCCESS in 2m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7fb2175068794cfead01ce9f24460d3a : SUCCESS in 2m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a551e8f0b571408da972fefb31a57414 : SUCCESS in 2m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/01229211933a4352b6784ccf5911d2c2 : SUCCESS in 2m 48s (non-voting)\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/9bc0ffb7fcc24ff5829ae455d4ebe07f : SUCCESS in 1h 00m 43s\n- refstack-client-devstack-yoga https://zuul.opendev.org/t/openstack/build/cdeceae5ddae41fd957ab7396f358163 : SUCCESS in 59m 44s\n- refstack-client-devstack-xena https://zuul.opendev.org/t/openstack/build/f4ab14e8bea040b79c04375fa7d248d6 : SUCCESS in 56m 39s\n- refstack-client-devstack-wallaby https://zuul.opendev.org/t/openstack/build/33db8fff11474dfea5ebedf1c17638d6 : SUCCESS in 1h 00m 37s\n- refstack-client-devstack-victoria https://zuul.opendev.org/t/openstack/build/740d934df1ca41f8b51db500c93170f2 : SUCCESS in 1h 01m 03s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/07350cc449dc4ff181dfde35d8fc80bb : SUCCESS in 2m 45s","accounts_in_message":[],"_revision_number":2},{"id":"c69b09a80502669b04b6013705faf5c1910a5b59","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-10-11 08:35:00.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e733fc548440857b4e4881f81bc4d1b6704a1d16","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 08:35:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"51a8c1fdc832907668e61b908acfc963cd3232f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 09:51:02.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2f8345cb74e48fe8189068eb7f2297b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d86995d98a234375ade7cb586b4e4ede : SUCCESS in 2m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/492a848ccbe54c7d90722964ffc510ac : SUCCESS in 3m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c7564c71da624083b7371723e07ead9d : SUCCESS in 3m 17s\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/83bbd6b2b24d465998cad460a7ea61d0 : SUCCESS in 54m 13s\n- refstack-client-devstack-yoga https://zuul.opendev.org/t/openstack/build/0948d52c915d471ab8fc68430a6cad67 : SUCCESS in 1h 08m 26s\n- refstack-client-devstack-xena https://zuul.opendev.org/t/openstack/build/9e4cab3a1ab04aaabc397b55000488f2 : SUCCESS in 1h 06m 14s\n- refstack-client-devstack-wallaby https://zuul.opendev.org/t/openstack/build/5c8d9638552342dc84a1080f6b7277d0 : SUCCESS in 1h 13m 59s\n- refstack-client-devstack-victoria https://zuul.opendev.org/t/openstack/build/9ceefe58b3aa482ebe224533a5644b16 : SUCCESS in 1h 12m 05s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/e741d90c32d34f81baca539484a61900 : SUCCESS in 2m 47s","accounts_in_message":[],"_revision_number":2},{"id":"96e8f39938dd9053cf2f1307cfb97d6e939a8707","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 09:51:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f26b86937a298a6ce2351167f764ef8f7586d03f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 09:52:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a41b599af55540f5adf0e28dffe6aab1\n\n- opendev-promote-docs https://zuul.opendev.org/t/openstack/build/e97443bfdd584ed1b3125d297261fa53 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"23d9f5d7dfb80c42f8154ff8dd1708f570595abc","revisions":{"2179463ef4bd497069a06bf4dd2e50023ecff0e4":{"kind":"REWORK","_number":1,"created":"2022-10-06 14:50:10.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/78/860578/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack-client","ref":"refs/changes/78/860578/1","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/78/860578/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/78/860578/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/78/860578/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack-client refs/changes/78/860578/1"}}},"commit":{"parents":[{"commit":"df3ce1694721d23852392cc18a07bf7bede3eebe","subject":"Update tempest version and drop U and T jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/df3ce1694721d23852392cc18a07bf7bede3eebe"}]}],"author":{"name":"Lukáš Piwowarski","email":"lpiwowar@redhat.com","date":"2022-10-06 12:24:25.000000000","tz":120},"committer":{"name":"Lukáš Piwowarski","email":"lpiwowar@redhat.com","date":"2022-10-06 14:50:08.000000000","tz":120},"subject":"Fix the default version of Python for \u0027-p 3\u0027 option","message":"Fix the default version of Python for \u0027-p 3\u0027 option\n\nWhen \u0027-p 3\u0027 option is used, the setup_env script checks whether\nthere is \u0027python3\u0027 installed on the system. If it is, then python\n3.8.10 is not installed by the script and whatever version that is\npointed to by the \u0027python3\u0027 command is used.\n\nThis is an issue as the setup_env script help message states that\npython 3.8.10 is used for the \u0027-p 3\u0027 option.\n\nThis patch fixes the issue by ensuring that PY_VERSION \u003d\u003d \"3.8.10\"\nwhen \u0027-p 3\u0027 is utilized and by checking the exact version of the\ninstalled python on the system.\n\nChange-Id: Ic4e32ca84f4c7bf63ae0a1deb11c1b4cb44166d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/2179463ef4bd497069a06bf4dd2e50023ecff0e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/2179463ef4bd497069a06bf4dd2e50023ecff0e4"}]},"branch":"refs/heads/master"},"23d9f5d7dfb80c42f8154ff8dd1708f570595abc":{"kind":"REWORK","_number":2,"created":"2022-10-07 11:09:00.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/78/860578/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack-client","ref":"refs/changes/78/860578/2","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/78/860578/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/78/860578/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/78/860578/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack-client refs/changes/78/860578/2"}}},"commit":{"parents":[{"commit":"df3ce1694721d23852392cc18a07bf7bede3eebe","subject":"Update tempest version and drop U and T jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/df3ce1694721d23852392cc18a07bf7bede3eebe"}]}],"author":{"name":"Lukáš Piwowarski","email":"lpiwowar@redhat.com","date":"2022-10-06 12:24:25.000000000","tz":120},"committer":{"name":"Lukáš Piwowarski","email":"lpiwowar@redhat.com","date":"2022-10-07 11:08:41.000000000","tz":120},"subject":"Fix the default version of Python for \u0027-p 3\u0027 option","message":"Fix the default version of Python for \u0027-p 3\u0027 option\n\nWhen \u0027-p 3\u0027 option is used, the setup_env script checks whether\nthere is \u0027python3\u0027 installed on the system. If it is, then python\n3.8.10 is not installed by the script and whatever version that is\npointed to by the \u0027python3\u0027 command is used.\n\nThis is an issue as the setup_env script help message states that\npython 3.8.10 is used for the \u0027-p 3\u0027 option.\n\nThis patch fixes the issue by ensuring that PY_VERSION \u003d\u003d \"3.8.10\"\nwhen \u0027-p 3\u0027 is utilized and by checking the exact version of the\ninstalled python on the system.\n\nChange-Id: Ic4e32ca84f4c7bf63ae0a1deb11c1b4cb44166d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/23d9f5d7dfb80c42f8154ff8dd1708f570595abc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/23d9f5d7dfb80c42f8154ff8dd1708f570595abc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}}]}],"submit_requirements":[]}
