)]}'
{"id":"openinfra%2Frefstack-client~866311","triplet_id":"openinfra%2Frefstack-client~master~Ib1a3147cbf1268d9062f091f55a4c32eb8d2f535","project":"openinfra/refstack-client","branch":"master","topic":"fix/issues_link","attention_set":{},"removed_from_attention_set":{"30674":{"account":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"last_update":"2022-12-13 19:34:24.000000000","reason":"Change was submitted"},"22873":{"account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"last_update":"2022-12-13 19:20:55.000000000","reason":"\u003cGERRIT_ACCOUNT_22873\u003e replied on the change","reason_account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}}},"hashtags":[],"change_id":"Ib1a3147cbf1268d9062f091f55a4c32eb8d2f535","subject":"Update link to the issues board","status":"MERGED","created":"2022-12-01 09:51:35.000000000","updated":"2022-12-13 19:35:27.000000000","submitted":"2022-12-13 19:34:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"866311-fix/issues_link","meta_rev_id":"f888caf79cae232d2ebae041d0e61153efdecc3c","_number":866311,"virtual_id_number":866311,"owner":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-13 19:34:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"all":[{"value":2,"date":"2022-12-01 18:15:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":2,"date":"2022-12-13 19:20:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":1,"date":"2022-12-13 19:20:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-01 10:00:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-01 10:02:21.000000000","updated_by":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2022-12-01 18:15:13.000000000","updated_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"}],"messages":[{"id":"6476bad99a6630b0279fdd56f75356088f814374","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2022-12-01 09:51:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5415993c0df915e9d651f47c9f6d4fbeb3746349","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 10:00:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9604a35bd0e4b85bd20aca5cccaaa7f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5166507ea748483b9bd475e5fd5606e6 : SUCCESS in 2m 50s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/c8328d789ffd497896f8e9951365ffc8 : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":1},{"id":"4b9f447036b3daf9b3ebc6e356b4aa6c25a524fc","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2022-12-01 18:15:13.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"aad3b8256ca54d8868575cd51d26d72c0d52a3a6","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-12-13 19:20:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cda44238d7fad2064dd724ee79b17e1fdba672ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-13 19:21:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"77f73521a95962c0409fb6e689b22e1dc0403cd0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-13 19:34:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ff5bd27a2cc4e0fab426dbd400904ce\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ff0ca3d69b046ce91e420d0f57cecca : SUCCESS in 3m 00s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/2827a1afa1f64187b2742e39b875ffdd : SUCCESS in 2m 41s","accounts_in_message":[],"_revision_number":1},{"id":"2f08fdfc50ecd693ea98488e1db585d345b551b3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-13 19:34:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f888caf79cae232d2ebae041d0e61153efdecc3c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-13 19:35:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0a2a595a56f4704bc4faf268933e7fd\n\n- opendev-promote-docs https://zuul.opendev.org/t/openstack/build/08957ee944f1428488dba5e571f48d36 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ee6161ddf7557ef4356031f6f59bf4b193e4421b","revisions":{"ee6161ddf7557ef4356031f6f59bf4b193e4421b":{"kind":"REWORK","_number":1,"created":"2022-12-01 09:51:35.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/11/866311/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack-client","ref":"refs/changes/11/866311/1","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/11/866311/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/11/866311/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/11/866311/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack-client refs/changes/11/866311/1"}}},"commit":{"parents":[{"commit":"340344526bff3dc2b298eb55f99ce5eb7bff1579","subject":"Fix parsing of tenant/project_name values","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/340344526bff3dc2b298eb55f99ce5eb7bff1579"}]}],"author":{"name":"Lukáš Piwowarski","email":"lpiwowar@redhat.com","date":"2022-12-01 09:42:21.000000000","tz":60},"committer":{"name":"Lukáš Piwowarski","email":"lpiwowar@redhat.com","date":"2022-12-01 09:45:50.000000000","tz":60},"subject":"Update link to the issues board","message":"Update link to the issues board\n\nThe link to the issues board should be updated as we use Taiga instead\nof StoryBoard to track refstack related issues [1].\n\n[1] https://tree.taiga.io/project/openstack-interop-working-group/issues\n\nChange-Id: Ib1a3147cbf1268d9062f091f55a4c32eb8d2f535\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/ee6161ddf7557ef4356031f6f59bf4b193e4421b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/ee6161ddf7557ef4356031f6f59bf4b193e4421b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}}]}],"submit_requirements":[]}
