)]}'
{"id":"openinfra%2Frefstack-client~878174","triplet_id":"openinfra%2Frefstack-client~master~I4c668a6a229a9afa941a1d7eaa6b7d8b6f08021f","project":"openinfra/refstack-client","branch":"master","topic":"experiment/role-project","attention_set":{},"removed_from_attention_set":{"30674":{"account":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"last_update":"2023-03-29 09:04:56.000000000","reason":"\u003cGERRIT_ACCOUNT_30674\u003e replied on the change","reason_account":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"}}},"hashtags":[],"change_id":"I4c668a6a229a9afa941a1d7eaa6b7d8b6f08021f","subject":"DNM: Experiment","status":"ABANDONED","created":"2023-03-21 21:12:18.000000000","updated":"2023-05-15 15:53:18.000000000","total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"841ed74a7ba0b3abd8b884cdd2d22943fe4027c7","_number":878174,"virtual_id_number":878174,"owner":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-04-05 08:55:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-21 22:46:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b293f8667920b9f71e0daa6f97def97333e49c4f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-03-21 21:12:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a61304edd111b78070c4e1ee7a991d3e9b1f8746","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-21 22:46:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dcfa55abac7645898c25a8ca9b2d9965\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e5b8dd8d1264ea68910b85c14839c46 : SUCCESS in 2m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a28b076289c4822a4d76e3333beacf2 : SUCCESS in 2m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/058581511c814d88862c2923bea6175d : SUCCESS in 2m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0c7740926324472ba4dff5dff05905c3 : SUCCESS in 2m 54s\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/0b537bd9d04946458b7793c77038a8ee : FAILURE in 1h 09m 58s\n- refstack-client-devstack-zed https://zuul.opendev.org/t/openstack/build/dffb6ded87164cf38e6c35f897f7ab9f : FAILURE in 1h 02m 12s\n- refstack-client-devstack-yoga https://zuul.opendev.org/t/openstack/build/8d326def592c433fa3d707b89e8acb13 : FAILURE in 1h 10m 14s\n- refstack-client-devstack-xena https://zuul.opendev.org/t/openstack/build/d3045a48d2cd45c88b158dedf03ac7d1 : FAILURE in 1h 19m 55s\n- refstack-client-devstack-wallaby https://zuul.opendev.org/t/openstack/build/411f696d52d148be956eb8da4f129174 : FAILURE in 1h 11m 03s\n- refstack-client-devstack-victoria https://zuul.opendev.org/t/openstack/build/c8c32e3d5c3b479a80b45b569b11d298 : FAILURE in 1h 26m 32s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/e4d2904d830a434f92b9697c72031dbd : SUCCESS in 2m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/137fa73317e341cdae5c5b82d018479a : SUCCESS in 3m 00s","accounts_in_message":[],"_revision_number":1},{"id":"f5888c6b0851f8dfba41cdc953eb961719bfc140","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-03-22 08:15:00.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"b8e8fc02ae55dd0725938cbf455e5620c5ddfcfa","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-03-22 08:23:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"af0f562f2172087652134414c755c530cedb2be9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-22 09:46:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6cac487d88e74cbdbeb9f442381279be\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3228bc17a29b4ec597c36643bb6ffcc4 : SUCCESS in 3m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acf4bbb53a1d4f06aecacff186cf89ff : SUCCESS in 3m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/996722b3f8fa4062a61098e6978b7052 : SUCCESS in 3m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/097251e7480242e58638ac4c17ae1b5d : SUCCESS in 3m 01s\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/8c63980219084eeeb822c442f212413e : FAILURE in 1h 21m 22s\n- refstack-client-devstack-zed https://zuul.opendev.org/t/openstack/build/30d37996a09a40c084776a537fcb4f62 : FAILURE in 1h 14m 13s\n- refstack-client-devstack-yoga https://zuul.opendev.org/t/openstack/build/a36ac5668907455fbaff6ca7ae202ee2 : FAILURE in 1h 06m 21s\n- refstack-client-devstack-xena https://zuul.opendev.org/t/openstack/build/236fd993743c47ed9c4225bc15f6f1ca : FAILURE in 1h 11m 22s\n- refstack-client-devstack-wallaby https://zuul.opendev.org/t/openstack/build/41cf36a6f71d490683ab199c5bf607e1 : FAILURE in 1h 16m 08s\n- refstack-client-devstack-victoria https://zuul.opendev.org/t/openstack/build/9d1f6580e0c74923831b18230853e23f : FAILURE in 1h 09m 35s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/77ecda563118483d8a9191c3ee995d16 : SUCCESS in 3m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7a50db3a9c234587b45545b7d67ac529 : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":2},{"id":"752643a8d9b55498859c4e039d3914362ea3b126","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-03-22 09:50:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d2e1e02c4859d37d5dc39da95c634ab893a2b441","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-22 11:22:43.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cfcd085dc647473e82c4f15557a7371f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a72c4ed69ce6433f8e32b97351620106 : SUCCESS in 3m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf99a34d4f3449abb828c67558576f3b : SUCCESS in 2m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c5734ccda6d04e5782277272600cd51f : SUCCESS in 3m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/caff6197c2d943c3ad4ac9176bcc4ca3 : SUCCESS in 3m 13s\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/8e10d0f69de740a9a468e1cbc34e8b7a : FAILURE in 1h 25m 49s\n- refstack-client-devstack-zed https://zuul.opendev.org/t/openstack/build/708b987981b94dffa985f4ba975ffe01 : FAILURE in 1h 30m 43s\n- refstack-client-devstack-yoga https://zuul.opendev.org/t/openstack/build/614f872f121a4123abc2ce79848d6871 : FAILURE in 1h 11m 45s\n- refstack-client-devstack-xena https://zuul.opendev.org/t/openstack/build/29121dabae064ea3be578f95f94a810a : FAILURE in 1h 06m 22s\n- refstack-client-devstack-wallaby https://zuul.opendev.org/t/openstack/build/7cc0e3415f08430b818132ae0669da9f : FAILURE in 1h 09m 50s\n- refstack-client-devstack-victoria https://zuul.opendev.org/t/openstack/build/4b4b043b100f4561897cf978bc2721c5 : FAILURE in 1h 09m 31s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/a20a5d90ba094f3e8d5de6bbde5ac24a : SUCCESS in 2m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c6572f7dfa5e4fb689010c0ce7445666 : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":2},{"id":"f88565cc5f5a5a17cb2132025dcee0dd2e4d1aa7","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-03-28 16:24:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f62f924cb501a45bc9eb047f5389920c3853c7e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 17:25:24.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e61e67aeefa640a887281596156f1be2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1a9987b16a3149358fd7f66707d0f04a : SUCCESS in 3m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df0fc75971094b4d860298769418e774 : SUCCESS in 3m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a0ee42b1ce2040169798728bf9a3b315 : SUCCESS in 2m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/398a0050e5f249c8a62d87dcac3a147a : SUCCESS in 5m 48s\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/33a0f5193b624a05875c6c52f07b4d01 : FAILURE in 33m 44s\n- refstack-client-devstack-zed https://zuul.opendev.org/t/openstack/build/2c0ac7ac72c845f89b52223f69c454c1 : FAILURE in 45m 31s\n- refstack-client-devstack-yoga https://zuul.opendev.org/t/openstack/build/f68af4e546c7427abdccb3e985a64611 : FAILURE in 41m 08s\n- refstack-client-devstack-xena https://zuul.opendev.org/t/openstack/build/a1e74271e62b427690ec8ea2bac7a4d7 : FAILURE in 31m 04s\n- refstack-client-devstack-wallaby https://zuul.opendev.org/t/openstack/build/d01dbf21a8434508824fc1bb7b79ba29 : FAILURE in 53m 16s\n- refstack-client-devstack-victoria https://zuul.opendev.org/t/openstack/build/f1e35ebc7d064b87bb7442db98a4dafd : FAILURE in 32m 57s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/5d997ac5036742cbb93bd1a14da2139b : SUCCESS in 2m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b8183a6cf020452991c48ed15a17416a : SUCCESS in 3m 00s","accounts_in_message":[],"_revision_number":2},{"id":"91d669cafa3eaaa510319d13ce25de0b18e04fc9","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-03-28 17:42:27.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4b6548353c2477c94cf197055273d19dc9531331","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 19:22:27.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/530de5dc3fcc4badbfd16c58697b089c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/16f092c1dec64c03934af20a2d2d8ad9 : SUCCESS in 3m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/480b026c808e4bdcbb37f3d77529e101 : SUCCESS in 2m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f4012b75e8f740b6962cca7409db48a3 : SUCCESS in 2m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eb4fa380e9b54238b9cc990846661a9d : SUCCESS in 2m 55s\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/808798e8de9343f4b60f13f5eb6be1b4 : FAILURE in 1h 26m 52s\n- refstack-client-devstack-zed https://zuul.opendev.org/t/openstack/build/2baf6b76f0684964ab290942c1a0aa0a : SUCCESS in 1h 16m 30s\n- refstack-client-devstack-yoga https://zuul.opendev.org/t/openstack/build/4e98a7f405c84f72804d7e036bd6ad94 : SUCCESS in 1h 21m 48s\n- refstack-client-devstack-xena https://zuul.opendev.org/t/openstack/build/fdde80457a0c4237a0b9c3744d378b93 : SUCCESS in 54m 07s\n- refstack-client-devstack-wallaby https://zuul.opendev.org/t/openstack/build/4e2874ae5ff14f27ab2424d277775020 : SUCCESS in 1h 33m 43s\n- refstack-client-devstack-victoria https://zuul.opendev.org/t/openstack/build/32e9ba1fa3c24517890e188b48968e31 : SUCCESS in 1h 17m 39s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/cc2df38bbda74d37aa65c06723945dbe : SUCCESS in 2m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4e0e0a5353114b19878136c10f064cc3 : SUCCESS in 2m 46s","accounts_in_message":[],"_revision_number":2},{"id":"1f2acab59dab1c3ff0d5f41d9ba115a119f0b0f9","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-03-29 09:04:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"67456e8565bb386a62b770b88a3aa0e1607cf3dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-29 10:31:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f43a4f87ac94b64a8b84c234583b225\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4dd9dbdad6dd47c6b006668a8edc127e : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b1059ae68974cc1a63f17c45061ace3 : SUCCESS in 3m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0d5206417dad494db7eb89775d553676 : SUCCESS in 3m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/757306abab154dcbaf19aa97c3d8e3e3 : SUCCESS in 3m 26s\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/0a18de8f61f1401cbe6d6a071f9c5434 : SUCCESS in 1h 20m 42s\n- refstack-client-devstack-zed https://zuul.opendev.org/t/openstack/build/1efa0cf6cc41407d82d26c68e3632998 : SUCCESS in 1h 13m 56s\n- refstack-client-devstack-yoga https://zuul.opendev.org/t/openstack/build/27ecb32b32684313be9fcf0675f58f87 : SUCCESS in 1h 21m 40s\n- refstack-client-devstack-xena https://zuul.opendev.org/t/openstack/build/cc52b14527f74143a1fa7e211ae607b3 : SUCCESS in 1h 19m 17s\n- refstack-client-devstack-wallaby https://zuul.opendev.org/t/openstack/build/464ab4979cea47ffaef63e1fdce0d55e : SUCCESS in 1h 22m 14s\n- refstack-client-devstack-victoria https://zuul.opendev.org/t/openstack/build/c7a578874328424f85c06d608b282b24 : SUCCESS in 1h 13m 20s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/7daf0c10ad1c4846957549c4c040c9a9 : SUCCESS in 3m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d24b3ed33c244ea6bf6a4617ad5a1e59 : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":2},{"id":"dd050b07b90bd20144fb7bfc12369d99a02c08c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-04-05 07:19:03.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d74c09cda4e7278ca54af5b8f4811cb712b4abef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-05 08:55:02.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d85b30f95794438cbdd2523b04e235f8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/40f6900168c04eb0b5c4aa899025ddd5 : SUCCESS in 3m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44a241c925384bee813f679cd050df81 : SUCCESS in 2m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c41e898d7a0e44c98011660ac2259e39 : SUCCESS in 2m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cc0b725ee95d43558797ad8901259d32 : SUCCESS in 2m 48s\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/1c1548d0c8f34230b6542ed6187d55c9 : SUCCESS in 1h 24m 17s\n- refstack-client-devstack-zed https://zuul.opendev.org/t/openstack/build/5c2245b3d2e64539bc217ed7ffedbe1b : SUCCESS in 1h 13m 32s\n- refstack-client-devstack-yoga https://zuul.opendev.org/t/openstack/build/48cd268b8a914adaa14104f741b8912a : SUCCESS in 1h 23m 02s\n- refstack-client-devstack-xena https://zuul.opendev.org/t/openstack/build/561876d9c8b6443b8d02d2141a07a84d : SUCCESS in 1h 17m 11s\n- refstack-client-devstack-wallaby https://zuul.opendev.org/t/openstack/build/2ab3b1dfb650492f8518d06576665ef0 : SUCCESS in 1h 14m 09s\n- refstack-client-devstack-victoria https://zuul.opendev.org/t/openstack/build/3fa3a7cca4d5446087ab3c1d38fae828 : SUCCESS in 1h 27m 56s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/adec0d7b01b640e19d9850d646d0e295 : SUCCESS in 2m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dc49b81679d74261aa194b066dd96c0c : SUCCESS in 2m 19s","accounts_in_message":[],"_revision_number":3},{"id":"841ed74a7ba0b3abd8b884cdd2d22943fe4027c7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-05-15 15:53:18.000000000","message":"Abandoned\n\nNo longer required.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f9ea8ffb57135dc67e86c88617e8cadc7018c441","revisions":{"796efb50c30796668d01d048ee867ba3340ecb64":{"kind":"REWORK","_number":1,"created":"2023-03-21 21:12:18.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/74/878174/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack-client","ref":"refs/changes/74/878174/1","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/74/878174/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/74/878174/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/74/878174/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack-client refs/changes/74/878174/1"}}},"commit":{"parents":[{"commit":"43b6fd208bbfdce16362a9b461316a42123812f5","subject":"Merge \"Stop to use the __future__ module.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/43b6fd208bbfdce16362a9b461316a42123812f5"}]}],"author":{"name":"Lukáš Piwowarski","email":"lpiwowar@redhat.com","date":"2023-03-21 21:02:14.000000000","tz":60},"committer":{"name":"Lukáš Piwowarski","email":"lpiwowar@redhat.com","date":"2023-03-21 21:03:48.000000000","tz":60},"subject":"DNM: Experiment","message":"DNM: Experiment\n\nChange-Id: I4c668a6a229a9afa941a1d7eaa6b7d8b6f08021f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/796efb50c30796668d01d048ee867ba3340ecb64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/796efb50c30796668d01d048ee867ba3340ecb64"}]},"branch":"refs/heads/master"},"9764a01f63dd28bf9aebd9376679fc641d3797ca":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-03-22 08:15:00.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/74/878174/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack-client","ref":"refs/changes/74/878174/2","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/74/878174/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/74/878174/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/74/878174/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack-client refs/changes/74/878174/2"}}},"commit":{"parents":[{"commit":"43b6fd208bbfdce16362a9b461316a42123812f5","subject":"Merge \"Stop to use the __future__ module.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/43b6fd208bbfdce16362a9b461316a42123812f5"}]}],"author":{"name":"Lukáš Piwowarski","email":"lpiwowar@redhat.com","date":"2023-03-21 21:02:14.000000000","tz":60},"committer":{"name":"Lukáš Piwowarski","email":"lpiwowar@redhat.com","date":"2023-03-22 08:14:43.000000000","tz":60},"subject":"DNM: Experiment","message":"DNM: Experiment\n\nDepends-On: I2a3cdca3b2b26946d6e5f910527ad1c7ada9b202\nChange-Id: I4c668a6a229a9afa941a1d7eaa6b7d8b6f08021f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/9764a01f63dd28bf9aebd9376679fc641d3797ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/9764a01f63dd28bf9aebd9376679fc641d3797ca"}]},"branch":"refs/heads/master"},"f9ea8ffb57135dc67e86c88617e8cadc7018c441":{"kind":"REWORK","_number":3,"created":"2023-04-05 07:19:03.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/74/878174/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack-client","ref":"refs/changes/74/878174/3","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/74/878174/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/74/878174/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/74/878174/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack-client refs/changes/74/878174/3"}}},"commit":{"parents":[{"commit":"43b6fd208bbfdce16362a9b461316a42123812f5","subject":"Merge \"Stop to use the __future__ module.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/43b6fd208bbfdce16362a9b461316a42123812f5"}]}],"author":{"name":"Lukáš Piwowarski","email":"lpiwowar@redhat.com","date":"2023-03-21 21:02:14.000000000","tz":60},"committer":{"name":"Lukáš Piwowarski","email":"lpiwowar@redhat.com","date":"2023-04-05 07:18:53.000000000","tz":120},"subject":"DNM: Experiment","message":"DNM: Experiment\n\nDepends-On: I2a3cdca3b2b26946d6e5f910527ad1c7ada9b202\nChange-Id: I4c668a6a229a9afa941a1d7eaa6b7d8b6f08021f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/f9ea8ffb57135dc67e86c88617e8cadc7018c441"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/f9ea8ffb57135dc67e86c88617e8cadc7018c441"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
