)]}'
{"id":"openinfra%2Frefstack-client~880829","triplet_id":"openinfra%2Frefstack-client~master~Ifbdbe55e2309c4855cc5f069487739aa91ee2a89","project":"openinfra/refstack-client","branch":"master","topic":"switch-to-bobcat-testing-runtime","attention_set":{},"removed_from_attention_set":{"35229":{"account":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"last_update":"2023-05-23 17:09:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifbdbe55e2309c4855cc5f069487739aa91ee2a89","subject":"Update jobs for Bobcat development cycle","status":"MERGED","created":"2023-04-19 14:16:38.000000000","updated":"2023-05-23 17:11:04.000000000","submitted":"2023-05-23 17:09:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880829-switch-to-bobcat-testing-runtime","meta_rev_id":"6240d2b4c8cbde19ba9eaaa28a7dae4e896326ff","_number":880829,"virtual_id_number":880829,"owner":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-23 17:09:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":2,"date":"2023-05-22 19:25:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":1,"date":"2023-05-22 19:25:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-19 14:17:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-21 13:28:52.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"CC"},{"updated":"2023-05-22 19:25:18.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"}],"messages":[{"id":"5a483bb088aba35db2a2e14d91451c5156636155","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"date":"2023-04-19 14:16:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8ee1c4d4ba1e5e94e32a17a4d30d7dfdaba5a125","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 14:17:08.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openinfra/refstack-client on branch master.  The error was:\n\n  The project template \"openstack-python3-bobcat-jobs\" was not found.\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - openstack-cover-jobs\n        - openstack-python3-bobcat-jobs\n      check:\n        jobs:\n          - refstack-client-devstack-master:\n              # Define a list of irrelevant files to use everywhere else\n              irrelevant-files: \u0026refstack-client-irrelevant-files\n                - ^.*\\.rst$\n                - ^doc/.*$\n                - ^.gitignore$\n                - ^.gitreview$\n          - refstack-client-devstack-antelope:\n              irrelevant-files: *refstack-client-irrelevant-files\n          - refstack-client-devstack-zed:\n              irrelevant-files: *refstack-client-irrelevant-files\n          - refstack-client-devstack-yoga:\n              irrelevant-files: *refstack-client-irrelevant-files\n          - opendev-tox-docs\n          - openstack-tox-py39\n      gate:\n        jobs:\n          - refstack-client-devstack-master:\n              irrelevant-files: *refstack-client-irrelevant-files\n          - refstack-client-devstack-antelope:\n              irrelevant-files: *refstack-client-irrelevant-files\n          - refstack-client-devstack-zed:\n              irrelevant-files: *refstack-client-irrelevant-files\n          - refstack-client-devstack-yoga:\n              irrelevant-files: *refstack-client-irrelevant-files\n          - opendev-tox-docs\n          - openstack-tox-py39\n      promote:\n        jobs:\n          - opendev-promote-docs\n\n  in \"openinfra/refstack-client/.zuul.yaml@master\", line 1, column 3","accounts_in_message":[],"_revision_number":1},{"id":"0989d9e43dc1e1901cc840949ca874d347524e4b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"date":"2023-04-19 15:27:47.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"91bc261bcf57952b69a3f5191427fe1baa88dfdc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 15:28:10.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openinfra/refstack-client on branch master.  The error was:\n\n  Job refstack-client-devstack-antelope not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - openstack-cover-jobs\n        - openstack-python3-antelope-jobs\n      check:\n        jobs:\n          - refstack-client-devstack-master:\n              # Define a list of irrelevant files to use everywhere else\n              irrelevant-files: \u0026refstack-client-irrelevant-files\n                - ^.*\\.rst$\n                - ^doc/.*$\n                - ^.gitignore$\n                - ^.gitreview$\n          - refstack-client-devstack-antelope:\n              irrelevant-files: *refstack-client-irrelevant-files\n          - refstack-client-devstack-zed:\n              irrelevant-files: *refstack-client-irrelevant-files\n          - refstack-client-devstack-yoga:\n              irrelevant-files: *refstack-client-irrelevant-files\n          - opendev-tox-docs\n          - openstack-tox-py39\n      gate:\n        jobs:\n          - refstack-client-devstack-master:\n              irrelevant-files: *refstack-client-irrelevant-files\n          - refstack-client-devstack-antelope:\n              irrelevant-files: *refstack-client-irrelevant-files\n          - refstack-client-devstack-zed:\n              irrelevant-files: *refstack-client-irrelevant-files\n          - refstack-client-devstack-yoga:\n              irrelevant-files: *refstack-client-irrelevant-files\n          - opendev-tox-docs\n          - openstack-tox-py39\n      promote:\n        jobs:\n          - opendev-promote-docs\n\n  in \"openinfra/refstack-client/.zuul.yaml@master\", line 1, column 3","accounts_in_message":[],"_revision_number":2},{"id":"6af49aba0889635799047b1e5f78fd42503edd23","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2023-04-21 13:28:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9c160f85efaa39230dcd4d294e629192c0712674","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"date":"2023-04-24 12:41:05.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"a5590dd9a36a40ded33c8cf79c9f767d7a425780","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-24 12:41:41.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openinfra/refstack-client on branch master.  The error was:\n\n  Job refstack-client-devstack-antelope not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - openstack-cover-jobs\n        - openstack-python3-antelope-jobs\n      check:\n        jobs:\n          - refstack-client-devstack-master:\n              # Define a list of irrelevant files to use everywhere else\n              irrelevant-files: \u0026refstack-client-irrelevant-files\n                - ^.*\\.rst$\n                - ^doc/.*$\n                - ^.gitignore$\n                - ^.gitreview$\n          - refstack-client-devstack-antelope:\n              irrelevant-files: *refstack-client-irrelevant-files\n          - refstack-client-devstack-zed:\n              irrelevant-files: *refstack-client-irrelevant-files\n          - refstack-client-devstack-yoga:\n              irrelevant-files: *refstack-client-irrelevant-files\n          - opendev-tox-docs\n          - openstack-tox-py39\n      gate:\n        jobs:\n          - refstack-client-devstack-master:\n              irrelevant-files: *refstack-client-irrelevant-files\n          - refstack-client-devstack-antelope:\n              irrelevant-files: *refstack-client-irrelevant-files\n          - refstack-client-devstack-zed:\n              irrelevant-files: *refstack-client-irrelevant-files\n          - refstack-client-devstack-yoga:\n              irrelevant-files: *refstack-client-irrelevant-files\n          - opendev-tox-docs\n          - openstack-tox-py39\n      promote:\n        jobs:\n          - opendev-promote-docs\n\n  in \"openinfra/refstack-client/.zuul.yaml@master\", line 1, column 3","accounts_in_message":[],"_revision_number":3},{"id":"e8c3125345f1cbd9022d62cfac5ef739ec923db3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"date":"2023-04-24 12:46:29.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"06edac84113e0fc4baa750f4dd2189260b686465","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-24 14:18:46.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da3a26553d2340b8b065e44c40f4787f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0b74f91c35134032a32b6d1fb8e49a75 : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a43543d386b04785805c009d27477907 : SUCCESS in 2m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/12d1443f50d24812acebb4afd9511360 : SUCCESS in 3m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0eb14d3452424c4d9211d7289524e5db : SUCCESS in 2m 43s\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/23a59e9d480248d6b3ea056e940756e7 : SUCCESS in 1h 22m 58s\n- refstack-client-devstack-2023-1 https://zuul.opendev.org/t/openstack/build/a040536955e14c5095a7d413c76e05c7 : SUCCESS in 1h 19m 20s\n- refstack-client-devstack-zed https://zuul.opendev.org/t/openstack/build/3b9059e96f3c4dab9e869dcf81be1b07 : SUCCESS in 1h 24m 20s\n- refstack-client-devstack-yoga https://zuul.opendev.org/t/openstack/build/98f7658cc1914e92adc316cdbdd7fb4a : SUCCESS in 1h 02m 47s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/02898c195e46451f87ff93021ff35a43 : SUCCESS in 2m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5eeadddbb8ae42f1bc0ff59714387360 : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":4},{"id":"d3a21f7d8f7d8b45c9e0d686fd221c9172f80636","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2023-05-22 19:25:18.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e8facd2a75c3b110936d99eaab17a6595f713601","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 19:25:30.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fc4ee2bab916478ba5f67f05ebefa7a4\n\n\nWarning:\n  Change 880868 in project openinfra/ansible-role-refstack-client does not share a change queue with 880829 in project openinfra/refstack-client","accounts_in_message":[],"_revision_number":4},{"id":"a54053c09007dab480f7f729c0afc134b0a474be","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2023-05-23 07:52:56.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b7f5ad3428317e409e718ae363dc180d679f5442","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-23 09:20:35.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/42d0fc6a04e64810a2718904674d642a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fec8f5a2244b4d4bb4935933a67c4f11 : SUCCESS in 2m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/690fd34fdb514a799a384a8285d5e027 : SUCCESS in 2m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/235682a92dfa4592abc102889890e113 : SUCCESS in 2m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d24c2c6323cb4162a637491dbfde1263 : SUCCESS in 2m 56s\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/7d0f85bdd3b9453dbaa62d53a25134ca : SUCCESS in 1h 19m 20s\n- refstack-client-devstack-2023-1 https://zuul.opendev.org/t/openstack/build/f230774f40374a9a8d524bcc630f04e5 : FAILURE in 1h 20m 16s\n- refstack-client-devstack-zed https://zuul.opendev.org/t/openstack/build/eee0d5cdc4d24ed5bb1c42b2649fb321 : SUCCESS in 1h 03m 58s\n- refstack-client-devstack-yoga https://zuul.opendev.org/t/openstack/build/0de59cd078a0481a93dcffd67b0567db : SUCCESS in 1h 22m 19s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/54f2fe902a9c44ac875a1f322f7adb88 : SUCCESS in 2m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/caa1d579ee6b4d9092d03e4a4839f677 : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":4},{"id":"55163a93bd8acb3b6a1a824d273adbbf26bce758","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2023-05-23 13:59:47.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"fe9bbd3b413724bc65ddf85a5d276389ac7f12d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-23 15:36:24.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2af568eee824bcbad84703e47aa09c5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5693ed1ea77345e983c55d83a350386a : SUCCESS in 2m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a349f13029084808bc7c8352598f2b56 : SUCCESS in 2m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f6812105b5354143aca990de451ef624 : SUCCESS in 2m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/42110d2f852b4ac8b47a5948f17f7d2c : SUCCESS in 2m 53s\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/8951e510b10c43b98178f3e5d6697688 : SUCCESS in 1h 27m 19s\n- refstack-client-devstack-2023-1 https://zuul.opendev.org/t/openstack/build/c8baf028b4ee4d50ba7adfa38ee46f70 : SUCCESS in 1h 18m 04s\n- refstack-client-devstack-zed https://zuul.opendev.org/t/openstack/build/52c93dad6f5b400cbe02a0ba14c3fb25 : SUCCESS in 1h 09m 29s\n- refstack-client-devstack-yoga https://zuul.opendev.org/t/openstack/build/11bbf93bb4a0437e8158fbd9a30e676d : SUCCESS in 1h 28m 47s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/898f92bb8a244849811c26998a9ad237 : SUCCESS in 2m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c5cbfdf54a3848b998b167ae10c9b088 : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":4},{"id":"90c45964d2cf90bc01d3335f34a9a0138f2a5676","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-23 15:37:06.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"ac4ff47a8d58b9b28ab72253a711aedeac61aba0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-23 17:09:10.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f941061e5ae64916a7b2aa8e88f74a85\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcf1a101a2174e6b88035d49640577e8 : SUCCESS in 2m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1142dfa7f93e40c688b83968d8fc0eed : SUCCESS in 3m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7d02e73682964850b30cd48b17e33776 : SUCCESS in 3m 05s\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/3db4b971238647119ecf4f730c40fb1a : SUCCESS in 1h 16m 41s\n- refstack-client-devstack-2023-1 https://zuul.opendev.org/t/openstack/build/56394020fbfc422f98bde043a933a9fa : SUCCESS in 1h 13m 52s\n- refstack-client-devstack-zed https://zuul.opendev.org/t/openstack/build/5f68b7d8b02f43e2b9d3ab63c40b5e64 : SUCCESS in 1h 24m 41s\n- refstack-client-devstack-yoga https://zuul.opendev.org/t/openstack/build/0d228fc6c3214d0896ea2bba93a83d2e : SUCCESS in 1h 25m 24s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/246eb6fd150d4a1895fb6d606df6fed9 : SUCCESS in 2m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/21100cc301fe4771b42504947355a40e : SUCCESS in 3m 02s","accounts_in_message":[],"_revision_number":4},{"id":"5a6d1a7ac3f6efeb48a60a2fbcb1972ec669fe0f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-23 17:09:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"6240d2b4c8cbde19ba9eaaa28a7dae4e896326ff","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-23 17:11:04.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/969de3a1acec4923bc9afcf3b76a263b\n\n- opendev-promote-docs https://zuul.opendev.org/t/openstack/build/1eb8bf8e2b6e42688b5958da0d4a2628 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"b41463f32cd65ee604e756439c671973c8aa6ed7","revisions":{"54d8e056712a9f6ac7f0c203f31814f976523383":{"kind":"REWORK","_number":1,"created":"2023-04-19 14:16:38.000000000","uploader":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"ref":"refs/changes/29/880829/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack-client","ref":"refs/changes/29/880829/1","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/29/880829/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/29/880829/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/29/880829/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack-client refs/changes/29/880829/1"}}},"commit":{"parents":[{"commit":"3f7a3edd1e950d3b3228531fe006d6387ec81ac0","subject":"Merge \"Change checkout commit in setup_env\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/3f7a3edd1e950d3b3228531fe006d6387ec81ac0"}]}],"author":{"name":"jskunda","email":"jskunda@redhat.com","date":"2023-04-19 14:15:07.000000000","tz":120},"committer":{"name":"jskunda","email":"jskunda@redhat.com","date":"2023-04-19 14:15:07.000000000","tz":120},"subject":"Update jobs for Bobcat development cycle","message":"Update jobs for Bobcat development cycle\n\nAdds an Antelope job as upstream has just released it and moved to\nthe Bobcat development cycle\n\nChange-Id: Ifbdbe55e2309c4855cc5f069487739aa91ee2a89\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/54d8e056712a9f6ac7f0c203f31814f976523383"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/54d8e056712a9f6ac7f0c203f31814f976523383"}]},"branch":"refs/heads/master"},"35e5b11d75054a955bc4b6b16c1a58ee69f63ec7":{"kind":"REWORK","_number":2,"created":"2023-04-19 15:27:47.000000000","uploader":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"ref":"refs/changes/29/880829/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack-client","ref":"refs/changes/29/880829/2","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/29/880829/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/29/880829/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/29/880829/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack-client refs/changes/29/880829/2"}}},"commit":{"parents":[{"commit":"3f7a3edd1e950d3b3228531fe006d6387ec81ac0","subject":"Merge \"Change checkout commit in setup_env\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/3f7a3edd1e950d3b3228531fe006d6387ec81ac0"}]}],"author":{"name":"jskunda","email":"jskunda@redhat.com","date":"2023-04-19 14:15:07.000000000","tz":120},"committer":{"name":"jskunda","email":"jskunda@redhat.com","date":"2023-04-19 15:27:31.000000000","tz":120},"subject":"Update jobs for Bobcat development cycle","message":"Update jobs for Bobcat development cycle\n\nAdds an Antelope job as upstream has just released it and moved to\nthe Bobcat development cycle\n\nChange-Id: Ifbdbe55e2309c4855cc5f069487739aa91ee2a89\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/35e5b11d75054a955bc4b6b16c1a58ee69f63ec7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/35e5b11d75054a955bc4b6b16c1a58ee69f63ec7"}]},"branch":"refs/heads/master"},"bd0264dce17dd46f9755388c52f8b71b250e790c":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-04-24 12:41:05.000000000","uploader":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"ref":"refs/changes/29/880829/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack-client","ref":"refs/changes/29/880829/3","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/29/880829/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/29/880829/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/29/880829/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack-client refs/changes/29/880829/3"}}},"commit":{"parents":[{"commit":"3f7a3edd1e950d3b3228531fe006d6387ec81ac0","subject":"Merge \"Change checkout commit in setup_env\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/3f7a3edd1e950d3b3228531fe006d6387ec81ac0"}]}],"author":{"name":"jskunda","email":"jskunda@redhat.com","date":"2023-04-19 14:15:07.000000000","tz":120},"committer":{"name":"Jakub Skunda","email":"jskunda@redhat.com","date":"2023-04-24 12:41:05.000000000","tz":0},"subject":"Update jobs for Bobcat development cycle","message":"Update jobs for Bobcat development cycle\n\nAdds an Antelope job as upstream has just released it and moved to\nthe Bobcat development cycle\n\nChange-Id: Ifbdbe55e2309c4855cc5f069487739aa91ee2a89\nDepends-On: https://review.opendev.org/c/openinfra/ansible-role-refstack-client/+/880868\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/bd0264dce17dd46f9755388c52f8b71b250e790c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/bd0264dce17dd46f9755388c52f8b71b250e790c"}]},"branch":"refs/heads/master","description":"Edit commit message"},"b41463f32cd65ee604e756439c671973c8aa6ed7":{"kind":"REWORK","_number":4,"created":"2023-04-24 12:46:29.000000000","uploader":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"ref":"refs/changes/29/880829/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack-client","ref":"refs/changes/29/880829/4","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/29/880829/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/29/880829/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/29/880829/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack-client refs/changes/29/880829/4"}}},"commit":{"parents":[{"commit":"3f7a3edd1e950d3b3228531fe006d6387ec81ac0","subject":"Merge \"Change checkout commit in setup_env\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/3f7a3edd1e950d3b3228531fe006d6387ec81ac0"}]}],"author":{"name":"jskunda","email":"jskunda@redhat.com","date":"2023-04-19 14:15:07.000000000","tz":120},"committer":{"name":"jskunda","email":"jskunda@redhat.com","date":"2023-04-24 12:46:02.000000000","tz":120},"subject":"Update jobs for Bobcat development cycle","message":"Update jobs for Bobcat development cycle\n\nAdds an Antelope job as upstream has just released it and moved to\nthe Bobcat development cycle\n\nChange-Id: Ifbdbe55e2309c4855cc5f069487739aa91ee2a89\nDepends-On: https://review.opendev.org/c/openinfra/ansible-role-refstack-client/+/880868\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/b41463f32cd65ee604e756439c671973c8aa6ed7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/b41463f32cd65ee604e756439c671973c8aa6ed7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
