)]}'
{"id":"openinfra%2Frefstack-client~881818","triplet_id":"openinfra%2Frefstack-client~master~I94f86bc5b2fa7a1038cfa8705d8b7dec72e4fa7e","project":"openinfra/refstack-client","branch":"master","topic":"add/fips-job","attention_set":{},"removed_from_attention_set":{"33732":{"account":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"last_update":"2023-07-17 18:58:54.000000000","reason":"Change was submitted"},"22873":{"account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"last_update":"2023-07-17 08:30:15.000000000","reason":"\u003cGERRIT_ACCOUNT_22873\u003e replied on the change","reason_account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}}},"hashtags":[],"change_id":"I94f86bc5b2fa7a1038cfa8705d8b7dec72e4fa7e","subject":"Add fips job","status":"MERGED","created":"2023-04-28 12:11:14.000000000","updated":"2023-07-17 19:00:31.000000000","submitted":"2023-07-17 18:58:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"881818-add/fips-job","meta_rev_id":"8f47e187eb27846fdecadc7b85252f44c10d4b93","_number":881818,"virtual_id_number":881818,"owner":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-17 18:58:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":2,"date":"2023-07-17 13:06:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":1,"date":"2023-07-17 13:06:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-28 13:44:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-21 16:07:57.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"}],"messages":[{"id":"f128e84e3df71864e0b160ad512a1db88eb8a7b8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2023-04-28 12:11:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f9d6d8715e0729d0a67c300c800a541c8614d8fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 13:44:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de68f3e373364342b8d1580c18f25950\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/18a215b57e614a54a0161408a315004c : SUCCESS in 2m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1c360600d914f409b87d460f91f2b8a : SUCCESS in 2m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5201c853a5a84a67b49cdb498635d4a9 : SUCCESS in 3m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cd002f545cf74db9b167009b6f5cb7d5 : SUCCESS in 4m 13s\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/110101c41d60469bb65ed054ff22ac25 : SUCCESS in 1h 18m 51s\n- refstack-client-devstack-master-fips https://zuul.opendev.org/t/openstack/build/ac9f8429eb50401c86ae45b19170c012 : RETRY_LIMIT in 5m 12s (non-voting)\n- refstack-client-devstack-zed https://zuul.opendev.org/t/openstack/build/9fdebb75b81340f28dd03f46fa460372 : SUCCESS in 1h 11m 48s\n- refstack-client-devstack-yoga https://zuul.opendev.org/t/openstack/build/59ceef0ef9064f3581ca1a27d6137dc4 : SUCCESS in 1h 16m 41s\n- refstack-client-devstack-xena https://zuul.opendev.org/t/openstack/build/c349380493ee4afa82860b4e18312696 : SUCCESS in 1h 18m 15s\n- refstack-client-devstack-wallaby https://zuul.opendev.org/t/openstack/build/473281d2dfd64e0cb2a5c8666de84154 : SUCCESS in 1h 12m 34s\n- refstack-client-devstack-victoria https://zuul.opendev.org/t/openstack/build/541f93478809409bb9f92c836b410ecc : SUCCESS in 1h 25m 29s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/fa5d1779376f4fdeaa910ee9d1e56e8e : SUCCESS in 2m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5ff46519d7df47868fdc21da64b98468 : SUCCESS in 2m 55s","accounts_in_message":[],"_revision_number":1},{"id":"4b5b52747010998d4c1f7a7e358a11e1510bd040","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2023-04-28 14:37:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"83057f1cf1bccf3e56e2da6676a8dd7eba0be1b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 16:21:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af7f98e5ac144b4081503c73e39f4766\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2734ccd5ac5b4ecca70afbed1242a921 : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8b02c14620246c290a7ec9f9fdb1a50 : SUCCESS in 2m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ac17ab1e7e2401cab826a3f16134eb6 : SUCCESS in 2m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/32c5ab5e45684992a26d36617f7a075e : SUCCESS in 3m 04s\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/bef51f6d89ce4cd88f76d54e5cdd8214 : SUCCESS in 1h 35m 14s\n- refstack-client-devstack-master-fips https://zuul.opendev.org/t/openstack/build/ba8afccad6aa4ffaa9251cd71702cac0 : RETRY_LIMIT in 5m 08s (non-voting)\n- refstack-client-devstack-zed https://zuul.opendev.org/t/openstack/build/d700fd36ca2343cd9bc3fe9a1af53006 : SUCCESS in 1h 17m 45s\n- refstack-client-devstack-yoga https://zuul.opendev.org/t/openstack/build/3c32285634254c06bdb6b9d15c2ae254 : SUCCESS in 1h 28m 34s\n- refstack-client-devstack-xena https://zuul.opendev.org/t/openstack/build/7943dc0929ff4f0095d9f92519dd2040 : SUCCESS in 1h 00m 58s\n- refstack-client-devstack-wallaby https://zuul.opendev.org/t/openstack/build/bb2bca8ace6f4499a3eec94fffa11f55 : SUCCESS in 1h 12m 20s\n- refstack-client-devstack-victoria https://zuul.opendev.org/t/openstack/build/63cfdb238f60409494502157fd9435fe : SUCCESS in 1h 14m 05s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/e3ccf3ec5faa440e93c10007051d67aa : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0af1f9cffd1349f2b961c716d84bfeee : SUCCESS in 2m 34s","accounts_in_message":[],"_revision_number":1},{"id":"6676c5486f514b40a954719274d744b3c4ffca26","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2023-05-29 10:41:33.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"1d2133e737c550c8e2fc5a81682423abe01e527d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-29 10:41:57.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":2},{"id":"534c7875cf4305e830ee9743fa40cbf8a9a1e112","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2023-05-29 10:53:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9232a0739883b92f9ee3ba029765dc2f64f0c557","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-29 10:54:03.000000000","message":"Patch Set 2:\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":2},{"id":"b768fc6f13726205ec63e2663c961ca3ded3f7a5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2023-06-02 09:58:21.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"d5d7ab75583942af84a812cccc5ece101b125b56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-02 09:58:42.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":3},{"id":"ebb1477a9fce64a53e116cfae27b1e0b1d79233d","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2023-06-02 09:59:35.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c613caed497820959ec2117ad753aeaebbe77357","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-02 10:00:09.000000000","message":"Patch Set 3:\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":3},{"id":"1a86a33e9a07afde2353698f97bfa995058e4d8b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2023-06-02 10:21:36.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"29ca43d1d91a21961058a423c3010e0ff3547859","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-02 11:57:38.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/de4fe2e844ce45a7b5d208ec968111f4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/236001d26fe248778e15254903b39919 : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a3714067a2044c4af2fdb6058009d27 : SUCCESS in 2m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/48d48f8398a34135874eba10173554b5 : SUCCESS in 2m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/611d09769527412780ed0fb8c32bb747 : SUCCESS in 3m 48s\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/08b3b1f820cf4e07a181cd6ef5a03a73 : SUCCESS in 1h 16m 19s\n- refstack-client-devstack-master-fips-centos9 https://zuul.opendev.org/t/openstack/build/0961d96a636f420480805c19a4ad3260 : FAILURE in 51m 24s (non-voting)\n- refstack-client-devstack-2023-1 https://zuul.opendev.org/t/openstack/build/453872fc79224280a157b35f4947ef66 : FAILURE in 1h 28m 43s\n- refstack-client-devstack-zed https://zuul.opendev.org/t/openstack/build/8cf2d406b974437784369a47f79171c4 : SUCCESS in 1h 18m 06s\n- refstack-client-devstack-yoga https://zuul.opendev.org/t/openstack/build/d0d0bc3828434badb7db523e967c78e3 : SUCCESS in 1h 22m 03s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/a09038bc432c47b88e6590adc6461b7e : SUCCESS in 2m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0785733d53de444baac73580e21ced77 : SUCCESS in 2m 50s","accounts_in_message":[],"_revision_number":4},{"id":"31a9af67140b456c32deeb3ab88de5c2cc782583","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2023-06-02 12:31:35.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7d6969962fcbd3bd63594a4dfc2bee410736c226","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-02 14:25:45.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/39b5f8b99af54069ac2b8e10bfe75182\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0bd18898c6db4be7a72a31a5d870b2f9 : SUCCESS in 2m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4605fff60c754f6f9d019da2658a82d3 : SUCCESS in 2m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d43482c5c51548639dcef0b941e1d4d1 : SUCCESS in 2m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b2f0a9f7350849ebb5799bbb48da820b : SUCCESS in 3m 44s\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/050cbae2851649768c94c789c584bbac : SUCCESS in 1h 18m 37s\n- refstack-client-devstack-master-fips-centos9 https://zuul.opendev.org/t/openstack/build/7ba17fcdce0646aea092d640792d1544 : SUCCESS in 1h 34m 38s (non-voting)\n- refstack-client-devstack-2023-1 https://zuul.opendev.org/t/openstack/build/57ea18c91c0d455b8dc38ce44a2c31fd : FAILURE in 1h 13m 41s\n- refstack-client-devstack-zed https://zuul.opendev.org/t/openstack/build/0ab80e74798c4e1aa667ed772149f215 : SUCCESS in 1h 14m 07s\n- refstack-client-devstack-yoga https://zuul.opendev.org/t/openstack/build/cb310578e414465b9e54e80e2efed177 : SUCCESS in 54m 20s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/bd0036ef3b5d4de2a2c45183b79668ad : SUCCESS in 3m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1c6cad79ad714e01b849c18cc0bd7a79 : SUCCESS in 2m 45s","accounts_in_message":[],"_revision_number":4},{"id":"af7353b7143d055fc42d3d79711e271acdbbf7ff","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2023-06-21 16:07:57.000000000","message":"Patch Set 4: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"ae155248ed0238f438ffdf75d7acb29568c76d89","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2023-06-28 16:48:57.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"b781c9f739d19dfbdb43f2902de334a5f163120b","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2023-07-13 09:46:57.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"34f72212a00df47f0665d98e8dc1908c5de71a7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-13 11:29:20.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/001b12d353e24cd2be2854230e883aca\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c712fc5cada244a3a78141591e6d0687 : SUCCESS in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9de898b74b9c419c8aca2b5ab0f39c33 : SUCCESS in 2m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/78c693bbb0c145ac8570bb68cb5b35f8 : SUCCESS in 2m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e59117b57bc349adaaa3b62dc5778736 : SUCCESS in 4m 01s\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/80ba5021631a42699f04628a199988a6 : SUCCESS in 1h 39m 44s\n- refstack-client-devstack-master-fips-centos9 https://zuul.opendev.org/t/openstack/build/3bcd457ff650450cbf796fc8a45e4684 : FAILURE in 33m 39s\n- refstack-client-devstack-2023-1 https://zuul.opendev.org/t/openstack/build/938b8def525c42b98e95a6bdf7d4c9ad : SUCCESS in 1h 12m 12s\n- refstack-client-devstack-zed https://zuul.opendev.org/t/openstack/build/f853d3b36cc24ae186cb066e18392feb : SUCCESS in 1h 22m 05s\n- refstack-client-devstack-yoga https://zuul.opendev.org/t/openstack/build/52b7119e1ca344a1a99f249f1216fd82 : SUCCESS in 1h 26m 56s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/6ef62ee337a74326a5b4324b450c812d : SUCCESS in 2m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9f44be8617d744079f6b7dc35a84107e : SUCCESS in 2m 49s","accounts_in_message":[],"_revision_number":5},{"id":"39e5d0648c0240c58cb1df23941c09397f0fbb40","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2023-07-14 19:56:34.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":5},{"id":"55dfd922083d2f845ddb16134c5342df3b28347e","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2023-07-17 08:30:15.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"66907bf7de2b6d6c90efde341473c3ed28f95c05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 09:48:46.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e90d8b1fd8fd4d0e9f263eed6eaf5e71\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/01e0a269ae7d48e7a9f9e0d68e9633ad : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/308b299f039241ab9720d59123a0a618 : SUCCESS in 2m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/846116b9ea714c8babf9f2ff090cb0f3 : SUCCESS in 2m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/58bec217a34f4e7e80281acbfb1cb00b : SUCCESS in 3m 51s\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/2892530525f74a95af2ae6991fbfda99 : SUCCESS in 1h 13m 44s\n- refstack-client-devstack-master-fips-centos9 https://zuul.opendev.org/t/openstack/build/3169d21e1be04015be16314172da388e : SUCCESS in 1h 12m 54s\n- refstack-client-devstack-2023-1 https://zuul.opendev.org/t/openstack/build/10f18c9f8307474f8437151daea430b5 : FAILURE in 1h 16m 51s\n- refstack-client-devstack-zed https://zuul.opendev.org/t/openstack/build/9b07664d97474563bf726c632892f301 : SUCCESS in 1h 14m 02s\n- refstack-client-devstack-yoga https://zuul.opendev.org/t/openstack/build/573b30eb48be47608fce5f951da8772f : SUCCESS in 1h 12m 29s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/5005f4b9c69c4fd1be37a6f648ab79c0 : SUCCESS in 2m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/67a5169337d24f62bcd1197cbddf4598 : SUCCESS in 2m 53s","accounts_in_message":[],"_revision_number":5},{"id":"4386da74614a996a919a5093b55a7ddae72d1886","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2023-07-17 13:06:25.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"1f416ec7426d09cb8565036d5437647f57f3ca95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 15:12:53.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4b7436f8a7784e5091a8b54f4e0a5f42\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b06ffdd2788b449b998a304e29ae408a : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a975eab505a84082b3fc35c5f6eea83c : SUCCESS in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f22ea967d14f4cada73a2f0ea5d36cae : SUCCESS in 3m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3b8af0ee7d4540bca0de173234bfa5fe : SUCCESS in 3m 05s\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/27423905049b48a9b5178f89f627f5ad : SUCCESS in 1h 31m 50s\n- refstack-client-devstack-master-fips-centos9 https://zuul.opendev.org/t/openstack/build/46b3c5e040c94d1d9bc350ceed0ca5f6 : SUCCESS in 1h 59m 03s\n- refstack-client-devstack-2023-1 https://zuul.opendev.org/t/openstack/build/39ca7b85021d4b56a99b1ac3232e9f3a : FAILURE in 1h 25m 46s\n- refstack-client-devstack-zed https://zuul.opendev.org/t/openstack/build/d0ecb53066ce4cb4823c5ace3e570a06 : SUCCESS in 1h 00m 09s\n- refstack-client-devstack-yoga https://zuul.opendev.org/t/openstack/build/f10d7fe3a298499b9514a08ebba5cac0 : SUCCESS in 1h 12m 22s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/db8ff7f6d79041e4accfa8f1f589fad0 : SUCCESS in 2m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ad06f1d3362e401b9a5f20122bd06a32 : SUCCESS in 3m 11s","accounts_in_message":[],"_revision_number":5},{"id":"dc400c43ca6d9024306a93d3269cc1bb0ad97ec4","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2023-07-17 15:38:25.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"8539b2fc706630187a89378e8224d1c1281f706f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 17:29:46.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a9b4557992344af9510334f7dc9e549\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d79caeb21ea3471daba78338b65c0224 : SUCCESS in 2m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02faa9a8e716449ea29a4d17f8c11637 : SUCCESS in 2m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/80be1dc091a6445da64e48945339f7d3 : SUCCESS in 2m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fb1cea349df4427d9261de5f98207801 : SUCCESS in 2m 39s\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/f61732d7074b4e818c67923f284be2ff : SUCCESS in 1h 18m 05s\n- refstack-client-devstack-master-fips-centos9 https://zuul.opendev.org/t/openstack/build/879e91d9d66243439060978db82562b0 : SUCCESS in 1h 45m 55s\n- refstack-client-devstack-2023-1 https://zuul.opendev.org/t/openstack/build/6a9de7f57a55441badea844d4f28345d : SUCCESS in 1h 24m 40s\n- refstack-client-devstack-zed https://zuul.opendev.org/t/openstack/build/295022eb97fc4cecb976f88008724a4a : SUCCESS in 1h 01m 17s\n- refstack-client-devstack-yoga https://zuul.opendev.org/t/openstack/build/8b7052e103b5423586b393cd8e099d15 : SUCCESS in 1h 02m 33s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/570b44f36fe74fd3860a6aac913c7003 : SUCCESS in 2m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/760102f80c914a55a641bc9c131bbef7 : SUCCESS in 2m 54s","accounts_in_message":[],"_revision_number":5},{"id":"4f98bcd90d16371502f167fbd057982ae695a4d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 17:30:26.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"8ca6f4e3637f581b237704f156cce99113bf14e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 18:58:54.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/609bcf18e78e4a89aabb1a9ffd31d877\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d16c92ee952e438eafd631b1e57d0eff : SUCCESS in 2m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/72d51d28773a472b902dcdad080d7db8 : SUCCESS in 2m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/88eac7e2dea44aec857972ea0dac187d : SUCCESS in 2m 44s\n- refstack-client-devstack-master https://zuul.opendev.org/t/openstack/build/9ec7a82c1e554769bfd006cb10f8af60 : SUCCESS in 1h 23m 29s\n- refstack-client-devstack-2023-1 https://zuul.opendev.org/t/openstack/build/5ae69e5a6ac14e7a8c8408c087ade4eb : SUCCESS in 1h 13m 11s\n- refstack-client-devstack-zed https://zuul.opendev.org/t/openstack/build/063dbd0e44ad4c8c8515b740f7a109d2 : SUCCESS in 52m 50s\n- refstack-client-devstack-yoga https://zuul.opendev.org/t/openstack/build/fd7b8117511c490a97c0c116811e94d0 : SUCCESS in 1h 04m 44s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/78ae72436f16419a801280959e69e5d0 : SUCCESS in 3m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/75a8a29b49c04a2d9e21cb77b62d149d : SUCCESS in 2m 35s","accounts_in_message":[],"_revision_number":5},{"id":"fd8e1d746b50058822defbb2692cb8d01d59f331","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 18:58:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"8f47e187eb27846fdecadc7b85252f44c10d4b93","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 19:00:31.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a8d7099c16c4a09b9ff6b513d8089f2\n\n- opendev-promote-docs https://zuul.opendev.org/t/openstack/build/7bd94cc928974f16a9f8c828ae7428b8 : SUCCESS in 33s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"adfd17a082be11ebd5f65c62594c4638f5aac44b","revisions":{"20b9677285dfb5a5cecb5786e2727e698628d8a0":{"kind":"REWORK","_number":1,"created":"2023-04-28 12:11:14.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/18/881818/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack-client","ref":"refs/changes/18/881818/1","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/18/881818/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/18/881818/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/18/881818/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack-client refs/changes/18/881818/1"}}},"commit":{"parents":[{"commit":"3f7a3edd1e950d3b3228531fe006d6387ec81ac0","subject":"Merge \"Change checkout commit in setup_env\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/3f7a3edd1e950d3b3228531fe006d6387ec81ac0"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2023-04-28 12:10:33.000000000","tz":120},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2023-04-28 12:10:33.000000000","tz":120},"subject":"Add fips job","message":"Add fips job\n\nDepends-On: https://review.opendev.org/c/openinfra/ansible-role-refstack-client/+/881815\nChange-Id: I94f86bc5b2fa7a1038cfa8705d8b7dec72e4fa7e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/20b9677285dfb5a5cecb5786e2727e698628d8a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/20b9677285dfb5a5cecb5786e2727e698628d8a0"}]},"branch":"refs/heads/master"},"2e62f09322c4bc10b60c631d76d1e2984728b70f":{"kind":"REWORK","_number":2,"created":"2023-05-29 10:41:33.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/18/881818/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack-client","ref":"refs/changes/18/881818/2","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/18/881818/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/18/881818/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/18/881818/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack-client refs/changes/18/881818/2"}}},"commit":{"parents":[{"commit":"b41463f32cd65ee604e756439c671973c8aa6ed7","subject":"Update jobs for Bobcat development cycle","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/b41463f32cd65ee604e756439c671973c8aa6ed7"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2023-04-28 12:10:33.000000000","tz":120},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2023-05-29 10:41:22.000000000","tz":120},"subject":"Add fips job","message":"Add fips job\n\nDepends-On: https://review.opendev.org/c/openinfra/ansible-role-refstack-client/+/881815\nChange-Id: I94f86bc5b2fa7a1038cfa8705d8b7dec72e4fa7e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/2e62f09322c4bc10b60c631d76d1e2984728b70f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/2e62f09322c4bc10b60c631d76d1e2984728b70f"}]},"branch":"refs/heads/master"},"ac4a262409703f3df6ca0f390ea038b1d4a15ff6":{"kind":"REWORK","_number":3,"created":"2023-06-02 09:58:21.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/18/881818/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack-client","ref":"refs/changes/18/881818/3","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/18/881818/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/18/881818/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/18/881818/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack-client refs/changes/18/881818/3"}}},"commit":{"parents":[{"commit":"b41463f32cd65ee604e756439c671973c8aa6ed7","subject":"Update jobs for Bobcat development cycle","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/b41463f32cd65ee604e756439c671973c8aa6ed7"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2023-04-28 12:10:33.000000000","tz":120},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2023-06-02 09:58:06.000000000","tz":120},"subject":"Add fips job","message":"Add fips job\n\nDepends-On: https://review.opendev.org/c/openinfra/ansible-role-refstack-client/+/881815\nChange-Id: I94f86bc5b2fa7a1038cfa8705d8b7dec72e4fa7e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/ac4a262409703f3df6ca0f390ea038b1d4a15ff6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/ac4a262409703f3df6ca0f390ea038b1d4a15ff6"}]},"branch":"refs/heads/master"},"9516ebfaf4b95b1ede499269047cc4bb47d4ccd8":{"kind":"REWORK","_number":4,"created":"2023-06-02 10:21:36.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/18/881818/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack-client","ref":"refs/changes/18/881818/4","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/18/881818/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/18/881818/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/18/881818/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack-client refs/changes/18/881818/4"}}},"commit":{"parents":[{"commit":"b41463f32cd65ee604e756439c671973c8aa6ed7","subject":"Update jobs for Bobcat development cycle","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/b41463f32cd65ee604e756439c671973c8aa6ed7"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2023-04-28 12:10:33.000000000","tz":120},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2023-06-02 10:21:25.000000000","tz":120},"subject":"Add fips job","message":"Add fips job\n\nDepends-On: https://review.opendev.org/c/openinfra/ansible-role-refstack-client/+/881815\nChange-Id: I94f86bc5b2fa7a1038cfa8705d8b7dec72e4fa7e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/9516ebfaf4b95b1ede499269047cc4bb47d4ccd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/9516ebfaf4b95b1ede499269047cc4bb47d4ccd8"}]},"branch":"refs/heads/master"},"adfd17a082be11ebd5f65c62594c4638f5aac44b":{"kind":"REWORK","_number":5,"created":"2023-06-28 16:48:57.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/18/881818/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack-client","ref":"refs/changes/18/881818/5","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/18/881818/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/18/881818/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack-client refs/changes/18/881818/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack-client refs/changes/18/881818/5"}}},"commit":{"parents":[{"commit":"b41463f32cd65ee604e756439c671973c8aa6ed7","subject":"Update jobs for Bobcat development cycle","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/b41463f32cd65ee604e756439c671973c8aa6ed7"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2023-04-28 12:10:33.000000000","tz":120},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2023-06-28 16:48:48.000000000","tz":120},"subject":"Add fips job","message":"Add fips job\n\nDepends-On: https://review.opendev.org/c/openinfra/ansible-role-refstack-client/+/881815\nChange-Id: I94f86bc5b2fa7a1038cfa8705d8b7dec72e4fa7e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/adfd17a082be11ebd5f65c62594c4638f5aac44b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack-client/commit/adfd17a082be11ebd5f65c62594c4638f5aac44b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
