)]}'
{"id":"openinfra%2Frefstack~813065","triplet_id":"openinfra%2Frefstack~master~I6d64cb67529585b458b28ed4fe011d50d1f23d6d","project":"openinfra/refstack","branch":"master","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2021-10-16 18:06:10.000000000","reason":"Goutham Pacha Ravi replied on the change"},"7160":{"account":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"last_update":"2021-10-16 18:18:41.000000000","reason":"Change was submitted"},"22873":{"account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"last_update":"2021-10-16 18:18:41.000000000","reason":"Change was submitted"},"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2021-10-16 18:18:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6d64cb67529585b458b28ed4fe011d50d1f23d6d","subject":"Switch testing to Yoga testing runtime","status":"MERGED","created":"2021-10-07 18:45:47.000000000","updated":"2021-10-16 18:22:26.000000000","submitted":"2021-10-16 18:18:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"813065","meta_rev_id":"dfc1bf4180a24943d06bd38435aead14ef5b8e6b","_number":813065,"virtual_id_number":813065,"owner":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-16 18:18:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2021-10-16 18:06:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2021-10-16 18:07:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}],"reviewers":{"REVIEWER":[{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-07 18:54:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-13 09:38:43.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2021-10-13 09:38:43.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2021-10-13 09:38:43.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"state":"REVIEWER"}],"messages":[{"id":"d4350b608322e9996a3d0b6298204a47648834f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-10-07 18:45:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"53f9506e4b421a81e6d75e7f7a689541d6eadc27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 18:54:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- nodejs-run-lint https://zuul.opendev.org/t/openstack/build/ea8fcc58fe70403785c863d9037a9ad0 : SUCCESS in 6m 09s\n- nodejs-run-test-browser https://zuul.opendev.org/t/openstack/build/382e19da93c94a08b78709dd410db865 : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4de95e28d9e43fea4f9b5daf979eddd : SUCCESS in 4m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6c2c6e057e8b4916a4dc7639d740bd35 : SUCCESS in 6m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9f3ef38606c4d818a252d6ef4cc9411 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df99d2fb5f124d569db0849f37bb2a73 : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/23d1dc86586e4a329caebba922539267 : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d1efa6daca8547aab0c84916f7eb37d7 : SUCCESS in 5m 30s (non-voting)\n- refstack-tox-functional https://zuul.opendev.org/t/openstack/build/903a3436f58943d4bd3a67c09fbbb2cd : SUCCESS in 6m 45s\n- refstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9e087e01ec2d435795e3f82bf9dfdee1 : FAILURE in 5m 00s","accounts_in_message":[],"_revision_number":1},{"id":"b08a4fd83d5ebfe667cd86de8c3c12e51a598bd8","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-10-08 18:59:33.000000000","message":"Patch Set 1:\n\npip-check-reqs is broken - https://github.com/r1chardj0n3s/pip-check-reqs/issues/70","accounts_in_message":[],"_revision_number":1},{"id":"5ad548c84b262c5006d45966bc536dd72fe9e3ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-10-08 19:13:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ef7d5267438115c4a6ad5fc9d9ab1585e22ac613","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-08 19:23:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- nodejs-run-lint https://zuul.opendev.org/t/openstack/build/2349153d2a6d44ebbbef2d1e022d1884 : SUCCESS in 7m 59s\n- nodejs-run-test-browser https://zuul.opendev.org/t/openstack/build/7777f03006db4141aac62cd09efbcb8b : SUCCESS in 6m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4617c9e8c43d47ee8815863d5cb09da2 : SUCCESS in 4m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/10475cd8474149ccbebafd13e4055ee4 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11a31c675f794a4fba09c3f0da2cfb9b : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/09e9b6d7f78749cc98998872b564d21f : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c1e89e25b14e414ba84f3dde50524fa7 : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ad576f2e513c408081ab01e9e6f85d16 : SUCCESS in 5m 27s (non-voting)\n- refstack-tox-functional https://zuul.opendev.org/t/openstack/build/75dfdcff548e420e8fb8589af476969f : SUCCESS in 5m 44s\n- refstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9db6196cdf2e482e858b81efe9ed8d33 : FAILURE in 5m 55s","accounts_in_message":[],"_revision_number":2},{"id":"15dabc067c97929d5e1c366de329d1b33b21932f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-10-11 06:57:48.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7e579a6bf658e1044108a508450f9b0b81b215e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 07:08:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs-run-lint https://zuul.opendev.org/t/openstack/build/fca928169cee4525aea2b76a35873a66 : SUCCESS in 5m 46s\n- nodejs-run-test-browser https://zuul.opendev.org/t/openstack/build/f399c8e9c9f7470a860537a9592b18e2 : SUCCESS in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ec836a9bbe6452b87b4524c60cddc53 : SUCCESS in 4m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/da74b0cfe7384a33b34940629fd1ccd7 : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/443c36b76f1046cea766e8d36a0671b2 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf38b97a79c149b39a1cf6ecb69160d1 : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e1a87151526e4acda33b5312f148d7a1 : SUCCESS in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e95236139ec4410eb1fbe4578264ab4b : SUCCESS in 4m 50s (non-voting)\n- refstack-tox-functional https://zuul.opendev.org/t/openstack/build/e23acf8037d445e1bdad4cc4024ccae8 : SUCCESS in 5m 03s\n- refstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4ac0bce2c81a4d1a8dd2301e363e98ec : FAILURE in 4m 52s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cf9e56356ed160eeff14cb7178f9ef69f966a131","author":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"date":"2021-10-13 20:07:18.000000000","message":"Patch Set 3: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"a5d35d0429690ed073309c1bd05db52c37c8fea8","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-10-14 07:09:21.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"c3672a9306dfbb9718ac6780d5ab6c8bd9c32b17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-10-14 07:09:49.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"503b412a2c3d708a6537d76f56e7555065d13acc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-14 07:23:15.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs-run-lint https://zuul.opendev.org/t/openstack/build/680a39e378a04cbda7b286630ce4a02b : SUCCESS in 5m 50s\n- nodejs-run-test-browser https://zuul.opendev.org/t/openstack/build/6d6ba6f367f0475e9c0d970deac306d3 : SUCCESS in 8m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee691a5e00e34666ae7cd1d959497408 : SUCCESS in 5m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/277db246b74c4ebea5de370ae5649f27 : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6555c241377a460ba3192512c6948772 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/94b7fc34b7024906bd01f0d23eb1f189 : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7bb071db7d294a6f8fcaa9613df76cd6 : SUCCESS in 6m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5bef7f0b4d2a47899f06707335d662cb : SUCCESS in 6m 16s (non-voting)\n- refstack-tox-functional https://zuul.opendev.org/t/openstack/build/5ccc5a4c891c4611b931d9977d8c9515 : SUCCESS in 6m 16s\n- refstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d9ee13937dbd4418a5340baa473cb493 : FAILURE in 5m 21s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8b328fc0b4a9780fb383ee0cd903bff9f5b652cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2021-10-14 13:11:51.000000000","message":"Uploaded patch set 5: New patch set was added with same tree, parent, and commit message as Patch Set 4.","accounts_in_message":[],"_revision_number":5},{"id":"8e263b8ffac9d128cbbd75e1cd3738862d614eb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-14 13:23:36.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- nodejs-run-lint https://zuul.opendev.org/t/openstack/build/c581ad69a0f84143ac79ec9c8ecfe499 : SUCCESS in 6m 13s\n- nodejs-run-test-browser https://zuul.opendev.org/t/openstack/build/f893059d978742beb034436295897779 : SUCCESS in 7m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5119cfa2f104b78b85c304ad4ef9618 : SUCCESS in 5m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e08dbe2f40549bdbe34569363563d44 : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39bef46a476e4be2b84346e6bc1b7780 : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c7e503dd4734110a52ac941bd8b9ab8 : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d854a268d066409bb0f48e354105ef33 : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0b8259ba3b2d45dc8d64037f544019c0 : SUCCESS in 6m 27s (non-voting)\n- refstack-tox-functional https://zuul.opendev.org/t/openstack/build/8370a7b047a0493d8a37ddf71860e481 : SUCCESS in 5m 13s\n- refstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5283a769dab64a0088a4c099ad286bf7 : FAILURE in 5m 01s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"521d1ebbb708000edebc1f200e8037d711807631","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-10-15 06:48:56.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"56f9426e0cfa14794f66e183fce6a43fc791d9d0","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-10-15 06:49:06.000000000","message":"Topic fix/docs removed","accounts_in_message":[],"_revision_number":6},{"id":"96647b93e8516ad5eee84be6997fa15bc8168380","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-15 07:04:26.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs-run-lint https://zuul.opendev.org/t/openstack/build/176848bb419f4ebd8c143287a49e5fe1 : SUCCESS in 5m 11s\n- nodejs-run-test-browser https://zuul.opendev.org/t/openstack/build/7caedd20382d49e19c0caceef2eb0081 : SUCCESS in 6m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f073a467c57941d79c075a74ac3302d0 : SUCCESS in 4m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e1ece3f3adde4185b1589ad38edb898b : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58dd94b4b00b43f2bba7602f55224dba : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6720f67012294e78821c5a7fb9934c3c : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8f06f32f5107427a81bc7d25e32829c1 : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6890cd4c4eb3443c923cb96eb42b1f9d : SUCCESS in 5m 19s (non-voting)\n- refstack-tox-functional https://zuul.opendev.org/t/openstack/build/1f395f7a69bf460fab062a6d227d6e3c : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":6},{"id":"d61371620e916cf356516a2d78bf84efee8f9e9a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-10-15 11:40:11.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"dab199dad11b93e8c0efb07f961584f7a8c01b9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-15 11:57:49.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs-run-lint https://zuul.opendev.org/t/openstack/build/97e1378149fd45888d68962301bb287f : SUCCESS in 5m 34s\n- nodejs-run-test-browser https://zuul.opendev.org/t/openstack/build/f37ad50bf29747c2b11d7e493fa9888c : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5d035a797e4427b8e643cc9b0263b9b : SUCCESS in 4m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/417f99f140cb4e9ea721f7de4741273e : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a57290e25c0c4005b6946b7fa13ea099 : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/742c675932cb4c55bffe653f92671c5b : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a112d3306eba4bf484dfdb452f89a1ed : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fd46a0946fe444559f36600404277878 : SUCCESS in 5m 26s (non-voting)\n- refstack-tox-functional https://zuul.opendev.org/t/openstack/build/b9073a29b4494c61a3441c5039dc6214 : SUCCESS in 6m 23s","accounts_in_message":[],"_revision_number":7},{"id":"988f67643fd28f19f6d5321f6c42c27b6f65f91f","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-10-16 18:06:10.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"786a34087c1bd5fad3ad848d66cdc0df95e3273a","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-10-16 18:07:41.000000000","message":"Patch Set 7: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"a0fd65423b806a6487c489657b02af67c48fbd2a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-16 18:08:05.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"1522cd349c4676a6d1a8659e9a4f9044de221044","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-16 18:18:40.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- nodejs-run-lint https://zuul.opendev.org/t/openstack/build/88eac9ac0d2b4e14b7679e76690767c5 : SUCCESS in 4m 31s\n- nodejs-run-test-browser https://zuul.opendev.org/t/openstack/build/a6203708c6b141c4899773ac061f0cce : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d73d5e1e33fc43309521b20bc05467ad : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bd80150f00a43249ae55e3d6bda2ecf : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4cad13bdf4b84ec5be02a188209fa617 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d591ad11691b4d2bac7f6cdfc23c4348 : SUCCESS in 4m 21s\n- refstack-tox-functional https://zuul.opendev.org/t/openstack/build/64f0130b8dc64458b73db320dc5c193a : SUCCESS in 4m 50s","accounts_in_message":[],"_revision_number":7},{"id":"d735810e2b2cf6df48014de221eea6168f335c8a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-16 18:18:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"dfc1bf4180a24943d06bd38435aead14ef5b8e6b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-16 18:22:26.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0e3e5f2f9644bd792c3b93091de17c3 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"1e6acbec852d2bc8af959303aafec61616039be9","revisions":{"549d219dffe16dfd56523af32229cf6d67570692":{"kind":"REWORK","_number":1,"created":"2021-10-07 18:45:47.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/65/813065/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack","ref":"refs/changes/65/813065/1","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/65/813065/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/65/813065/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/65/813065/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack refs/changes/65/813065/1"}}},"commit":{"parents":[{"commit":"5790634ff48dd00444c051e20b7fe24cc82fb350","subject":"Update documentation, add playbook for developers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/5790634ff48dd00444c051e20b7fe24cc82fb350"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2021-10-07 18:45:14.000000000","tz":0},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2021-10-07 18:45:14.000000000","tz":0},"subject":"Switch testing to Yoga testing runtime","message":"Switch testing to Yoga testing runtime\n\nUpating the tetsing template to Yoga testing runtime:\nhttps://governance.openstack.org/tc/reference/runtimes/yoga.html\n\nChange-Id: I6d64cb67529585b458b28ed4fe011d50d1f23d6d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/549d219dffe16dfd56523af32229cf6d67570692"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/549d219dffe16dfd56523af32229cf6d67570692"}]},"branch":"refs/heads/master"},"fb243faaaba4d94552b35fe3d707c0238e1c2746":{"kind":"REWORK","_number":2,"created":"2021-10-08 19:13:09.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/65/813065/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack","ref":"refs/changes/65/813065/2","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/65/813065/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/65/813065/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/65/813065/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack refs/changes/65/813065/2"}}},"commit":{"parents":[{"commit":"5790634ff48dd00444c051e20b7fe24cc82fb350","subject":"Update documentation, add playbook for developers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/5790634ff48dd00444c051e20b7fe24cc82fb350"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2021-10-07 18:45:14.000000000","tz":0},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2021-10-08 19:12:13.000000000","tz":0},"subject":"Switch testing to Yoga testing runtime","message":"Switch testing to Yoga testing runtime\n\nUpating the tetsing template to Yoga testing runtime:\nhttps://governance.openstack.org/tc/reference/runtimes/yoga.html\n\nChange-Id: I6d64cb67529585b458b28ed4fe011d50d1f23d6d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/fb243faaaba4d94552b35fe3d707c0238e1c2746"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/fb243faaaba4d94552b35fe3d707c0238e1c2746"}]},"branch":"refs/heads/master"},"6a970e95c72931d139cf942615ccb23b43fe2007":{"kind":"REWORK","_number":3,"created":"2021-10-11 06:57:48.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/65/813065/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack","ref":"refs/changes/65/813065/3","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/65/813065/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/65/813065/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/65/813065/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack refs/changes/65/813065/3"}}},"commit":{"parents":[{"commit":"5790634ff48dd00444c051e20b7fe24cc82fb350","subject":"Update documentation, add playbook for developers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/5790634ff48dd00444c051e20b7fe24cc82fb350"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2021-10-07 18:45:14.000000000","tz":0},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2021-10-11 06:56:36.000000000","tz":0},"subject":"Switch testing to Yoga testing runtime","message":"Switch testing to Yoga testing runtime\n\nUpating the tetsing template to Yoga testing runtime:\nhttps://governance.openstack.org/tc/reference/runtimes/yoga.html\n\nChange-Id: I6d64cb67529585b458b28ed4fe011d50d1f23d6d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/6a970e95c72931d139cf942615ccb23b43fe2007"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/6a970e95c72931d139cf942615ccb23b43fe2007"}]},"branch":"refs/heads/master"},"9bd742ff319bda7d1d4cec700f2ec49e0db39d57":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2021-10-14 07:09:49.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/65/813065/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack","ref":"refs/changes/65/813065/4","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/65/813065/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/65/813065/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/65/813065/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack refs/changes/65/813065/4"}}},"commit":{"parents":[{"commit":"5790634ff48dd00444c051e20b7fe24cc82fb350","subject":"Update documentation, add playbook for developers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/5790634ff48dd00444c051e20b7fe24cc82fb350"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2021-10-07 18:45:14.000000000","tz":0},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2021-10-14 07:09:49.000000000","tz":0},"subject":"Switch testing to Yoga testing runtime","message":"Switch testing to Yoga testing runtime\n\nUpdating the testing template to Yoga testing runtime:\nhttps://governance.openstack.org/tc/reference/runtimes/yoga.html\n\nChange-Id: I6d64cb67529585b458b28ed4fe011d50d1f23d6d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/9bd742ff319bda7d1d4cec700f2ec49e0db39d57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/9bd742ff319bda7d1d4cec700f2ec49e0db39d57"}]},"branch":"refs/heads/master","description":"Edit commit message"},"4be9baae2e0d0c62239864dd490a28ca4aa22cbd":{"kind":"NO_CHANGE","_number":5,"created":"2021-10-14 13:11:51.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/65/813065/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack","ref":"refs/changes/65/813065/5","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/65/813065/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/65/813065/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/65/813065/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack refs/changes/65/813065/5"}}},"commit":{"parents":[{"commit":"5790634ff48dd00444c051e20b7fe24cc82fb350","subject":"Update documentation, add playbook for developers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/5790634ff48dd00444c051e20b7fe24cc82fb350"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2021-10-07 18:45:14.000000000","tz":0},"committer":{"name":"lpiwowar","email":"lpiwowar@redhat.com","date":"2021-10-14 13:08:34.000000000","tz":120},"subject":"Switch testing to Yoga testing runtime","message":"Switch testing to Yoga testing runtime\n\nUpdating the testing template to Yoga testing runtime:\nhttps://governance.openstack.org/tc/reference/runtimes/yoga.html\n\nChange-Id: I6d64cb67529585b458b28ed4fe011d50d1f23d6d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/4be9baae2e0d0c62239864dd490a28ca4aa22cbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/4be9baae2e0d0c62239864dd490a28ca4aa22cbd"}]},"branch":"refs/heads/master"},"b05ec0515b29b16b8cb2c1b863fd81a0c1f96c55":{"kind":"REWORK","_number":6,"created":"2021-10-15 06:48:56.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/65/813065/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack","ref":"refs/changes/65/813065/6","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/65/813065/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/65/813065/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/65/813065/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack refs/changes/65/813065/6"}}},"commit":{"parents":[{"commit":"5790634ff48dd00444c051e20b7fe24cc82fb350","subject":"Update documentation, add playbook for developers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/5790634ff48dd00444c051e20b7fe24cc82fb350"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2021-10-07 18:45:14.000000000","tz":0},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2021-10-15 06:47:19.000000000","tz":0},"subject":"Switch testing to Yoga testing runtime","message":"Switch testing to Yoga testing runtime\n\nUpdating the testing template to Yoga testing runtime:\nhttps://governance.openstack.org/tc/reference/runtimes/yoga.html\n\nThe commmit also removes refstack-tox-pip-check-reqs job due to a\nbug in the pip-check-reqs tool:\nhttps://github.com/r1chardj0n3s/pip-check-reqs/issues/70\n\nChange-Id: I6d64cb67529585b458b28ed4fe011d50d1f23d6d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/b05ec0515b29b16b8cb2c1b863fd81a0c1f96c55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/b05ec0515b29b16b8cb2c1b863fd81a0c1f96c55"}]},"branch":"refs/heads/master"},"1e6acbec852d2bc8af959303aafec61616039be9":{"kind":"REWORK","_number":7,"created":"2021-10-15 11:40:11.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/65/813065/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack","ref":"refs/changes/65/813065/7","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/65/813065/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/65/813065/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/65/813065/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack refs/changes/65/813065/7"}}},"commit":{"parents":[{"commit":"5790634ff48dd00444c051e20b7fe24cc82fb350","subject":"Update documentation, add playbook for developers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/5790634ff48dd00444c051e20b7fe24cc82fb350"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2021-10-07 18:45:14.000000000","tz":0},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2021-10-15 11:38:07.000000000","tz":0},"subject":"Switch testing to Yoga testing runtime","message":"Switch testing to Yoga testing runtime\n\nUpdating the testing template to Yoga testing runtime:\nhttps://governance.openstack.org/tc/reference/runtimes/yoga.html\n\nThe commmit also removes refstack-tox-pip-check-reqs job due to a\nbug in the pip-check-reqs tool:\nhttps://github.com/r1chardj0n3s/pip-check-reqs/issues/70\nand the fact the tool is not used in the gates by openstack\nprojects - still can be executed manualy though.\n\nChange-Id: I6d64cb67529585b458b28ed4fe011d50d1f23d6d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/1e6acbec852d2bc8af959303aafec61616039be9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/1e6acbec852d2bc8af959303aafec61616039be9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
