)]}'
{"id":"openinfra%2Frefstack~813597","triplet_id":"openinfra%2Frefstack~master~I4ae3a0274b1f3201a44bbac516c00f16dec5ce55","project":"openinfra/refstack","branch":"master","topic":"fix/docs","attention_set":{},"removed_from_attention_set":{"30674":{"account":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"last_update":"2021-10-19 02:44:27.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2021-10-19 02:44:27.000000000","reason":"Change was submitted"},"7160":{"account":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"last_update":"2021-10-19 02:33:38.000000000","reason":"arkady kanevsky replied on the change"},"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2021-10-19 02:44:27.000000000","reason":"Change was submitted"},"22873":{"account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"last_update":"2021-10-18 15:30:26.000000000","reason":"Martin Kopec replied on the change"}},"hashtags":[],"change_id":"I4ae3a0274b1f3201a44bbac516c00f16dec5ce55","subject":"Fix documentation structure","status":"MERGED","created":"2021-10-12 09:01:18.000000000","updated":"2021-10-19 02:48:53.000000000","submitted":"2021-10-19 02:44:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"813597-fix/docs","meta_rev_id":"e6eb9f976943ebdbed92e8b6a18c592c462a70b7","_number":813597,"virtual_id_number":813597,"owner":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-19 02:44:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"all":[{"value":2,"date":"2021-10-19 02:33:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"value":2,"date":"2021-10-18 15:30:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"all":[{"value":1,"date":"2021-10-19 02:33:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}],"reviewers":{"REVIEWER":[{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-12 09:12:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-12 14:36:41.000000000","updated_by":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2021-10-18 15:30:26.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2021-10-18 15:30:26.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"state":"REVIEWER"},{"updated":"2021-10-18 15:30:26.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"}],"messages":[{"id":"4e8251e48121c6c1f30609723ecf2399771f7361","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2021-10-12 09:01:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d57402d2493fe9d2612af04b860d74ec50aa7181","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 09:12:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- nodejs-run-lint https://zuul.opendev.org/t/openstack/build/e4c1f6b9c4fd4f44a8900e5d5208687e : SUCCESS in 4m 40s\n- nodejs-run-test-browser https://zuul.opendev.org/t/openstack/build/bae17b0ac86e450fb1bf4a1483170807 : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7e621e8facc41658e927fe10988cf3e : SUCCESS in 4m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f1a35f13cbf3418faea48dec89c13bab : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48d2aef451674136be0525d963eda70e : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fbf210b9eb014e179c83b3a34e4cd1b4 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/53250f60b42d4376acde2edc822c752e : SUCCESS in 5m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/56a89359de0a4fb6872e0abca2c3554d : SUCCESS in 5m 19s (non-voting)\n- refstack-tox-functional https://zuul.opendev.org/t/openstack/build/c93c0017ead54ca1a584e7c15f44809b : SUCCESS in 5m 07s\n- refstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e60805c1a0d644a69c17256ccf9e4962 : FAILURE in 4m 32s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/64a663b35164429fa7e5f726dbe80975 : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":1},{"id":"cffb152a27f1ec799c1b1cfbfdf04048827178c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2021-10-12 10:39:35.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"f52f5dba75c66b98d830aee1d025ad2e84dd0f1f","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2021-10-12 10:40:01.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":2},{"id":"7d3ef394474298b27666a8971e821d291d737840","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 10:50:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- nodejs-run-lint https://zuul.opendev.org/t/openstack/build/44c2dea4d8eb46828543db0140f79a18 : SUCCESS in 4m 41s\n- nodejs-run-test-browser https://zuul.opendev.org/t/openstack/build/985a85bdaa0e4d80a4da47939f66b1ba : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7e8c97bbd684a568df9c587392a6d9d : SUCCESS in 6m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/42111ece69234418a288ea0d2092e2d5 : SUCCESS in 7m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/820a0c7d309c4e23a69cd80f7da08ce7 : SUCCESS in 6m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f456df47f6c448268310e8d3b4ecfdb7 : SUCCESS in 7m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dec87ca742394ba2b21536e5a38ceb4a : SUCCESS in 7m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/696940650a864a1a95f9e4ba9ecb3ba4 : SUCCESS in 7m 22s (non-voting)\n- refstack-tox-functional https://zuul.opendev.org/t/openstack/build/8c8ed8c99e0743baa4762f07250b03db : SUCCESS in 7m 27s\n- refstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5abd041d9c64484bb698b4bf635d53d8 : FAILURE in 4m 20s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/3b3a1f1ebecc453ba65c0806e71c71bb : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":2},{"id":"ede74c2c1280db4b43195b719a5f36c4a01fa6a2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2021-10-12 11:12:58.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6beb893cbba6d9996e55ce5e38a6c84c50a61648","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 11:23:28.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- nodejs-run-lint https://zuul.opendev.org/t/openstack/build/24f9b0574ee2486286849df90360e106 : SUCCESS in 4m 58s\n- nodejs-run-test-browser https://zuul.opendev.org/t/openstack/build/3da3575722294a83b25e66293503dc9f : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71a0c2f4a0f84725a582aa6d119b7d0e : SUCCESS in 4m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8da2fafe98ea4097a2be74d8791ab1e8 : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2711c105076042f0b2bf01cf49a6a82c : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb456078c4aa46dca28d1745f5941916 : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/581096bdcb8c46429bb112496d679186 : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fc6b2aa7376445d1acdf32799b2db944 : SUCCESS in 4m 37s (non-voting)\n- refstack-tox-functional https://zuul.opendev.org/t/openstack/build/af59ab1a8ed541fc9f7cb9de0af2ef09 : SUCCESS in 5m 06s\n- refstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c235926239e94c69877a4eb9ae78faaf : FAILURE in 4m 50s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/fb90cc970c044f9ab6f3f6f12f80d0fe : SUCCESS in 4m 10s","accounts_in_message":[],"_revision_number":3},{"id":"aa6057b226bc3a66329a15bb03f269d7aecffe71","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2021-10-12 14:29:33.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"76c7f9d4eb7d1216bf1ace6787be7a3d0dbc1a84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 14:30:09.000000000","message":"Patch Set 4: Verified-1\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":4},{"id":"28ab0b353d39904273ff4bc5be4fd29963eae9ae","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2021-10-12 14:33:28.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"860451288189ac132429c27fabbd33ec94cfdfa7","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2021-10-12 14:36:41.000000000","message":"Patch Set 5:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":5},{"id":"384f49353cbe15a688f31a7565369c9b9aceac4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 14:47:54.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- nodejs-run-lint https://zuul.opendev.org/t/openstack/build/239128d67d964201a25fec09bfa12070 : SUCCESS in 4m 54s\n- nodejs-run-test-browser https://zuul.opendev.org/t/openstack/build/4f2c5011929f48239b59b40b37e16d06 : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92da6f6055da4614b7670c3d5142499b : SUCCESS in 4m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bbcdc1596b5240c08d02caf5920377ed : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50860ed1874348b1b634464747cccb12 : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dab777e98c3b429bb61af4075387b442 : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/412e581540494982a37030141432a099 : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/41dddc729b4341cebe1eb45983afb5d1 : SUCCESS in 5m 17s (non-voting)\n- refstack-tox-functional https://zuul.opendev.org/t/openstack/build/0d640cf4e96547edb1ccf1fe114c01bc : SUCCESS in 5m 25s\n- refstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6e6168ba1dc8435584320e9d3b336008 : FAILURE in 4m 51s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/99dee82726f34df79cfa85eba9a718ad : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":5},{"id":"a83605f2ab1050ac010681cd2aa680ec9bac0db5","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-10-13 12:03:12.000000000","message":"Patch Set 5: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"c15c099f39d85e97cda989201be384d2cb4274f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2021-10-14 11:22:44.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"e5d8110979fb68749a390b3b1c040e896b5e9f51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-14 11:47:57.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- nodejs-run-lint https://zuul.opendev.org/t/openstack/build/d9bc695dcd8546239b8ad8fe2bee24a5 : SUCCESS in 5m 15s\n- nodejs-run-test-browser https://zuul.opendev.org/t/openstack/build/c80671638a6d4050ae0811ff711da804 : SUCCESS in 6m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0d5d9c8df0fd4864823fe91e3b9d5a9d : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bf2cbaf710b4f0eac2e5f4525fec248 : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/58c4db775a944899822d5e472b4b52f0 : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e9a91054c61f408f8813b8a465c5220d : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/84589f1db3ea48aeb8418f7b162558f6 : SUCCESS in 5m 19s (non-voting)\n- refstack-tox-functional https://zuul.opendev.org/t/openstack/build/5079d4abc91143428d9a84da60a03d76 : SUCCESS in 6m 31s\n- refstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/cc62bcd5974045fa9aa8718ea609c3f4 : FAILURE in 5m 43s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/44295fee27fb4069bec9f89678cc4146 : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":6},{"id":"6bb4dd5b5889fdd273261d2b179bdb01d0982339","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2021-10-14 12:11:24.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"a80eabea251ad09cd4193f9f66d3af78fc4828c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2021-10-14 12:13:05.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"9dd212f0a65d708e94d3ae4ac5fff68aa62617df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-14 12:31:18.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- nodejs-run-lint https://zuul.opendev.org/t/openstack/build/7e6b88c2eb80422ea4f40e248567851e : SUCCESS in 6m 08s\n- nodejs-run-test-browser https://zuul.opendev.org/t/openstack/build/4f4a32c151064f129d0a1f3b9d5648bf : SUCCESS in 6m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/211d870b0deb4180aa2d9a589af9684c : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90e8906181d84a87856cee0a802f3fb0 : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ba8708bb2e6e4d21b769ee6789f5280f : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c2d96c6fc9e4c18b364009416eecf1b : SUCCESS in 5m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dd0c4737a2494e9f8d4736a3b1951370 : SUCCESS in 6m 09s (non-voting)\n- refstack-tox-functional https://zuul.opendev.org/t/openstack/build/332fc413d98a4dfbbbf7abcd9fa8d55e : SUCCESS in 6m 00s\n- refstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/936d446e2d9f4c74bdfeacfd468a7178 : FAILURE in 5m 58s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/f798f6f3ba5d4a78950cfb0b5f56c2ac : SUCCESS in 6m 05s","accounts_in_message":[],"_revision_number":8},{"id":"314d8501c12807ff5303ba0d550f60b590b21c8e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2021-10-14 13:11:51.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"08191a3ab7295763c21b0fe737937cec4b656d62","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2021-10-14 13:14:11.000000000","message":"Uploaded patch set 10: Commit message was updated.","accounts_in_message":[],"_revision_number":10},{"id":"39360ee1460778a9e58de1791a96df058d6ca9f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-14 13:25:56.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs-run-lint https://zuul.opendev.org/t/openstack/build/508b39e8c8164f93949c99fe224bf345 : SUCCESS in 5m 07s\n- nodejs-run-test-browser https://zuul.opendev.org/t/openstack/build/fd706616bec04eddb4b9efdd35ac23eb : SUCCESS in 8m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0bf77886f4e647489419515ac706660e : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/466e9a5795414da2b0edaa440f07631b : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7eea31cd512e49a490b9baee15a91d1a : SUCCESS in 6m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b184f23a66f425eaf87a7c1d8c9183e : SUCCESS in 5m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f61f66faf3e64e6d80f0cf8e62f40d70 : SUCCESS in 5m 25s (non-voting)\n- refstack-tox-functional https://zuul.opendev.org/t/openstack/build/29f12812ab7941c4932b2e0af107a419 : SUCCESS in 5m 41s\n- refstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9bdadd8cb2314c558fdfb670c62ea228 : FAILURE in 5m 48s (non-voting)\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/347e139071fd469f9fdc7f32288f982a : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":10},{"id":"e864fb4df16ce9d722d1e861cc57aeded00828c8","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-10-15 08:24:24.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"7b64e7fc09391a60a4336c9893babf28a81da566","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-10-15 08:24:29.000000000","message":"Patch Set 10: Code-Review-1","accounts_in_message":[],"_revision_number":10},{"id":"c7cd038181c4a6cf4536b07ec30153360e3a31f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2021-10-18 09:29:26.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"a800ec10e45bbb6ea968cfa72df85d6b0087ec08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-18 09:38:17.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs-run-lint https://zuul.opendev.org/t/openstack/build/7f24b056ff894f67bafd6fb39f73e0c7 : SUCCESS in 5m 25s\n- nodejs-run-test-browser https://zuul.opendev.org/t/openstack/build/540dbd7395b74ae49efd6b1b152b67af : SUCCESS in 6m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/36dbcbaadcb24ae2bc74435d4dd3df23 : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85b83fd708874079ae92c0a43704a214 : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa9182e8616f4ed8acda039c70d53ebe : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cfcf8788ebf947039bcb78cd47e1a768 : SUCCESS in 5m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/736c01c3c5cd4dae98723041be193df7 : SUCCESS in 5m 43s (non-voting)\n- refstack-tox-functional https://zuul.opendev.org/t/openstack/build/579e8813a5004cfdb05289160397245b : SUCCESS in 6m 36s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/1f03d4b20eef4ae7b394b43fe6f5c9bf : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":11},{"id":"67459ad077ce8470728136ba81bad6cc54efd164","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2021-10-18 09:39:55.000000000","message":"Patch Set 11:\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"8b17b7d4670c35f81934a26dcfe1e3384b4e5755","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-10-18 15:30:26.000000000","message":"Patch Set 11: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"5bfe9836519bef773359dec27c4a202ec3e85f18","author":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"date":"2021-10-19 02:33:38.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"5d5e5459c4ec67341caf8ce80cc193b09e54be28","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-19 02:33:59.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"a1e68833376a11478033facc5e4d14ccc048b3b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-19 02:44:26.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- nodejs-run-lint https://zuul.opendev.org/t/openstack/build/a9be6cc1c67840df8d9f0e25976e8472 : SUCCESS in 4m 42s\n- nodejs-run-test-browser https://zuul.opendev.org/t/openstack/build/6bff9be31e0f46ce893bf4d6ac676f20 : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7074bc349b0f42ceb286ea4a7d55179a : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/991acd4b85f846f0bbf543d4eb361ba5 : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/da5c092a3d01427d85b2bc5bd9152c67 : SUCCESS in 4m 48s\n- refstack-tox-functional https://zuul.opendev.org/t/openstack/build/621ae04d451e4c75bfd27ca139990568 : SUCCESS in 4m 58s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/f4a8948228a04edabced33e27e19e706 : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":11},{"id":"0df8542fe0920ed1d93245215ea811a8e94d43c9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-19 02:44:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":11},{"id":"e6eb9f976943ebdbed92e8b6a18c592c462a70b7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-19 02:48:53.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/241896fcc8de44d98a17787f9e5c80ae : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"42df3599900b9a98b7e7aa8b623424919e6c321f","revisions":{"9943eb3e4003fafa4b70bb47b9561efca95eafd1":{"kind":"REWORK","_number":1,"created":"2021-10-12 09:01:18.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/97/813597/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack","ref":"refs/changes/97/813597/1","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/97/813597/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/97/813597/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/97/813597/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack refs/changes/97/813597/1"}}},"commit":{"parents":[{"commit":"5790634ff48dd00444c051e20b7fe24cc82fb350","subject":"Update documentation, add playbook for developers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/5790634ff48dd00444c051e20b7fe24cc82fb350"}]}],"author":{"name":"lpiwowar","email":"lpiwowar@redhat.com","date":"2021-10-05 14:33:03.000000000","tz":120},"committer":{"name":"lpiwowar","email":"lpiwowar@redhat.com","date":"2021-10-12 09:00:53.000000000","tz":120},"subject":"[WIP] - Documentation fix","message":"[WIP] - Documentation fix\n\nChange-Id: I4ae3a0274b1f3201a44bbac516c00f16dec5ce55\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/9943eb3e4003fafa4b70bb47b9561efca95eafd1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/9943eb3e4003fafa4b70bb47b9561efca95eafd1"}]},"branch":"refs/heads/master"},"50dea7ba3d8083c70703deb72655a22316265362":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-10-12 10:39:35.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/97/813597/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack","ref":"refs/changes/97/813597/2","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/97/813597/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/97/813597/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/97/813597/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack refs/changes/97/813597/2"}}},"commit":{"parents":[{"commit":"e4494b52faa3e917de57b60c1168c6ac1bb85ccd","subject":"[WIP] - Add docs job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/e4494b52faa3e917de57b60c1168c6ac1bb85ccd"}]}],"author":{"name":"lpiwowar","email":"lpiwowar@redhat.com","date":"2021-10-05 14:33:03.000000000","tz":120},"committer":{"name":"lpiwowar","email":"lpiwowar@redhat.com","date":"2021-10-12 10:39:12.000000000","tz":120},"subject":"[WIP] - Documentation fix","message":"[WIP] - Documentation fix\n\nChange-Id: I4ae3a0274b1f3201a44bbac516c00f16dec5ce55\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/50dea7ba3d8083c70703deb72655a22316265362"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/50dea7ba3d8083c70703deb72655a22316265362"}]},"branch":"refs/heads/master"},"226ad4c1265f4978d373a340e784d312468f17d7":{"kind":"REWORK","_number":3,"created":"2021-10-12 11:12:58.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/97/813597/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack","ref":"refs/changes/97/813597/3","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/97/813597/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/97/813597/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/97/813597/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack refs/changes/97/813597/3"}}},"commit":{"parents":[{"commit":"f6a781374b5b30d5688649cab20ab4242a4475ad","subject":"Publish docs to docs.opendev.org","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/f6a781374b5b30d5688649cab20ab4242a4475ad"}]}],"author":{"name":"lpiwowar","email":"lpiwowar@redhat.com","date":"2021-10-05 14:33:03.000000000","tz":120},"committer":{"name":"lpiwowar","email":"lpiwowar@redhat.com","date":"2021-10-12 11:12:14.000000000","tz":120},"subject":"Fix documentation structure","message":"Fix documentation structure\n\nThis patch fixes the structure of the Refstack documentation and\ntries to make it more readable.\n\nChange-Id: I4ae3a0274b1f3201a44bbac516c00f16dec5ce55\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/226ad4c1265f4978d373a340e784d312468f17d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/226ad4c1265f4978d373a340e784d312468f17d7"}]},"branch":"refs/heads/master"},"56a264230839504ffacd70b7ee95e83f70b0b99c":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-10-12 14:29:33.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/97/813597/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack","ref":"refs/changes/97/813597/4","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/97/813597/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/97/813597/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/97/813597/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack refs/changes/97/813597/4"}}},"commit":{"parents":[{"commit":"5de5d602f2c816caeae046575e9f6ed418798da4","subject":"Publish docs to docs.opendev.org","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/5de5d602f2c816caeae046575e9f6ed418798da4"}]}],"author":{"name":"lpiwowar","email":"lpiwowar@redhat.com","date":"2021-10-05 14:33:03.000000000","tz":120},"committer":{"name":"lpiwowar","email":"lpiwowar@redhat.com","date":"2021-10-12 14:28:14.000000000","tz":120},"subject":"Fix documentation structure","message":"Fix documentation structure\n\nThis patch fixes the structure of the Refstack documentation and\ntries to make it more readable.\n\nChange-Id: I4ae3a0274b1f3201a44bbac516c00f16dec5ce55\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/56a264230839504ffacd70b7ee95e83f70b0b99c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/56a264230839504ffacd70b7ee95e83f70b0b99c"}]},"branch":"refs/heads/master"},"9ead087c5c54ccaf5f05ff8fa4479d9ffbf1b8cd":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-10-12 14:33:28.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/97/813597/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack","ref":"refs/changes/97/813597/5","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/97/813597/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/97/813597/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/97/813597/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack refs/changes/97/813597/5"}}},"commit":{"parents":[{"commit":"8422577f6664d3b645bdce2c1447a10399643121","subject":"Publish docs to docs.opendev.org","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/8422577f6664d3b645bdce2c1447a10399643121"}]}],"author":{"name":"lpiwowar","email":"lpiwowar@redhat.com","date":"2021-10-05 14:33:03.000000000","tz":120},"committer":{"name":"lpiwowar","email":"lpiwowar@redhat.com","date":"2021-10-12 14:32:50.000000000","tz":120},"subject":"Fix documentation structure","message":"Fix documentation structure\n\nThis patch fixes the structure of the Refstack documentation and\ntries to make it more readable.\n\nChange-Id: I4ae3a0274b1f3201a44bbac516c00f16dec5ce55\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/9ead087c5c54ccaf5f05ff8fa4479d9ffbf1b8cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/9ead087c5c54ccaf5f05ff8fa4479d9ffbf1b8cd"}]},"branch":"refs/heads/master"},"7a6d1c5e681c30120a668edd0d0212daf6086db1":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2021-10-14 11:22:44.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/97/813597/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack","ref":"refs/changes/97/813597/6","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/97/813597/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/97/813597/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/97/813597/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack refs/changes/97/813597/6"}}},"commit":{"parents":[{"commit":"bb880eced0beae129becf1165a6a38175e4d79ef","subject":"Publish docs to docs.opendev.org","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/bb880eced0beae129becf1165a6a38175e4d79ef"}]}],"author":{"name":"lpiwowar","email":"lpiwowar@redhat.com","date":"2021-10-05 14:33:03.000000000","tz":120},"committer":{"name":"lpiwowar","email":"lpiwowar@redhat.com","date":"2021-10-14 11:21:07.000000000","tz":120},"subject":"Fix documentation structure","message":"Fix documentation structure\n\nThis patch fixes the structure of the Refstack documentation and\ntries to make it more readable.\n\nChange-Id: I4ae3a0274b1f3201a44bbac516c00f16dec5ce55\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/7a6d1c5e681c30120a668edd0d0212daf6086db1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/7a6d1c5e681c30120a668edd0d0212daf6086db1"}]},"branch":"refs/heads/master"},"080665d285e5957a6547a6932c27944e296b1419":{"kind":"REWORK","_number":7,"created":"2021-10-14 12:11:24.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/97/813597/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack","ref":"refs/changes/97/813597/7","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/97/813597/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/97/813597/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/97/813597/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack refs/changes/97/813597/7"}}},"commit":{"parents":[{"commit":"bb880eced0beae129becf1165a6a38175e4d79ef","subject":"Publish docs to docs.opendev.org","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/bb880eced0beae129becf1165a6a38175e4d79ef"}]}],"author":{"name":"lpiwowar","email":"lpiwowar@redhat.com","date":"2021-10-05 14:33:03.000000000","tz":120},"committer":{"name":"lpiwowar","email":"lpiwowar@redhat.com","date":"2021-10-14 12:09:31.000000000","tz":120},"subject":"Fix documentation structure","message":"Fix documentation structure\n\nThis patch fixes the structure of the Refstack documentation and\ntries to make it more readable.\n\nChange-Id: I4ae3a0274b1f3201a44bbac516c00f16dec5ce55\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/080665d285e5957a6547a6932c27944e296b1419"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/080665d285e5957a6547a6932c27944e296b1419"}]},"branch":"refs/heads/master"},"83a67154628aa835d266288aa736c7c69b7e8a5f":{"kind":"REWORK","_number":8,"created":"2021-10-14 12:13:05.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/97/813597/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack","ref":"refs/changes/97/813597/8","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/97/813597/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/97/813597/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/97/813597/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack refs/changes/97/813597/8"}}},"commit":{"parents":[{"commit":"bb880eced0beae129becf1165a6a38175e4d79ef","subject":"Publish docs to docs.opendev.org","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/bb880eced0beae129becf1165a6a38175e4d79ef"}]}],"author":{"name":"lpiwowar","email":"lpiwowar@redhat.com","date":"2021-10-05 14:33:03.000000000","tz":120},"committer":{"name":"lpiwowar","email":"lpiwowar@redhat.com","date":"2021-10-14 12:12:55.000000000","tz":120},"subject":"Fix documentation structure","message":"Fix documentation structure\n\nThis patch fixes the structure of the Refstack documentation and\ntries to make it more readable.\n\nChange-Id: I4ae3a0274b1f3201a44bbac516c00f16dec5ce55\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/83a67154628aa835d266288aa736c7c69b7e8a5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/83a67154628aa835d266288aa736c7c69b7e8a5f"}]},"branch":"refs/heads/master"},"08a4be07db537b8eaf7b597f4974fd2c676254a5":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":9,"created":"2021-10-14 13:11:51.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/97/813597/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack","ref":"refs/changes/97/813597/9","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/97/813597/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/97/813597/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/97/813597/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack refs/changes/97/813597/9"}}},"commit":{"parents":[{"commit":"18c07ff73ec346aeee2dc6b351a8eb704c891154","subject":"Publish docs to docs.opendev.org","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/18c07ff73ec346aeee2dc6b351a8eb704c891154"}]}],"author":{"name":"lpiwowar","email":"lpiwowar@redhat.com","date":"2021-10-05 14:33:03.000000000","tz":120},"committer":{"name":"lpiwowar","email":"lpiwowar@redhat.com","date":"2021-10-14 13:10:41.000000000","tz":120},"subject":"Fix documentation structure","message":"Fix documentation structure\n\nThis patch fixes the structure of the Refstack documentation and\ntries to make it more readable.\n\nChange-Id: I4ae3a0274b1f3201a44bbac516c00f16dec5ce55\n(cherry picked from commit 83a67154628aa835d266288aa736c7c69b7e8a5f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/08a4be07db537b8eaf7b597f4974fd2c676254a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/08a4be07db537b8eaf7b597f4974fd2c676254a5"}]},"branch":"refs/heads/master"},"c872804224a8eb2af57680f6ab51d2094fc60c7b":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2021-10-14 13:14:11.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/97/813597/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack","ref":"refs/changes/97/813597/10","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/97/813597/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/97/813597/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/97/813597/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack refs/changes/97/813597/10"}}},"commit":{"parents":[{"commit":"ae207a698b63d036514c416f1e9d68243ed1a03a","subject":"Publish docs to docs.opendev.org","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/ae207a698b63d036514c416f1e9d68243ed1a03a"}]}],"author":{"name":"lpiwowar","email":"lpiwowar@redhat.com","date":"2021-10-05 14:33:03.000000000","tz":120},"committer":{"name":"lpiwowar","email":"lpiwowar@redhat.com","date":"2021-10-14 13:13:36.000000000","tz":120},"subject":"Fix documentation structure","message":"Fix documentation structure\n\nThis patch fixes the structure of the Refstack documentation and\ntries to make it more readable.\n\nChange-Id: I4ae3a0274b1f3201a44bbac516c00f16dec5ce55\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/c872804224a8eb2af57680f6ab51d2094fc60c7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/c872804224a8eb2af57680f6ab51d2094fc60c7b"}]},"branch":"refs/heads/master"},"42df3599900b9a98b7e7aa8b623424919e6c321f":{"kind":"REWORK","_number":11,"created":"2021-10-18 09:29:26.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/97/813597/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack","ref":"refs/changes/97/813597/11","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/97/813597/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/97/813597/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/97/813597/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack refs/changes/97/813597/11"}}},"commit":{"parents":[{"commit":"7bbd4efa371cbd0b0183f45de594d0333c11ec16","subject":"Publish docs to docs.opendev.org","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/7bbd4efa371cbd0b0183f45de594d0333c11ec16"}]}],"author":{"name":"lpiwowar","email":"lpiwowar@redhat.com","date":"2021-10-05 14:33:03.000000000","tz":120},"committer":{"name":"lpiwowar","email":"lpiwowar@redhat.com","date":"2021-10-18 09:27:11.000000000","tz":120},"subject":"Fix documentation structure","message":"Fix documentation structure\n\nThis patch fixes the structure of the Refstack documentation and\ntries to make it more readable.\n\nChange-Id: I4ae3a0274b1f3201a44bbac516c00f16dec5ce55\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/42df3599900b9a98b7e7aa8b623424919e6c321f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/42df3599900b9a98b7e7aa8b623424919e6c321f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
