)]}'
{"id":"openinfra%2Frefstack~814273","triplet_id":"openinfra%2Frefstack~master~Ifebeb374c17223b4b69ff06956dcb849bfe11267","project":"openinfra/refstack","branch":"master","topic":"project-rename","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2021-10-18 09:08:52.000000000","reason":"Change was submitted"},"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2021-10-18 09:08:52.000000000","reason":"Change was submitted"},"22873":{"account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"last_update":"2021-10-18 08:48:46.000000000","reason":"Martin Kopec replied on the change"}},"hashtags":[],"change_id":"Ifebeb374c17223b4b69ff06956dcb849bfe11267","subject":"[project-rename] Update .gitreview","status":"MERGED","created":"2021-10-16 17:52:33.000000000","updated":"2021-10-18 09:10:22.000000000","submitted":"2021-10-18 09:08:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"814273-project-rename","meta_rev_id":"4b670f52faa92631c0b85c803f8dcefb7bc66f86","_number":814273,"virtual_id_number":814273,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-18 09:08:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"all":[{"value":2,"date":"2021-10-16 19:43:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"value":2,"date":"2021-10-18 08:48:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":0,"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"value":1,"date":"2021-10-18 08:48:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}],"reviewers":{"REVIEWER":[{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-16 18:04:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-16 19:43:24.000000000","updated_by":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"reviewer":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"state":"REVIEWER"},{"updated":"2021-10-16 19:43:38.000000000","updated_by":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2021-10-16 19:43:38.000000000","updated_by":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"}],"messages":[{"id":"85fc80bac5e4fb922ce7cec578f96da3cab93cee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-10-16 17:52:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2e2016adac98c95e5ddb409127a33f0963bfa318","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-16 18:04:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- nodejs-run-lint https://zuul.opendev.org/t/openstack/build/1ac8b7986b5d436fba305c643948383c : SUCCESS in 4m 49s\n- nodejs-run-test-browser https://zuul.opendev.org/t/openstack/build/b524de26994e4abb80a4d3d8bdc8b185 : SUCCESS in 6m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36b3e7b064e64b4998e8530464657a60 : SUCCESS in 4m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1302b0f52797442daaca751dc7cf8c6f : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/173d355a0bc94333ba74d3d1fe23346b : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e80867a7fd16454881590b92c0061ff4 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/feb7f0e280cd4546946727e32df43888 : SUCCESS in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ba65cce0620c4d64beca23a5d096448e : SUCCESS in 5m 26s (non-voting)\n- refstack-tox-functional https://zuul.opendev.org/t/openstack/build/561d9be1146f4c76a7eef3494eba8b59 : SUCCESS in 5m 13s\n- refstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a67dc3a423bc48b98db77bde07bb938c : FAILURE in 5m 11s","accounts_in_message":[],"_revision_number":1},{"id":"fa7be4b091a7ff696f3a56c124b06e18b8e871cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-10-16 18:07:56.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"12aa8b01527d8e1a45059749d0a39a0a12c7f3a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-16 18:18:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs-run-lint https://zuul.opendev.org/t/openstack/build/9a3f1d8036bc40f3b9e449d14e67edee : SUCCESS in 4m 15s\n- nodejs-run-test-browser https://zuul.opendev.org/t/openstack/build/d771df7b20704651b2aeb4e58240d825 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d238a97036e940ab9a95729da1edc1f2 : SUCCESS in 3m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3110cd61443441a6bbc2c035ff7c082a : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/951e4200c66b40efaa51a3d3810dde23 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/23443e47b0414570b56f25ef255c026a : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e7be0db07f3d4ead985610f75038314c : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/204e7189da104f0ebb63e4fae0417a59 : SUCCESS in 4m 22s (non-voting)\n- refstack-tox-functional https://zuul.opendev.org/t/openstack/build/2d7f626db3b749d9af038dcb6a1e1d38 : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":2},{"id":"a0f9fb9e341d0d830610aa029a683e399fbef0b1","author":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"date":"2021-10-16 19:43:24.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7c9227740953f43b1bdfc2b549ce2bbb601edc8b","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-10-18 08:48:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"eefa709c94c734c901f8214c285184adb5fc017a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-18 08:50:42.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"caa5282d9806e1589499d345c1cb16158086afe3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-18 09:08:52.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- nodejs-run-lint https://zuul.opendev.org/t/openstack/build/862382ceac90476e8778d9e549b2e2af : SUCCESS in 5m 18s\n- nodejs-run-test-browser https://zuul.opendev.org/t/openstack/build/02829994569742ed8cfd10919ee9c352 : SUCCESS in 6m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3240f64c0c34d46ab24df28d9faf25c : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37ea56831e35417aa5d877bcd53c827c : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/63687eac30c14405ad2ca6f9d76c9307 : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/798a27b65f6c481ba214c57112ce61a7 : SUCCESS in 4m 31s\n- refstack-tox-functional https://zuul.opendev.org/t/openstack/build/7c7b1c00661d47249390fe3cbb4242af : SUCCESS in 6m 23s","accounts_in_message":[],"_revision_number":2},{"id":"a7914953bc2784c6a4cec629d02c250b5a757c58","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-18 09:08:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"4b670f52faa92631c0b85c803f8dcefb7bc66f86","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-18 09:10:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd04b42014ba46979a639fd41f9f013c : SUCCESS in 58s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2f508df44d2f83c70b742ac53e7edc88102e2201","revisions":{"2921cfbc1390c8e8677e5e3f680ed328b4f3268c":{"kind":"REWORK","_number":1,"created":"2021-10-16 17:52:33.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/73/814273/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack","ref":"refs/changes/73/814273/1","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/73/814273/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/73/814273/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/73/814273/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack refs/changes/73/814273/1"}}},"commit":{"parents":[{"commit":"5790634ff48dd00444c051e20b7fe24cc82fb350","subject":"Update documentation, add playbook for developers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/5790634ff48dd00444c051e20b7fe24cc82fb350"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-10-16 17:51:56.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-10-16 17:51:56.000000000","tz":-420},"subject":"[project-rename] Update .gitreview","message":"[project-rename] Update .gitreview\n\nThis project was moved from the osf/ namespace\nto openinfra/.\n\nChange-Id: Ifebeb374c17223b4b69ff06956dcb849bfe11267\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/2921cfbc1390c8e8677e5e3f680ed328b4f3268c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/2921cfbc1390c8e8677e5e3f680ed328b4f3268c"}]},"branch":"refs/heads/master"},"2f508df44d2f83c70b742ac53e7edc88102e2201":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-10-16 18:07:56.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/73/814273/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack","ref":"refs/changes/73/814273/2","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/73/814273/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/73/814273/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/73/814273/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack refs/changes/73/814273/2"}}},"commit":{"parents":[{"commit":"1e6acbec852d2bc8af959303aafec61616039be9","subject":"Switch testing to Yoga testing runtime","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/1e6acbec852d2bc8af959303aafec61616039be9"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-10-16 17:51:56.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-10-16 18:07:56.000000000","tz":0},"subject":"[project-rename] Update .gitreview","message":"[project-rename] Update .gitreview\n\nThis project was moved from the osf/ namespace\nto openinfra/.\n\nChange-Id: Ifebeb374c17223b4b69ff06956dcb849bfe11267\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/2f508df44d2f83c70b742ac53e7edc88102e2201"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/2f508df44d2f83c70b742ac53e7edc88102e2201"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
