)]}'
{"id":"openinfra%2Frefstack~815417","triplet_id":"openinfra%2Frefstack~master~Ie70c85febbe723562f16633b6175ebc7a50925d8","project":"openinfra/refstack","branch":"master","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2021-10-26 17:11:35.000000000","reason":"Change was submitted"},"7160":{"account":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"last_update":"2021-10-26 17:11:35.000000000","reason":"Change was submitted"},"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2021-10-26 17:11:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie70c85febbe723562f16633b6175ebc7a50925d8","subject":"Cap jsonschema to 3.2.0","status":"MERGED","created":"2021-10-26 07:55:23.000000000","updated":"2021-10-26 17:13:11.000000000","submitted":"2021-10-26 17:11:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"815417","meta_rev_id":"5cdead65e3dcfd7adb0fa793b9169c2ddf5ba420","_number":815417,"virtual_id_number":815417,"owner":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-26 17:11:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":2,"date":"2021-10-26 17:00:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":1,"date":"2021-10-26 17:00:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}],"reviewers":{"REVIEWER":[{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-26 08:03:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-26 08:12:05.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2021-10-26 08:12:05.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2021-10-26 08:12:05.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"state":"REVIEWER"}],"messages":[{"id":"945427a49549a53cde5a3be8438189f0ea1f5b06","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-10-26 07:55:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"221d7a6437593f60f7944afe16b328f0b41b3135","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 08:03:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs-run-lint https://zuul.opendev.org/t/openstack/build/564668c5a2644515ad979531cb226b84 : SUCCESS in 6m 17s\n- nodejs-run-test-browser https://zuul.opendev.org/t/openstack/build/800aa80335ea44ab8aa07d178abf898b : SUCCESS in 6m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9eac424102d84453b78b7539bad40263 : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08d74dc3dd934618a20faf4bbd5c1f26 : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0236eea526134473b3203bab63d2617f : SUCCESS in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7fd90673772f451a84a9070dc7297fd4 : SUCCESS in 6m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a70584a005a340789bbcfa36d604478e : SUCCESS in 6m 22s (non-voting)\n- refstack-tox-functional https://zuul.opendev.org/t/openstack/build/0719d38d591c4573b23a86798557a1db : SUCCESS in 6m 31s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/681d73fd4703419a8bde07b22d010e77 : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":1},{"id":"8b9a8e88bb13e107e0cc77121d2cf168d3549d65","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-10-26 17:00:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b308497637af854403d870319e6eb81388514b09","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 17:01:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"44a10fbeb5e577f5e625177ebf2b86964e8a4e8d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 17:11:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- nodejs-run-lint https://zuul.opendev.org/t/openstack/build/9fd7fe37ad7c4799b963a418cfc0ae8d : SUCCESS in 5m 33s\n- nodejs-run-test-browser https://zuul.opendev.org/t/openstack/build/940df36a1e1d4cd5a476a8e91327965c : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6344afde8e348f5b4afcd6ce102b6ae : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b0c02696973b467fabbe8fb165c91c54 : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7bc41639aefb4a89975d5bacd328e46e : SUCCESS in 5m 07s\n- refstack-tox-functional https://zuul.opendev.org/t/openstack/build/834e607e0b6243c8b3150f4bc4fb1ff7 : SUCCESS in 4m 37s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/d0f8e19ca05d42fabc466517ffe18845 : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":1},{"id":"d867100abf91dd053efbaffa21a395af4eef2e3e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 17:11:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5cdead65e3dcfd7adb0fa793b9169c2ddf5ba420","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 17:13:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- opendev-promote-docs https://zuul.opendev.org/t/openstack/build/c11f3daaceaf4f1d8fadfd181beca75e : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bb0e985734690394e227d16af893b8e4ac8d3724","revisions":{"bb0e985734690394e227d16af893b8e4ac8d3724":{"kind":"REWORK","_number":1,"created":"2021-10-26 07:55:23.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/17/815417/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack","ref":"refs/changes/17/815417/1","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/17/815417/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/17/815417/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/17/815417/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack refs/changes/17/815417/1"}}},"commit":{"parents":[{"commit":"4d2c7db14dad84335b612a266b32b2410fa69399","subject":"Merge \"Reflect osf/ to openinfra/ namespace rename\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/4d2c7db14dad84335b612a266b32b2410fa69399"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2021-10-26 07:50:08.000000000","tz":120},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2021-10-26 07:53:15.000000000","tz":120},"subject":"Cap jsonschema to 3.2.0","message":"Cap jsonschema to 3.2.0\n\nrefstack fails with jsonschema\u003e3.2.0 on result schema validation,\nuntil we will have resolved that the commit caps the jsonschema to\nthe last working version - 3.2.0\n\nStory: 2009327\nTask: 43762\nChange-Id: Ie70c85febbe723562f16633b6175ebc7a50925d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/bb0e985734690394e227d16af893b8e4ac8d3724"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/bb0e985734690394e227d16af893b8e4ac8d3724"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
