)]}'
{"id":"openinfra%2Frefstack~824066","triplet_id":"openinfra%2Frefstack~master~I1d12ffe4cc9f6a9a489dd664481274d25dc40fac","project":"openinfra/refstack","branch":"master","attention_set":{},"removed_from_attention_set":{"22873":{"account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"last_update":"2022-01-14 00:21:37.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I1d12ffe4cc9f6a9a489dd664481274d25dc40fac","subject":"DNM - testing","status":"ABANDONED","created":"2022-01-10 21:35:06.000000000","updated":"2022-01-14 00:21:37.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"2021138933f0662b9712ce7ed2e56871cad138ef","_number":824066,"virtual_id_number":824066,"owner":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-01-11 12:52:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2022-01-10 21:47:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9ff4c6e6310602f5f6203e44e5fec3b3ac7fb667","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-01-10 21:35:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5b4434280a066f0548f248dc1f3eaf8a320d989f","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-01-10 21:35:29.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"7787493e979dd4b9902c8bef9951ff69b98a409a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 21:47:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- nodejs-run-lint https://zuul.opendev.org/t/openstack/build/2c5c72d8b0fd451e89dff66de1f8472e : SUCCESS in 4m 36s\n- nodejs-run-test-browser https://zuul.opendev.org/t/openstack/build/50bd276bbd164320b6f1cab3ce6bdecb : SUCCESS in 5m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6047efd1fb904413ab3959923486d373 : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/063b6c1314984a97aad813e9ea1b40a1 : FAILURE in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5c862cdc8d144f39aab655ad9c439553 : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0e334afafe2340ae8d70e134b695af15 : SUCCESS in 4m 50s\n- refstack-tox-functional https://zuul.opendev.org/t/openstack/build/3b22c059e1bc4c4cbf9b745287ab60d0 : SUCCESS in 5m 34s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/fdb671253be742b0adb8578d58a24c80 : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":1},{"id":"9cb629163c53215658eee3ddddf793f713be89cb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-01-11 12:34:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9b307691a4555903dbc6c2df709b50dee40dec80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 12:52:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs-run-lint https://zuul.opendev.org/t/openstack/build/cbb93a7ede344bda8f7d5f91fa2e3b68 : SUCCESS in 5m 03s\n- nodejs-run-test-browser https://zuul.opendev.org/t/openstack/build/ca18bf97ee6849bb8e20aeff7a9207a4 : SUCCESS in 6m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bf737ab50ac741d5b06014473b29698c : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fbbcb3e1a8d43e8b828832a27e02181 : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa0f8d93fb834018a425a4541e151538 : SUCCESS in 5m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/30fc736906e44dc4b1118b094350aa40 : SUCCESS in 5m 59s\n- refstack-tox-functional https://zuul.opendev.org/t/openstack/build/1a98c7558d0647849cfeaef16c2750ff : SUCCESS in 6m 08s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/a614ad112cea4841a77722bccae6e2a4 : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":2},{"id":"2021138933f0662b9712ce7ed2e56871cad138ef","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-01-14 00:21:37.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ac3206b164d54dafc77e999072945c35d0791e8c","revisions":{"b9fc56e6caba746c357fdde37873c58919a0055b":{"kind":"REWORK","_number":1,"created":"2022-01-10 21:35:06.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/66/824066/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack","ref":"refs/changes/66/824066/1","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/66/824066/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/66/824066/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/66/824066/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack refs/changes/66/824066/1"}}},"commit":{"parents":[{"commit":"8daea3e9086ecd4bbb394cb1878587b3760a5d0a","subject":"Add contributing page and some updates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/8daea3e9086ecd4bbb394cb1878587b3760a5d0a"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2022-01-10 21:30:15.000000000","tz":0},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2022-01-10 21:30:15.000000000","tz":0},"subject":"DNM - testing","message":"DNM - testing\n\nChange-Id: I1d12ffe4cc9f6a9a489dd664481274d25dc40fac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/b9fc56e6caba746c357fdde37873c58919a0055b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/b9fc56e6caba746c357fdde37873c58919a0055b"}]},"branch":"refs/heads/master"},"ac3206b164d54dafc77e999072945c35d0791e8c":{"kind":"REWORK","_number":2,"created":"2022-01-11 12:34:06.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/66/824066/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack","ref":"refs/changes/66/824066/2","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/66/824066/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/66/824066/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/66/824066/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack refs/changes/66/824066/2"}}},"commit":{"parents":[{"commit":"8daea3e9086ecd4bbb394cb1878587b3760a5d0a","subject":"Add contributing page and some updates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/8daea3e9086ecd4bbb394cb1878587b3760a5d0a"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2022-01-10 21:30:15.000000000","tz":0},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2022-01-11 12:29:29.000000000","tz":0},"subject":"DNM - testing","message":"DNM - testing\n\nChange-Id: I1d12ffe4cc9f6a9a489dd664481274d25dc40fac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/ac3206b164d54dafc77e999072945c35d0791e8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/ac3206b164d54dafc77e999072945c35d0791e8c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
