)]}'
{"id":"openinfra%2Frefstack~852861","triplet_id":"openinfra%2Frefstack~master~I33fcd3a39b58b07585f21950f58ea35cd8419282","project":"openinfra/refstack","branch":"master","hashtags":[],"change_id":"I33fcd3a39b58b07585f21950f58ea35cd8419282","subject":"Fix dash-separated options in setup.cfg warning","status":"MERGED","created":"2022-08-11 14:13:07.000000000","updated":"2022-08-15 07:36:21.000000000","submitted":"2022-08-15 07:34:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"852861","meta_rev_id":"a4feea0550e9ade7db99817c7044297ce4cce64f","_number":852861,"virtual_id_number":852861,"owner":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-15 07:34:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":2,"date":"2022-08-15 07:21:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":1,"date":"2022-08-15 07:21:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-11 14:25:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"07b2493f8eb09a808a037c6f5cbc1d500cae746d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-08-11 14:13:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e99dd23526eafbf9d07ef3102c4edc5325964054","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 14:25:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/679146f751964ca8bde393532b7c2d05\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4b0b71c3b00d4a408072516ca70c149f : SUCCESS in 2m 31s\n- nodejs-run-lint https://zuul.opendev.org/t/openstack/build/7e5328e098ab4887b008391606e15ed7 : SUCCESS in 4m 31s\n- nodejs-run-test-browser https://zuul.opendev.org/t/openstack/build/91518a52543148ffbafe2ee75bf77985 : SUCCESS in 7m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2f7adae90c4f492b9dcf6edc34140f5f : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c15a8b57c2b4cd6b9efd78c599ebac9 : SUCCESS in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e6e74345131c4dbda3cbee537dc61271 : SUCCESS in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/702daeb15fa04e99b50bdb78a7dbe95b : SUCCESS in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/60ee68f20ef94581a95fa9b2640914f4 : SUCCESS in 7m 10s (non-voting)\n- refstack-tox-functional https://zuul.opendev.org/t/openstack/build/030dc9bbf07b40b7a01634ba9d7170aa : SUCCESS in 7m 39s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/4c77032b0615474388a699bb5bc739ef : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":1},{"id":"0b39bb009d4d10b3aefd24dfae3686442c46a98d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-08-11 14:38:41.000000000","message":"Topic interop-jobs removed","accounts_in_message":[],"_revision_number":1},{"id":"4dbea144c079fc94d6e53659942544a00d9ee40d","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-08-15 07:21:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"08b5cb276672b7f298bec1be8c1073b61ccad7e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 07:21:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"18ce7c90ddb89ac4dfbfbae3135269f1d96e7ddd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 07:34:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d508d1eb7845431dbcdb0ba70eb2b8c5\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0964576c028a47ed80ba4845da32a4d8 : SUCCESS in 2m 42s\n- nodejs-run-lint https://zuul.opendev.org/t/openstack/build/e960ab1afd79434dbe2f3623b1d43437 : SUCCESS in 4m 53s\n- nodejs-run-test-browser https://zuul.opendev.org/t/openstack/build/3b25e4b5ae0844b993f05dd71f3bd48f : SUCCESS in 7m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/113cfddf93664ba6b0bd68d35b4e2e4c : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e873ae45a02e43af8e57168c4775b02e : SUCCESS in 5m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9b388594b3754e42841f58df0b97bd1a : SUCCESS in 5m 24s\n- refstack-tox-functional https://zuul.opendev.org/t/openstack/build/322e1d90dc8d42188542002adb8c21d5 : SUCCESS in 7m 22s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/e1194b2d96bd4fcbb8332eb47830dd3f : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":1},{"id":"89892b86c91dd0b765cca6305ad4feb21cfff30c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 07:34:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a4feea0550e9ade7db99817c7044297ce4cce64f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 07:36:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97762670a2454e328cdeeed896029aa6\n\n- opendev-promote-docs https://zuul.opendev.org/t/openstack/build/5a36a7d75f6943d3b80514b357bbd42c : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f222f3fb9540f0df4bf0cb8736c2470740659b18","revisions":{"f222f3fb9540f0df4bf0cb8736c2470740659b18":{"kind":"REWORK","_number":1,"created":"2022-08-11 14:13:07.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/61/852861/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack","ref":"refs/changes/61/852861/1","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/61/852861/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/61/852861/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/61/852861/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack refs/changes/61/852861/1"}}},"commit":{"parents":[{"commit":"fbe48dbad7be549ebbc4ae859823e5405f1e88d5","subject":"Add support for barbican","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/fbe48dbad7be549ebbc4ae859823e5405f1e88d5"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2022-08-11 14:11:08.000000000","tz":120},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2022-08-11 14:11:08.000000000","tz":120},"subject":"Fix dash-separated options in setup.cfg warning","message":"Fix dash-separated options in setup.cfg warning\n\nSetuptools v54.1.0 introduces a warning that the use of dash-separated\noptions in \u0027setup.cfg\u0027 will not be supported in a future version [1].\nLet\u0027s get ahead of the issue by replacing the dashes with underscores.\nWithout this, we see \u0027UserWarning\u0027 messages like the following on new\nenough versions of setuptools:\n\n  UserWarning: Usage of dash-separated \u0027description-file\u0027 will not be\n  supported in future versions. Please use the underscore name\n  \u0027description_file\u0027 instead\n\n[1] https://github.com/pypa/setuptools/commit/a2e9ae4cba\n\nChange-Id: I33fcd3a39b58b07585f21950f58ea35cd8419282\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/f222f3fb9540f0df4bf0cb8736c2470740659b18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/f222f3fb9540f0df4bf0cb8736c2470740659b18"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}}]}],"submit_requirements":[]}
