)]}'
{"id":"openinfra%2Frefstack~863641","triplet_id":"openinfra%2Frefstack~master~If98670475b371d1ece7c877a0eea3158f6c1b3f5","project":"openinfra/refstack","branch":"master","topic":"fix/verifier","attention_set":{},"removed_from_attention_set":{"30674":{"account":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"last_update":"2023-02-01 19:43:30.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2023-02-01 19:21:23.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"7160":{"account":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"last_update":"2023-02-01 19:43:30.000000000","reason":"Change was submitted"},"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2023-02-01 19:43:30.000000000","reason":"Change was submitted"},"22873":{"account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"last_update":"2023-02-01 14:52:53.000000000","reason":"\u003cGERRIT_ACCOUNT_22873\u003e replied on the change","reason_account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}}},"hashtags":[],"change_id":"If98670475b371d1ece7c877a0eea3158f6c1b3f5","subject":"Make refstack jobs work again","status":"MERGED","created":"2022-11-04 12:43:06.000000000","updated":"2023-02-01 19:44:29.000000000","submitted":"2023-02-01 19:43:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"863641-fix/verifier","meta_rev_id":"285c8528df17a2db8426c75c4888e8c2e7b24744","_number":863641,"virtual_id_number":863641,"owner":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-01 19:43:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2023-02-01 19:21:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2023-02-01 14:52:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2023-02-01 19:21:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}],"reviewers":{"REVIEWER":[{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-04 14:06:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-04 16:37:34.000000000","updated_by":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"CC"},{"updated":"2023-02-01 08:50:05.000000000","updated_by":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2023-02-01 14:52:53.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2023-02-01 14:52:53.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"state":"REVIEWER"},{"updated":"2023-02-01 14:52:53.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"}],"messages":[{"id":"b9bd44048abda1b614c7b34ffa038366262df345","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2022-11-04 12:43:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ac4d39a30ff05d1114b9f64c916a16e9537226b6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2022-11-04 13:03:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"31b00f3fecdc696924178a0d96ec842b83c268d5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2022-11-04 13:21:50.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"51ce7a20428a508bf156a5f173aa6492fcc6e63e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2022-11-04 13:34:44.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ebe7de92a50b45a1ca96d8ef8e223d27b0f1076e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 14:06:52.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8bf0bbb30f594aaca74e82839f648fd7\n\n- nodejs-run-lint https://zuul.opendev.org/t/openstack/build/25e7d3bb73da42d2a4ba346871e0135c : SUCCESS in 3m 45s\n- nodejs-run-test-browser https://zuul.opendev.org/t/openstack/build/0287739c13f344ff917530a700f20bb2 : SUCCESS in 5m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bffb7a7e585b49c3be06f5ed64c91594 : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bf0f6fde24443e4b43ab1ab59f4d0ed : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dc0cdbee18cb4fee94484e8539246511 : FAILURE in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ce57e204daa64260b1421b652c5b1a99 : FAILURE in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dde61aa0588048099d4423feb7d3047f : FAILURE in 3m 29s (non-voting)\n- refstack-tox-functional https://zuul.opendev.org/t/openstack/build/4f68b2422fc44ad7806950478734ca19 : TIMED_OUT in 30m 47s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/c5dc9f886141405da6ed32cfe6822da3 : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":4},{"id":"1f6215fa0378bb7bb21376456f296558ba311db6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2022-11-04 16:12:10.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"1badff959c4db46fcf65c8b3d4b42307e90b7b57","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2022-11-04 16:37:34.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b59375d1c77da4c663f8f7d5c8f7fbad0d0b0c98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 16:44:09.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/203d1e46169d47978ca0c08fcb665d84\n\n- nodejs-run-lint https://zuul.opendev.org/t/openstack/build/603620f852cf49f393529727e37148ac : SUCCESS in 4m 07s\n- nodejs-run-test-browser https://zuul.opendev.org/t/openstack/build/ae5fc0ad702a4e7da956d1a787489b72 : SUCCESS in 22m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1867d0aab55447e396e05b2078202687 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cd44d42f6304e0fbf316592ae4fbe28 : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cfc33c09b02541ce8b0132dafa048890 : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0249d686c5ad479794faaef1210f1f07 : SUCCESS in 4m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a472f9878e504a4eacf2d25f44fa71e0 : SUCCESS in 4m 32s (non-voting)\n- refstack-tox-functional https://zuul.opendev.org/t/openstack/build/52a617e6a6914edeace895d4e150312d : TIMED_OUT in 30m 48s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/6e83f13f9b824df7b7f503ee73991f0c : SUCCESS in 2m 59s","accounts_in_message":[],"_revision_number":5},{"id":"2434626758cd30853b9c4a140d01d317d887a851","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2022-11-04 16:51:46.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6c8eb1fb2f247b0d488aec51734e6718cf8104ab","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-01-30 10:28:05.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"0712dcab02ae7cef8fcd0ed33c59ff1146d0b092","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-30 10:46:03.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/40fb30968bc14a3895cff2a271f5702a\n\n- nodejs-run-lint https://zuul.opendev.org/t/openstack/build/88834233db924fc29cc79819ea140a85 : SUCCESS in 3m 28s\n- nodejs-run-test-browser https://zuul.opendev.org/t/openstack/build/b2fec3301d384b0988da5cc359f66464 : SUCCESS in 5m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f9062496b96041d4b0a8489279baad09 : FAILURE in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b60f611b37944449ed2ec5522313aed : FAILURE in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b7d997f457e1416e92c6516309a05e77 : FAILURE in 3m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/72b00327dad24ec8941b70fa5d6a30e9 : FAILURE in 3m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/69299194eba64660930c86f9c860266a : FAILURE in 2m 56s (non-voting)\n- refstack-tox-functional https://zuul.opendev.org/t/openstack/build/717b5d63fd504e64ada455a6c895b797 : FAILURE in 4m 01s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/881c3b6a715c439dbe1804b714a578c8 : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":6},{"id":"65d915d51ef5e167887320d6443274a3af2f88ef","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-01-30 10:52:37.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"0243070e9c44ecb2eea6178370b39e297a8092c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-30 11:03:59.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d5f0399843a4493baa975549eb002596\n\n- nodejs-run-lint https://zuul.opendev.org/t/openstack/build/e95270e4c8c44ae095b0d22a102ac5c1 : SUCCESS in 3m 28s\n- nodejs-run-test-browser https://zuul.opendev.org/t/openstack/build/5c8746a9749c4bddb6717205fc51df96 : SUCCESS in 5m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1816a52e8ee14b9b94f163dd1301698f : FAILURE in 3m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/687ae44a1d454012a21050838f4688c3 : FAILURE in 3m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/512a3a6226a64c68a99610517db6ced6 : FAILURE in 2m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a1940a2092fe46b28560bffd367087c0 : FAILURE in 2m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/74e3a427a0894dbfbbaa59d1fab003c0 : FAILURE in 2m 42s (non-voting)\n- refstack-tox-functional https://zuul.opendev.org/t/openstack/build/9d15437a71d0425eadaa27c0c802114e : SUCCESS in 4m 30s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/6c483dadfc11493fb63b22495bb3fad4 : SUCCESS in 2m 37s","accounts_in_message":[],"_revision_number":7},{"id":"e4aac586c5d245e494e7be397b6635ed4bd90c23","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-01-30 16:30:38.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"2a58bf309a4bec0cb3ccc75f1750dd835b22b0f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-30 16:51:00.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/72cf017859554c29ad3f25f261c83bd9\n\n- nodejs-run-lint https://zuul.opendev.org/t/openstack/build/78573867baf24606b301ad6a4b956662 : SUCCESS in 3m 36s\n- nodejs-run-test-browser https://zuul.opendev.org/t/openstack/build/345a609e056a4faaab75fa66b1a3d404 : SUCCESS in 6m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4ea25717cd9e47d1b6a026d492fe9562 : FAILURE in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f595f4f5d83c4c63a3471f5c95ccbd34 : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/93623ea2afb044d5a53902feb0e52005 : FAILURE in 2m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d36bc7633f8e4feaa01b732cc6325390 : FAILURE in 2m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/58f4468b21ee45e4a116e14e95c7c63f : FAILURE in 2m 47s (non-voting)\n- refstack-tox-functional https://zuul.opendev.org/t/openstack/build/f68272af2bbc47cab7b06e77ec00422a : SUCCESS in 4m 08s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/1f9a50d32d19416ea3c483e2d49b5347 : SUCCESS in 2m 57s","accounts_in_message":[],"_revision_number":8},{"id":"7c899c825d713807a841418d0f229d576b6107f3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-01-30 17:10:14.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"6a72d001d1fd632385ae546cabd98608a5cd8e28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-30 17:30:54.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/159599e7f8644f73b8dbb89051a36e71\n\n- nodejs-run-lint https://zuul.opendev.org/t/openstack/build/b7f6c3f37f77471db789cdf454081900 : SUCCESS in 3m 33s\n- nodejs-run-test-browser https://zuul.opendev.org/t/openstack/build/60f44352d30d4e9d9fccabef2747744c : SUCCESS in 5m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d6f724dab8ed48dba19c98d3a6b5400a : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83743191296e49b5aa705957e83238ed : SUCCESS in 3m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/79d46193e8424efdb4ea671963d50f93 : SUCCESS in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6a206febccc64c7fb0f01c746e7dcd1c : SUCCESS in 4m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/024e94a8b745451fb602291d0c0e01de : SUCCESS in 4m 23s (non-voting)\n- refstack-tox-functional https://zuul.opendev.org/t/openstack/build/381bbdacffed412ab8401be23f2759d3 : SUCCESS in 4m 54s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/6b21cccc026e4621810e10471257aa2f : SUCCESS in 2m 54s","accounts_in_message":[],"_revision_number":9},{"id":"999c53ec70c24ce47f0e26b6cfaab5dfc69a070a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-01-31 15:54:22.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"f5cc19f1a2d5a30c29acbcbb3d7a3f487d4924d5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-01-31 15:55:20.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"d868622a4e05d86d5064237bab9b3605638f93e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 16:17:06.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be9e874563c84516b95996c9a39d5f09\n\n- nodejs-run-lint https://zuul.opendev.org/t/openstack/build/d7b048a644ca4df782a486588c02edea : SUCCESS in 3m 51s\n- nodejs-run-test-browser https://zuul.opendev.org/t/openstack/build/f4ff3c08fa3b436688ef339e79145068 : SUCCESS in 5m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e3e765cfbf4a441ea01f171640b11944 : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/563ae112b554425ba984514e34a755d4 : SUCCESS in 3m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e5cff37f5c74446a9a42bf3c88ef8fae : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/21445c67cf4149fa92f7bd5862621757 : SUCCESS in 4m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cd50c758230e4437900ff826785a636a : SUCCESS in 3m 51s (non-voting)\n- refstack-tox-functional https://zuul.opendev.org/t/openstack/build/75e421cc90d94528b1b259bae8b70632 : SUCCESS in 9m 46s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/4001a6622bbe45e09c6880d5c134e7ff : SUCCESS in 2m 41s","accounts_in_message":[],"_revision_number":11},{"id":"0deae9389522cf181d156c2daf5c7661ea842df4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-01-31 16:49:17.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"24267406012f429c0571fa20c2688ed7a61b922e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-01-31 17:01:41.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"6fbf00c7152ea2d2eef0b4399a381006a99a0801","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 17:38:06.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d1464fdf94d74e56b2a187070cf71108\n\n- nodejs-run-lint https://zuul.opendev.org/t/openstack/build/897be48ff38a4cf581e4fe726cf9d25f : SUCCESS in 3m 44s\n- nodejs-run-test-browser https://zuul.opendev.org/t/openstack/build/9c394933f91d4a1088d814f6ac3582ce : RETRY_LIMIT in 3m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/98ec6b12b6eb4eb3b86c34596b5b0a51 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/256d338805264e2dab4c439b20c19d07 : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4ca7900b60d74fb582a9b38e507fdf00 : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/68d591d81f674084b762232ebc53a007 : SUCCESS in 3m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dbc10139ba04467a840340890eb7e563 : SUCCESS in 3m 31s (non-voting)\n- refstack-tox-functional https://zuul.opendev.org/t/openstack/build/59fdde33379f4e84973a34d6d2d4fe01 : SUCCESS in 4m 47s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/df90032f165945bdad373c576a00224a : SUCCESS in 2m 50s","accounts_in_message":[],"_revision_number":13},{"id":"b4cde7620988f9da746b84bc8a8787a3dfc11ca3","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-02-01 08:29:09.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"94008c07e2fa81d010151a40770abaab59d0509e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 08:44:51.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01fd194023fb40db9cfb809386bde8ab\n\n- nodejs-run-lint https://zuul.opendev.org/t/openstack/build/03c42c0c03c14976aef4fc0c2ab9cb0b : SUCCESS in 3m 58s\n- nodejs-run-test-browser https://zuul.opendev.org/t/openstack/build/9ac8301e37a1403eacb7d183aa1f83d3 : SUCCESS in 6m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b1d241c2aefc42a4817702d9c47d9a14 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cba60e043ed44bdab8b2c99c045d97e : SUCCESS in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/638f2bb7bcd1496785b35900dea397f3 : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab15a98490fb4243b622e3f6f0e373b7 : SUCCESS in 3m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9333eeeb7e534ca4aaf540f28ed94e6b : SUCCESS in 4m 17s (non-voting)\n- refstack-tox-functional https://zuul.opendev.org/t/openstack/build/3a82e8958ce64a8eaf2899d5bcf1ed32 : SUCCESS in 5m 09s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/190df2556f394c00a6b28e2914a59f3f : SUCCESS in 2m 46s","accounts_in_message":[],"_revision_number":13},{"id":"6baf6db17b9885963e4d1902d92cacce210fe4e4","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-02-01 08:50:05.000000000","message":"Patch Set 13:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":13},{"id":"fd291ab664707596514df83e3ae726591fca5462","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2023-02-01 10:53:43.000000000","message":"Patch Set 13: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":13},{"id":"c9b2e3346cc0175418cdc3dadfcea577142e4ba3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-02-01 12:05:55.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"f99bd0b8b62876c42d27b9575b762463d85a51f1","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-02-01 12:17:54.000000000","message":"Patch Set 13:\n\n(3 comments)","accounts_in_message":[],"_revision_number":13},{"id":"bd68ff445f27ab6668080170621e14e60de0a20d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 12:21:11.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba64714ee4a54c029a078c5da24fb9ad\n\n- nodejs-run-lint https://zuul.opendev.org/t/openstack/build/ce64dd51819d442d8571f4ceae7f0d34 : SUCCESS in 3m 42s\n- nodejs-run-test-browser https://zuul.opendev.org/t/openstack/build/07d5507ccbb840998afceca34d60db77 : SUCCESS in 6m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cbd7ada31dd84ac98b8e370b6f7676b9 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56176b1b84c64349a0e3f39b4175ac26 : SUCCESS in 3m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f1a01ab1a3604c7596e267ecfdbf2584 : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d5a8cd86a39b4970a1009c09eccb3c79 : SUCCESS in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/94b8751c3ff54772a7a373fa2f1e6f7a : SUCCESS in 3m 55s (non-voting)\n- refstack-tox-functional https://zuul.opendev.org/t/openstack/build/4206e847817b4436be41bdb28eb4e6a4 : SUCCESS in 4m 20s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/7ef0fc9535cc4d91b35d5973b1b4e3f4 : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":14},{"id":"25a66b4efcb88c59614a7834534f7cfcac0190c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-02-01 14:31:50.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"2391221bbaf169996e90e74217190fad3bb9e3fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 14:47:03.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d33890deac804dde833355621903674e\n\n- nodejs-run-lint https://zuul.opendev.org/t/openstack/build/d65fd2cb54bc4010a717113370b93da8 : SUCCESS in 4m 17s\n- nodejs-run-test-browser https://zuul.opendev.org/t/openstack/build/0b28c84e20d24b13bfd23ba28900cc00 : SUCCESS in 6m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9098b301542d41baaee9c296f80103d2 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/067ffd59575845f99a8526f23c79cdf6 : SUCCESS in 3m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ef5a920d15e4f7e9fb8b0a6aae4986c : SUCCESS in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cf6cfd4a7b604ff1ba3ccb43dd9a3139 : SUCCESS in 4m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e052819c52f84a029cd0e5df9d300e5f : SUCCESS in 3m 38s (non-voting)\n- refstack-tox-functional https://zuul.opendev.org/t/openstack/build/a3bc53f6dfcd4500b19b0bd9c98bb57e : SUCCESS in 4m 15s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/4524270e3889494fba812f87d364eedc : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":15},{"id":"53ca8d6a45e594a90acecdc4f78ec44bededd5ff","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2023-02-01 14:52:53.000000000","message":"Patch Set 15: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"3be4736945f73d85f5cb185caef0db629a13cbaf","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-02-01 19:21:23.000000000","message":"Patch Set 15: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"45e491161c12a0302974054286956ff994f74a5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 19:21:47.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"6ba829563b257c6e736579c9e2e83c01a3cc3ce1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 19:43:29.000000000","message":"Patch Set 15: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5919a69ead594eb0b83cea4cafdd7058\n\n- nodejs-run-lint https://zuul.opendev.org/t/openstack/build/ba1c0c00030843f7975a048d3b508706 : SUCCESS in 4m 28s\n- nodejs-run-test-browser https://zuul.opendev.org/t/openstack/build/d7c84e0758fd47e887dbd926181a9da2 : SUCCESS in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2445c5df244477597d7034ffffe79b5 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7513ec31a36a4de486a2517d906cd718 : SUCCESS in 3m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e9c1daa6c976406c9537554d8e4d491b : SUCCESS in 4m 25s\n- refstack-tox-functional https://zuul.opendev.org/t/openstack/build/4df9077af2514ec7b470eb73e3813257 : SUCCESS in 4m 03s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/3f800194450b4cb98f9740a54ab7e0c9 : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":15},{"id":"d9833710bae43061cc9e0900d17cbe9da6c4223b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 19:43:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":15},{"id":"285c8528df17a2db8426c75c4888e8c2e7b24744","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 19:44:29.000000000","message":"Patch Set 15:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a64ab8ee8fae4ea0953b25110e278697\n\n- opendev-promote-docs https://zuul.opendev.org/t/openstack/build/6aba8beeb23b427399148a9222544517 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"d18f8ad221c41e287c247d888122cb7f93dd8f46","revisions":{"336110b039781232e9ddec9692ec4f050226ee89":{"kind":"REWORK","_number":1,"created":"2022-11-04 12:43:06.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/41/863641/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack","ref":"refs/changes/41/863641/1","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/41/863641/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/41/863641/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/41/863641/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack refs/changes/41/863641/1"}}},"commit":{"parents":[{"commit":"f222f3fb9540f0df4bf0cb8736c2470740659b18","subject":"Fix dash-separated options in setup.cfg warning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/f222f3fb9540f0df4bf0cb8736c2470740659b18"}]}],"author":{"name":"Lukáš Piwowarski","email":"lpiwowar@redhat.com","date":"2022-11-04 12:42:44.000000000","tz":60},"committer":{"name":"Lukáš Piwowarski","email":"lpiwowar@redhat.com","date":"2022-11-04 12:43:04.000000000","tz":60},"subject":"DNM: verifier -\u003e verify","message":"DNM: verifier -\u003e verify\n\nChange-Id: If98670475b371d1ece7c877a0eea3158f6c1b3f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/336110b039781232e9ddec9692ec4f050226ee89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/336110b039781232e9ddec9692ec4f050226ee89"}]},"branch":"refs/heads/master"},"3b59a94399b3d18f145e5242a5879467754a31c3":{"kind":"REWORK","_number":2,"created":"2022-11-04 13:03:08.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/41/863641/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack","ref":"refs/changes/41/863641/2","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/41/863641/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/41/863641/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/41/863641/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack refs/changes/41/863641/2"}}},"commit":{"parents":[{"commit":"f222f3fb9540f0df4bf0cb8736c2470740659b18","subject":"Fix dash-separated options in setup.cfg warning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/f222f3fb9540f0df4bf0cb8736c2470740659b18"}]}],"author":{"name":"Lukáš Piwowarski","email":"lpiwowar@redhat.com","date":"2022-11-04 12:42:44.000000000","tz":60},"committer":{"name":"Lukáš Piwowarski","email":"lpiwowar@redhat.com","date":"2022-11-04 13:03:01.000000000","tz":60},"subject":"DNM: verifier -\u003e verify","message":"DNM: verifier -\u003e verify\n\nChange-Id: If98670475b371d1ece7c877a0eea3158f6c1b3f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/3b59a94399b3d18f145e5242a5879467754a31c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/3b59a94399b3d18f145e5242a5879467754a31c3"}]},"branch":"refs/heads/master"},"1a78cd3048958bb95b33e11e696ccf0228e19d5b":{"kind":"REWORK","_number":3,"created":"2022-11-04 13:21:50.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/41/863641/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack","ref":"refs/changes/41/863641/3","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/41/863641/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/41/863641/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/41/863641/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack refs/changes/41/863641/3"}}},"commit":{"parents":[{"commit":"f222f3fb9540f0df4bf0cb8736c2470740659b18","subject":"Fix dash-separated options in setup.cfg warning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/f222f3fb9540f0df4bf0cb8736c2470740659b18"}]}],"author":{"name":"Lukáš Piwowarski","email":"lpiwowar@redhat.com","date":"2022-11-04 12:42:44.000000000","tz":60},"committer":{"name":"Lukáš Piwowarski","email":"lpiwowar@redhat.com","date":"2022-11-04 13:21:43.000000000","tz":60},"subject":"DNM: verifier -\u003e verify","message":"DNM: verifier -\u003e verify\n\nChange-Id: If98670475b371d1ece7c877a0eea3158f6c1b3f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/1a78cd3048958bb95b33e11e696ccf0228e19d5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/1a78cd3048958bb95b33e11e696ccf0228e19d5b"}]},"branch":"refs/heads/master"},"9aed70e79905cf40007ba024cf3826c0fb370328":{"kind":"REWORK","_number":4,"created":"2022-11-04 13:34:44.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/41/863641/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack","ref":"refs/changes/41/863641/4","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/41/863641/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/41/863641/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/41/863641/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack refs/changes/41/863641/4"}}},"commit":{"parents":[{"commit":"f222f3fb9540f0df4bf0cb8736c2470740659b18","subject":"Fix dash-separated options in setup.cfg warning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/f222f3fb9540f0df4bf0cb8736c2470740659b18"}]}],"author":{"name":"Lukáš Piwowarski","email":"lpiwowar@redhat.com","date":"2022-11-04 12:42:44.000000000","tz":60},"committer":{"name":"Lukáš Piwowarski","email":"lpiwowar@redhat.com","date":"2022-11-04 13:34:36.000000000","tz":60},"subject":"DNM: verifier -\u003e verify","message":"DNM: verifier -\u003e verify\n\nChange-Id: If98670475b371d1ece7c877a0eea3158f6c1b3f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/9aed70e79905cf40007ba024cf3826c0fb370328"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/9aed70e79905cf40007ba024cf3826c0fb370328"}]},"branch":"refs/heads/master"},"08e9b0152f2e3f824067849bc3916e22ef14a73f":{"kind":"REWORK","_number":5,"created":"2022-11-04 16:12:10.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/41/863641/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack","ref":"refs/changes/41/863641/5","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/41/863641/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/41/863641/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/41/863641/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack refs/changes/41/863641/5"}}},"commit":{"parents":[{"commit":"f222f3fb9540f0df4bf0cb8736c2470740659b18","subject":"Fix dash-separated options in setup.cfg warning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/f222f3fb9540f0df4bf0cb8736c2470740659b18"}]}],"author":{"name":"Lukáš Piwowarski","email":"lpiwowar@redhat.com","date":"2022-11-04 12:42:44.000000000","tz":60},"committer":{"name":"Lukáš Piwowarski","email":"lpiwowar@redhat.com","date":"2022-11-04 16:10:19.000000000","tz":60},"subject":"Fix verification of signatures","message":"Fix verification of signatures\n\nThe code of refstack API used verifier() method for the verification of\nsignatures. This method was deprecated and now is entirely removed\nfrom the cryptography library [1].\n\nThis patch makes sure that we use the verify() method which is supposed\nto be replacement for the verifier() method.\n\n[1] https://github.com/pyca/cryptography/commit/e71c0df30102611d1d4ea5ebd5c8a0f4d42204db\nTaiga Issue: https://tree.taiga.io/project/openstack-interop-working-group/issue/77\n\nChange-Id: If98670475b371d1ece7c877a0eea3158f6c1b3f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/08e9b0152f2e3f824067849bc3916e22ef14a73f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/08e9b0152f2e3f824067849bc3916e22ef14a73f"}]},"branch":"refs/heads/master"},"75cff7ecdfc1dcad5dbd4748b89d631a17969adf":{"kind":"REWORK","_number":6,"created":"2023-01-30 10:28:05.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/41/863641/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack","ref":"refs/changes/41/863641/6","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/41/863641/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/41/863641/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/41/863641/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack refs/changes/41/863641/6"}}},"commit":{"parents":[{"commit":"f222f3fb9540f0df4bf0cb8736c2470740659b18","subject":"Fix dash-separated options in setup.cfg warning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/f222f3fb9540f0df4bf0cb8736c2470740659b18"}]}],"author":{"name":"Lukáš Piwowarski","email":"lpiwowar@redhat.com","date":"2022-11-04 12:42:44.000000000","tz":60},"committer":{"name":"Lukáš Piwowarski","email":"lpiwowar@redhat.com","date":"2023-01-30 10:24:52.000000000","tz":60},"subject":"[WIP] Fix TIMED OUT for refstack-tox-functional job","message":"[WIP] Fix TIMED OUT for refstack-tox-functional job\n\nThe code of refstack API used verifier() method for the verification of\nsignatures. This method was deprecated and now is entirely removed\nfrom the cryptography library [1].\n\nThis patch makes sure that we use the verify() method which is supposed\nto be replacement for the verifier() method.\n\n[1] https://github.com/pyca/cryptography/commit/e71c0df30102611d1d4ea5ebd5c8a0f4d42204db\nTaiga Issue: https://tree.taiga.io/project/openstack-interop-working-group/issue/77\n\nChange-Id: If98670475b371d1ece7c877a0eea3158f6c1b3f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/75cff7ecdfc1dcad5dbd4748b89d631a17969adf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/75cff7ecdfc1dcad5dbd4748b89d631a17969adf"}]},"branch":"refs/heads/master"},"843ef03f950453017f0263f41894983c01c719f8":{"kind":"REWORK","_number":7,"created":"2023-01-30 10:52:37.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/41/863641/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack","ref":"refs/changes/41/863641/7","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/41/863641/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/41/863641/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/41/863641/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack refs/changes/41/863641/7"}}},"commit":{"parents":[{"commit":"f222f3fb9540f0df4bf0cb8736c2470740659b18","subject":"Fix dash-separated options in setup.cfg warning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/f222f3fb9540f0df4bf0cb8736c2470740659b18"}]}],"author":{"name":"Lukáš Piwowarski","email":"lpiwowar@redhat.com","date":"2022-11-04 12:42:44.000000000","tz":60},"committer":{"name":"Lukáš Piwowarski","email":"lpiwowar@redhat.com","date":"2023-01-30 10:51:59.000000000","tz":60},"subject":"[WIP] Fix TIMED OUT for refstack-tox-functional job","message":"[WIP] Fix TIMED OUT for refstack-tox-functional job\n\nThe code of refstack API used verifier() method for the verification of\nsignatures. This method was deprecated and now is entirely removed\nfrom the cryptography library [1].\n\nThis patch makes sure that we use the verify() method which is supposed\nto be replacement for the verifier() method.\n\n[1] https://github.com/pyca/cryptography/commit/e71c0df30102611d1d4ea5ebd5c8a0f4d42204db\nTaiga Issue: https://tree.taiga.io/project/openstack-interop-working-group/issue/77\n\nChange-Id: If98670475b371d1ece7c877a0eea3158f6c1b3f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/843ef03f950453017f0263f41894983c01c719f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/843ef03f950453017f0263f41894983c01c719f8"}]},"branch":"refs/heads/master"},"893a1a5a6329bf90877b77aaff728e92df13848d":{"kind":"REWORK","_number":8,"created":"2023-01-30 16:30:38.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/41/863641/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack","ref":"refs/changes/41/863641/8","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/41/863641/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/41/863641/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/41/863641/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack refs/changes/41/863641/8"}}},"commit":{"parents":[{"commit":"f222f3fb9540f0df4bf0cb8736c2470740659b18","subject":"Fix dash-separated options in setup.cfg warning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/f222f3fb9540f0df4bf0cb8736c2470740659b18"}]}],"author":{"name":"Lukáš Piwowarski","email":"lpiwowar@redhat.com","date":"2022-11-04 12:42:44.000000000","tz":60},"committer":{"name":"Lukáš Piwowarski","email":"lpiwowar@redhat.com","date":"2023-01-30 16:30:00.000000000","tz":60},"subject":"[WIP] Fix TIMED OUT for refstack-tox-functional job","message":"[WIP] Fix TIMED OUT for refstack-tox-functional job\n\nThe code of refstack API used verifier() method for the verification of\nsignatures. This method was deprecated and now is entirely removed\nfrom the cryptography library [1].\n\nThis patch makes sure that we use the verify() method which is supposed\nto be replacement for the verifier() method.\n\n[1] https://github.com/pyca/cryptography/commit/e71c0df30102611d1d4ea5ebd5c8a0f4d42204db\nTaiga Issue: https://tree.taiga.io/project/openstack-interop-working-group/issue/77\n\nChange-Id: If98670475b371d1ece7c877a0eea3158f6c1b3f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/893a1a5a6329bf90877b77aaff728e92df13848d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/893a1a5a6329bf90877b77aaff728e92df13848d"}]},"branch":"refs/heads/master"},"49af00ce5ea836b587fd1758252b12393d5eb15e":{"kind":"REWORK","_number":9,"created":"2023-01-30 17:10:14.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/41/863641/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack","ref":"refs/changes/41/863641/9","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/41/863641/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/41/863641/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/41/863641/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack refs/changes/41/863641/9"}}},"commit":{"parents":[{"commit":"f222f3fb9540f0df4bf0cb8736c2470740659b18","subject":"Fix dash-separated options in setup.cfg warning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/f222f3fb9540f0df4bf0cb8736c2470740659b18"}]}],"author":{"name":"Lukáš Piwowarski","email":"lpiwowar@redhat.com","date":"2022-11-04 12:42:44.000000000","tz":60},"committer":{"name":"Lukáš Piwowarski","email":"lpiwowar@redhat.com","date":"2023-01-30 17:09:51.000000000","tz":60},"subject":"[WIP] Fix TIMED OUT for refstack-tox-functional job","message":"[WIP] Fix TIMED OUT for refstack-tox-functional job\n\nThe code of refstack API used verifier() method for the verification of\nsignatures. This method was deprecated and now is entirely removed\nfrom the cryptography library [1].\n\nThis patch makes sure that we use the verify() method which is supposed\nto be replacement for the verifier() method.\n\n[1] https://github.com/pyca/cryptography/commit/e71c0df30102611d1d4ea5ebd5c8a0f4d42204db\nTaiga Issue: https://tree.taiga.io/project/openstack-interop-working-group/issue/77\n\nChange-Id: If98670475b371d1ece7c877a0eea3158f6c1b3f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/49af00ce5ea836b587fd1758252b12393d5eb15e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/49af00ce5ea836b587fd1758252b12393d5eb15e"}]},"branch":"refs/heads/master"},"6cfedc68ee0d7ca1c420606d18cfe6233544a75e":{"kind":"REWORK","_number":10,"created":"2023-01-31 15:54:22.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/41/863641/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack","ref":"refs/changes/41/863641/10","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/41/863641/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/41/863641/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/41/863641/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack refs/changes/41/863641/10"}}},"commit":{"parents":[{"commit":"f222f3fb9540f0df4bf0cb8736c2470740659b18","subject":"Fix dash-separated options in setup.cfg warning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/f222f3fb9540f0df4bf0cb8736c2470740659b18"}]}],"author":{"name":"Lukáš Piwowarski","email":"lpiwowar@redhat.com","date":"2022-11-04 12:42:44.000000000","tz":60},"committer":{"name":"Lukáš Piwowarski","email":"lpiwowar@redhat.com","date":"2023-01-31 15:53:49.000000000","tz":60},"subject":"[WIP] Fix TIMED OUT for refstack-tox-functional job","message":"[WIP] Fix TIMED OUT for refstack-tox-functional job\n\nThe code of refstack API used verifier() method for the verification of\nsignatures. This method was deprecated and now is entirely removed\nfrom the cryptography library [1].\n\nThis patch makes sure that we use the verify() method which is supposed\nto be replacement for the verifier() method.\n\n[1] https://github.com/pyca/cryptography/commit/e71c0df30102611d1d4ea5ebd5c8a0f4d42204db\nTaiga Issue: https://tree.taiga.io/project/openstack-interop-working-group/issue/77\n\nChange-Id: If98670475b371d1ece7c877a0eea3158f6c1b3f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/6cfedc68ee0d7ca1c420606d18cfe6233544a75e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/6cfedc68ee0d7ca1c420606d18cfe6233544a75e"}]},"branch":"refs/heads/master"},"cc0010ec6fc8042d91fad897e3b2dd69cc996d26":{"kind":"REWORK","_number":11,"created":"2023-01-31 15:55:20.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/41/863641/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack","ref":"refs/changes/41/863641/11","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/41/863641/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/41/863641/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/41/863641/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack refs/changes/41/863641/11"}}},"commit":{"parents":[{"commit":"f222f3fb9540f0df4bf0cb8736c2470740659b18","subject":"Fix dash-separated options in setup.cfg warning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/f222f3fb9540f0df4bf0cb8736c2470740659b18"}]}],"author":{"name":"Lukáš Piwowarski","email":"lpiwowar@redhat.com","date":"2022-11-04 12:42:44.000000000","tz":60},"committer":{"name":"Lukáš Piwowarski","email":"lpiwowar@redhat.com","date":"2023-01-31 15:55:13.000000000","tz":60},"subject":"[WIP] Fix TIMED OUT for refstack-tox-functional job","message":"[WIP] Fix TIMED OUT for refstack-tox-functional job\n\nThe code of refstack API used verifier() method for the verification of\nsignatures. This method was deprecated and now is entirely removed\nfrom the cryptography library [1].\n\nThis patch makes sure that we use the verify() method which is supposed\nto be replacement for the verifier() method.\n\n[1] https://github.com/pyca/cryptography/commit/e71c0df30102611d1d4ea5ebd5c8a0f4d42204db\nTaiga Issue: https://tree.taiga.io/project/openstack-interop-working-group/issue/77\n\nChange-Id: If98670475b371d1ece7c877a0eea3158f6c1b3f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/cc0010ec6fc8042d91fad897e3b2dd69cc996d26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/cc0010ec6fc8042d91fad897e3b2dd69cc996d26"}]},"branch":"refs/heads/master"},"272d5e94d385abb2cdcbe399e97e86ee19b5b10f":{"kind":"REWORK","_number":12,"created":"2023-01-31 16:49:17.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/41/863641/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack","ref":"refs/changes/41/863641/12","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/41/863641/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/41/863641/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/41/863641/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack refs/changes/41/863641/12"}}},"commit":{"parents":[{"commit":"f222f3fb9540f0df4bf0cb8736c2470740659b18","subject":"Fix dash-separated options in setup.cfg warning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/f222f3fb9540f0df4bf0cb8736c2470740659b18"}]}],"author":{"name":"Lukáš Piwowarski","email":"lpiwowar@redhat.com","date":"2022-11-04 12:42:44.000000000","tz":60},"committer":{"name":"Lukáš Piwowarski","email":"lpiwowar@redhat.com","date":"2023-01-31 16:46:41.000000000","tz":60},"subject":"Make refstack jobs work again","message":"Make refstack jobs work again\n\nThere were three main issues in the code from refstack repository that\nwere preventing refstack jobs from working:\n\n1) Improper handling of sessions in the database API.\n2) PEP8 job was failing because of outdated version of flake8.\n3) The new version of cryptography library doesn\u0027t support signer() and\n   verifier() functions.\n\nIssue #1 was solved by using the get_session() function as a context\nmanager instead of using session.begin() as a context manager. Using\nsession.begin() as a context manager does not ensure that the session\nwill be closed at the end of the context (see \"Opening and Closing\na Session\" and \"Framing out a begin / commit / rollback block\"\nhere [1]).\n\nIssue #2 was solved by updating the libraries in\ntest-requirements.txt file. This change also forces flake8 to ignore\nsome pep8 errors (similar to the ones ignored in tempest project).\n\nIssue #3 was solved by using the sign() and verify() functions instead\nof verifier() and signer() functions [2].\n\nRelated Taiga issues:\n - https://tree.taiga.io/project/openstack-interop-working-group/issue/77\n - https://tree.taiga.io/project/openstack-interop-working-group/issue/79\n\n[1] https://docs.sqlalchemy.org/en/14/orm/session_basics.html\n[2] https://github.com/pyca/cryptography/commit/e71c0df30102611d1d4ea5ebd5c8a0f4d42204db\n\nChange-Id: If98670475b371d1ece7c877a0eea3158f6c1b3f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/272d5e94d385abb2cdcbe399e97e86ee19b5b10f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/272d5e94d385abb2cdcbe399e97e86ee19b5b10f"}]},"branch":"refs/heads/master"},"87a12dfc18b119f25dad063191d1dd01b13889ac":{"kind":"REWORK","_number":13,"created":"2023-01-31 17:01:41.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/41/863641/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack","ref":"refs/changes/41/863641/13","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/41/863641/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/41/863641/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/41/863641/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack refs/changes/41/863641/13"}}},"commit":{"parents":[{"commit":"f222f3fb9540f0df4bf0cb8736c2470740659b18","subject":"Fix dash-separated options in setup.cfg warning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/f222f3fb9540f0df4bf0cb8736c2470740659b18"}]}],"author":{"name":"Lukáš Piwowarski","email":"lpiwowar@redhat.com","date":"2022-11-04 12:42:44.000000000","tz":60},"committer":{"name":"Lukáš Piwowarski","email":"lpiwowar@redhat.com","date":"2023-01-31 17:01:29.000000000","tz":60},"subject":"Make refstack jobs work again","message":"Make refstack jobs work again\n\nThere were three main issues in the code from refstack repository that\nwere preventing refstack jobs from working:\n\n1) Improper handling of sessions in the database API.\n2) PEP8 job was failing because of outdated version of flake8.\n3) The new version of cryptography library doesn\u0027t support signer() and\n   verifier() functions.\n\nIssue #1 was solved by using the get_session() function as a context\nmanager instead of using session.begin() as a context manager. Using\nsession.begin() as a context manager does not ensure that the session\nwill be closed at the end of the context (see \"Opening and Closing\na Session\" and \"Framing out a begin / commit / rollback block\"\nhere [1]).\n\nIssue #2 was solved by updating the libraries in\ntest-requirements.txt file. This change also forces flake8 to ignore\nsome pep8 errors (similar to the ones ignored in tempest project).\n\nIssue #3 was solved by using the sign() and verify() functions instead\nof verifier() and signer() functions [2].\n\nRelated Taiga issues:\n - https://tree.taiga.io/project/openstack-interop-working-group/issue/77\n - https://tree.taiga.io/project/openstack-interop-working-group/issue/79\n\n[1] https://docs.sqlalchemy.org/en/14/orm/session_basics.html\n[2] https://github.com/pyca/cryptography/commit/e71c0df30102611d1d4ea5ebd5c8a0f4d42204db\n\nChange-Id: If98670475b371d1ece7c877a0eea3158f6c1b3f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/87a12dfc18b119f25dad063191d1dd01b13889ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/87a12dfc18b119f25dad063191d1dd01b13889ac"}]},"branch":"refs/heads/master"},"18e3003eda588d634777b5d86f04775b9b4f8497":{"kind":"REWORK","_number":14,"created":"2023-02-01 12:05:55.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/41/863641/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack","ref":"refs/changes/41/863641/14","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/41/863641/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/41/863641/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/41/863641/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack refs/changes/41/863641/14"}}},"commit":{"parents":[{"commit":"f222f3fb9540f0df4bf0cb8736c2470740659b18","subject":"Fix dash-separated options in setup.cfg warning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/f222f3fb9540f0df4bf0cb8736c2470740659b18"}]}],"author":{"name":"Lukáš Piwowarski","email":"lpiwowar@redhat.com","date":"2022-11-04 12:42:44.000000000","tz":60},"committer":{"name":"Lukáš Piwowarski","email":"lpiwowar@redhat.com","date":"2023-02-01 11:47:42.000000000","tz":60},"subject":"Make refstack jobs work again","message":"Make refstack jobs work again\n\nThere were three main issues in the code from refstack repository that\nwere preventing refstack jobs from working:\n\n1) Improper handling of sessions in the database API.\n2) PEP8 job was failing because of outdated version of flake8.\n3) The new version of cryptography library doesn\u0027t support signer() and\n   verifier() functions.\n\nIssue #1 was solved by using the get_session() function as a context\nmanager instead of using session.begin() as a context manager. Using\nsession.begin() as a context manager does not ensure that the session\nwill be closed at the end of the context (see \"Opening and Closing\na Session\" and \"Framing out a begin / commit / rollback block\"\nhere [1]).\n\nIssue #2 was solved by updating the libraries in\ntest-requirements.txt file. This change also forces flake8 to ignore\nsome pep8 errors (similar to the ones ignored in tempest project).\n\nIssue #3 was solved by using the sign() and verify() functions instead\nof verifier() and signer() functions [2].\n\nRelated Taiga issues:\n - https://tree.taiga.io/project/openstack-interop-working-group/issue/77\n - https://tree.taiga.io/project/openstack-interop-working-group/issue/79\n\n[1] https://docs.sqlalchemy.org/en/14/orm/session_basics.html\n[2] https://github.com/pyca/cryptography/commit/e71c0df30102611d1d4ea5ebd5c8a0f4d42204db\n\nChange-Id: If98670475b371d1ece7c877a0eea3158f6c1b3f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/18e3003eda588d634777b5d86f04775b9b4f8497"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/18e3003eda588d634777b5d86f04775b9b4f8497"}]},"branch":"refs/heads/master"},"d18f8ad221c41e287c247d888122cb7f93dd8f46":{"kind":"REWORK","_number":15,"created":"2023-02-01 14:31:50.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/41/863641/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack","ref":"refs/changes/41/863641/15","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/41/863641/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/41/863641/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/41/863641/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack refs/changes/41/863641/15"}}},"commit":{"parents":[{"commit":"f222f3fb9540f0df4bf0cb8736c2470740659b18","subject":"Fix dash-separated options in setup.cfg warning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/f222f3fb9540f0df4bf0cb8736c2470740659b18"}]}],"author":{"name":"Lukáš Piwowarski","email":"lpiwowar@redhat.com","date":"2022-11-04 12:42:44.000000000","tz":60},"committer":{"name":"Lukáš Piwowarski","email":"lpiwowar@redhat.com","date":"2023-02-01 14:29:36.000000000","tz":60},"subject":"Make refstack jobs work again","message":"Make refstack jobs work again\n\nThere were three main issues in the code from refstack repository that\nwere preventing refstack jobs from working:\n\n1) Improper handling of sessions in the database API.\n2) PEP8 job was failing because of outdated version of flake8.\n3) The new version of cryptography library doesn\u0027t support signer() and\n   verifier() functions.\n\nIssue #1 was solved by using the get_session() function as a context\nmanager instead of using session.begin() as a context manager. Using\nsession.begin() as a context manager does not ensure that the session\nwill be closed at the end of the context (see \"Opening and Closing\na Session\" and \"Framing out a begin / commit / rollback block\"\nhere [1]).\n\nIssue #2 was solved by updating the libraries in\ntest-requirements.txt file. This change also forces flake8 to ignore\nsome pep8 errors (similar to the ones ignored in tempest project).\n\nIssue #3 was solved by using the sign() and verify() functions instead\nof verifier() and signer() functions [2].\n\nRelated Taiga issues:\n - https://tree.taiga.io/project/openstack-interop-working-group/issue/77\n - https://tree.taiga.io/project/openstack-interop-working-group/issue/79\n\n[1] https://docs.sqlalchemy.org/en/14/orm/session_basics.html\n[2] https://github.com/pyca/cryptography/commit/e71c0df30102611d1d4ea5ebd5c8a0f4d42204db\n\nChange-Id: If98670475b371d1ece7c877a0eea3158f6c1b3f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/d18f8ad221c41e287c247d888122cb7f93dd8f46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/d18f8ad221c41e287c247d888122cb7f93dd8f46"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[]}
