)]}'
{"id":"openinfra%2Frefstack~880001","triplet_id":"openinfra%2Frefstack~master~I8d9a351697408d020700ce76986d4cabdc846334","project":"openinfra/refstack","branch":"master","topic":"fix/about_page","attention_set":{},"removed_from_attention_set":{"35229":{"account":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"last_update":"2023-04-11 10:27:02.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I8d9a351697408d020700ce76986d4cabdc846334","subject":"Fix bug in refstack-ui","status":"ABANDONED","created":"2023-04-10 22:17:56.000000000","updated":"2023-04-11 10:27:02.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c432699fbe26b6140a1a929c0db452d0a270265c","_number":880001,"virtual_id_number":880001,"owner":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-04-10 23:13:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-10 22:31:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3ba444b5f90c5a12fca0abd5b7a81b6490de88f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"date":"2023-04-10 22:17:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c30fe3e198cf8c042a52c02574b607ad2f7dd1b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-10 22:31:02.000000000","message":"Patch Set 1: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5ea0e0f5328049a4ba1aa961919d7da5\n\n- nodejs-run-lint https://zuul.opendev.org/t/openstack/build/d083dcdf5e1b4628b2a5d63e1517aed4 : FAILURE in 3m 16s\n- nodejs-run-test-browser https://zuul.opendev.org/t/openstack/build/a3b9d1e5ac6241ab82ccf78fa9a0af96 : FAILURE in 5m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0dff2eb0ce1d406f9aa9b30600b19ab0 : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71be49e69b63487ca5b50ef394806502 : FAILURE in 2m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/79a6046a3ad845d1b6d9b49b8cd75fbb : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3397cd6b5e8f4304a941c2a9999c6163 : SUCCESS in 3m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8822b91d17ab4f7881572f063deeca27 : SUCCESS in 3m 38s (non-voting)\n- refstack-tox-functional https://zuul.opendev.org/t/openstack/build/13ba1b3dbc4a49928f515b325831204a : SUCCESS in 4m 09s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/60382aa4690b45968473f5c5d74cfcf5 : SUCCESS in 2m 37s","accounts_in_message":[],"_revision_number":1},{"id":"6be008a593429d9fa4545c1509dd378a9cd909a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"date":"2023-04-10 22:50:27.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"7f0ca2125c6e6702811051d5bb9bd01a6e218a50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-10 23:13:20.000000000","message":"Patch Set 2: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/43c75667964b442da81b2b8ee91c7e66\n\n- nodejs-run-lint https://zuul.opendev.org/t/openstack/build/b16410d928fc4c9e948ee92ba6b7273a : SUCCESS in 3m 37s\n- nodejs-run-test-browser https://zuul.opendev.org/t/openstack/build/8e647d432ced4f01b475bebdb103c8ab : FAILURE in 5m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b1c2f4c302ed4d758d535b4440a996fb : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42d2b1bfc70740f2ac7d44e001ec24c7 : FAILURE in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0dc767e91fff46aea0a47a186d6e717f : SUCCESS in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/394e447a7f3443b898c91dad19435316 : SUCCESS in 4m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7820d4217aa241a69a5a69e3ffdb8d5f : SUCCESS in 3m 49s (non-voting)\n- refstack-tox-functional https://zuul.opendev.org/t/openstack/build/02c2d1a05d8c4ab99c00b80fdcbe4e54 : SUCCESS in 7m 59s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/b420cd82074242fb84aff352c0ed0ad1 : SUCCESS in 2m 23s","accounts_in_message":[],"_revision_number":2},{"id":"c432699fbe26b6140a1a929c0db452d0a270265c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"date":"2023-04-11 10:27:02.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ff6701632307ef1beef0ead989ede657f925f364","revisions":{"d0e5e1d1448b84c1341efab80f3dd7dc38aa4c31":{"kind":"REWORK","_number":1,"created":"2023-04-10 22:17:56.000000000","uploader":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"ref":"refs/changes/01/880001/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack","ref":"refs/changes/01/880001/1","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/01/880001/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/01/880001/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/01/880001/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack refs/changes/01/880001/1"}}},"commit":{"parents":[{"commit":"a255af64e7b7f5805ef3ecea28f0f2fc183c24b5","subject":"Fix for bug in refstack-ui","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/a255af64e7b7f5805ef3ecea28f0f2fc183c24b5"}]}],"author":{"name":"jskunda","email":"jskunda@redhat.com","date":"2023-04-10 21:47:21.000000000","tz":120},"committer":{"name":"jskunda","email":"jskunda@redhat.com","date":"2023-04-10 21:47:21.000000000","tz":120},"subject":"Fix bug in refstack-ui","message":"Fix bug in refstack-ui\n\nRefstack webpage is not rendered correctly. The bug is tracked in\nhere: https://issues.redhat.com/browse/OSP-22171\n\nThis patch will update documentation build process to generate html\nfiles.\n- Modify convert-docs.py to use Sphinx instead of Docutils\n- Change paths to html files in about.html\n\nChange-Id: I8d9a351697408d020700ce76986d4cabdc846334\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/d0e5e1d1448b84c1341efab80f3dd7dc38aa4c31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/d0e5e1d1448b84c1341efab80f3dd7dc38aa4c31"}]},"branch":"refs/heads/master"},"ff6701632307ef1beef0ead989ede657f925f364":{"kind":"REWORK","_number":2,"created":"2023-04-10 22:50:27.000000000","uploader":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"ref":"refs/changes/01/880001/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack","ref":"refs/changes/01/880001/2","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/01/880001/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/01/880001/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/01/880001/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack refs/changes/01/880001/2"}}},"commit":{"parents":[{"commit":"a255af64e7b7f5805ef3ecea28f0f2fc183c24b5","subject":"Fix for bug in refstack-ui","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/a255af64e7b7f5805ef3ecea28f0f2fc183c24b5"}]}],"author":{"name":"jskunda","email":"jskunda@redhat.com","date":"2023-04-10 21:47:21.000000000","tz":120},"committer":{"name":"jskunda","email":"jskunda@redhat.com","date":"2023-04-10 22:50:04.000000000","tz":120},"subject":"Fix bug in refstack-ui","message":"Fix bug in refstack-ui\n\nRefstack webpage is not rendered correctly. The bug is tracked in\nhere: https://issues.redhat.com/browse/OSP-22171\n\nThis patch will update documentation build process to generate html\nfiles.\n- Modify convert-docs.py to use Sphinx instead of Docutils\n- Change paths to html files in about.html\n\nChange-Id: I8d9a351697408d020700ce76986d4cabdc846334\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/ff6701632307ef1beef0ead989ede657f925f364"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/ff6701632307ef1beef0ead989ede657f925f364"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
