)]}'
{"id":"openinfra%2Frefstack~880034","triplet_id":"openinfra%2Frefstack~master~Id296b9cecf75bb289b0da41bec880a13694298e1","project":"openinfra/refstack","branch":"master","attention_set":{},"removed_from_attention_set":{"30674":{"account":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"last_update":"2023-04-20 08:26:30.000000000","reason":"\u003cGERRIT_ACCOUNT_30674\u003e replied on the change","reason_account":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"}},"22873":{"account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"last_update":"2023-05-09 14:42:43.000000000","reason":"\u003cGERRIT_ACCOUNT_22873\u003e replied on the change","reason_account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},"35229":{"account":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"last_update":"2023-05-09 14:56:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id296b9cecf75bb289b0da41bec880a13694298e1","subject":"Fix missing about content in refstack-ui","status":"MERGED","created":"2023-04-11 10:26:18.000000000","updated":"2023-05-09 14:57:40.000000000","submitted":"2023-05-09 14:56:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":21,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880034","meta_rev_id":"dfa3f4810076c61a70898881454d77633b195995","_number":880034,"virtual_id_number":880034,"owner":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-09 14:56:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"recommended":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"all":[{"value":1,"date":"2023-04-20 08:26:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},{"value":2,"date":"2023-05-09 14:42:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":0,"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},{"value":1,"date":"2023-05-09 14:42:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-11 10:40:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-18 13:43:27.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2023-04-19 09:23:07.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"state":"REVIEWER"}],"messages":[{"id":"37806d666d717b3ed7fc3f7d8b5083db49e0d143","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"date":"2023-04-11 10:26:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0fb87521bc834783cfbd5fc7695b4cefb06ba5fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-11 10:40:17.000000000","message":"Patch Set 1: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c94eb09fd4a44814afb6279c1b529dc0\n\n- nodejs-run-lint https://zuul.opendev.org/t/openstack/build/040c387dae2d4fc59dbdab5c5c52f19e : SUCCESS in 4m 20s\n- nodejs-run-test-browser https://zuul.opendev.org/t/openstack/build/f08dd42870a54abbbcb486ae8555c667 : SUCCESS in 5m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db65b7fb6ec644bbbc8be97530448989 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d69c043243ce4c22994d279b282e0ecd : FAILURE in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dbadef38602245119075e764cb397ebe : SUCCESS in 3m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a8bf9504fc404986bbf625aa03b0640e : SUCCESS in 4m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cbb84252e4c644e782181d3d46b831ba : SUCCESS in 3m 32s (non-voting)\n- refstack-tox-functional https://zuul.opendev.org/t/openstack/build/f25ca633bc744032b16a86db89621195 : SUCCESS in 4m 27s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/7499e9746d314106930c708668fe50bd : SUCCESS in 3m 28s","accounts_in_message":[],"_revision_number":1},{"id":"3d7ffec7bd6dc47af2fba9ac9f2670b89dab150b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"date":"2023-04-11 12:30:54.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"acee4589ee73aef9c14d3196ee0ca8786b99189c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-11 12:38:14.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a507e3ab35f44951a9ccd709a374c102\n\n- nodejs-run-lint https://zuul.opendev.org/t/openstack/build/b01ead30426844d0a5986a6c7035ef09 : SUCCESS in 3m 32s\n- nodejs-run-test-browser https://zuul.opendev.org/t/openstack/build/7eb1abdd6daf4c10be020ebb941b874c : SUCCESS in 5m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cfbc1a5509c14e67bac146fc36fcf9b8 : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c2c3b36e0a14d07a788dceae53300a8 : FAILURE in 3m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/49cefc1fc194478a95965944b060d80f : SUCCESS in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eaca6fa117574f1dbc5e22ad3c231abb : SUCCESS in 3m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b91e64552a234c1a9be66890b39ef6c3 : SUCCESS in 3m 35s (non-voting)\n- refstack-tox-functional https://zuul.opendev.org/t/openstack/build/cb95800cf0314891a3dfa8d0145c3cb3 : SUCCESS in 4m 11s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/6d71234ed6af4108b60c425dc1aec6c8 : SUCCESS in 2m 36s","accounts_in_message":[],"_revision_number":2},{"id":"0be93c6c1fbffeec45242d89e035d53645c56c90","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"date":"2023-04-16 19:32:30.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"c865aec70c9091b4c57abf3f0cd92b12409027ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-16 19:43:27.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dc180ceb485b47e091be466734f143a0\n\n- nodejs-run-lint https://zuul.opendev.org/t/openstack/build/5b43c50be633438ab166b3cad78b0ef0 : SUCCESS in 3m 20s\n- nodejs-run-test-browser https://zuul.opendev.org/t/openstack/build/0b81d4ec256247a283029c1035e18e4f : SUCCESS in 4m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9735e1293bfc4481b2dc6daff44b8e03 : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56e968cf609248dda0d0c02f824e48c5 : FAILURE in 3m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1235d53afcf148b9b2d781eb0395a954 : SUCCESS in 3m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/577a8c65794f4e8c8409d21ce5f63525 : SUCCESS in 3m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c42e7237c10d4bb0a240a8c492215ea4 : SUCCESS in 3m 28s (non-voting)\n- refstack-tox-functional https://zuul.opendev.org/t/openstack/build/de1a03dac6874f049c9f3e8fd414f4ce : SUCCESS in 4m 53s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/a1a0c191ae2749aeb750a4f3cb4dc5f7 : SUCCESS in 2m 31s","accounts_in_message":[],"_revision_number":3},{"id":"72f52bbf473a21850d7d1f78d764dc20d53c30c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"date":"2023-04-17 11:30:44.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"7ce04e7fde7f947eac0616a8f2711cb2f9477b15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 11:43:23.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/51999070c37c4c38ba99b822437cf217\n\n- nodejs-run-lint https://zuul.opendev.org/t/openstack/build/7106c13f50ee49e1bed05aef31a2ee70 : SUCCESS in 6m 06s\n- nodejs-run-test-browser https://zuul.opendev.org/t/openstack/build/690d4419408e4e55914e8f7197237e5a : SUCCESS in 5m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d21317ea84eb424ca9135f0b99a7305a : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca362ce3de3544659d7878c44349a007 : FAILURE in 3m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7fe09bacd24f4fd8a9e8948f62e1a305 : SUCCESS in 3m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9e47cd9b4348410989a05c7fabfd0f18 : SUCCESS in 3m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/905cf473a13a408e8e1f7c055f63e2ec : SUCCESS in 3m 45s (non-voting)\n- refstack-tox-functional https://zuul.opendev.org/t/openstack/build/3f5294b6d8ac46d4860e80ddc4aa35ce : SUCCESS in 3m 51s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/fa9c2f9b922f427bafa8d3a0981ab6ab : SUCCESS in 2m 45s","accounts_in_message":[],"_revision_number":4},{"id":"045e930a3b84c453bb521e092b73a71d97b74394","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2023-04-18 13:43:27.000000000","message":"Patch Set 4: Code-Review-1\n\n(9 comments)","accounts_in_message":[],"_revision_number":4},{"id":"d47f50f32df063cf559c9795c7cfe74147fed387","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"date":"2023-04-18 14:06:08.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"62ff0fbae30072cefdca9bc0ca562064f0e2ac0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-18 14:20:08.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51c4e51dc654417d89ec6585c598ce4c\n\n- nodejs-run-lint https://zuul.opendev.org/t/openstack/build/a1886c64fd0b45e78a181f2d3abf24ab : SUCCESS in 6m 15s\n- nodejs-run-test-browser https://zuul.opendev.org/t/openstack/build/ded97918f1214e95bc39a79d7e45e2bc : SUCCESS in 9m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7b10b3986b9d40e18ba59d8c4d784e6b : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae21602cab4e4e54a546f1a200ba5b59 : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c1aed8db6190495a9a9723b25d31b9dc : SUCCESS in 5m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cb24aeb6f9cf4e3091f429f309772603 : SUCCESS in 7m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f1add6f710444571b54854921b10fc53 : SUCCESS in 4m 40s (non-voting)\n- refstack-tox-functional https://zuul.opendev.org/t/openstack/build/c4d0fd3b30f444499e30013b3af9cf4e : SUCCESS in 5m 02s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/971c93e80aec471086c348aacd3e61dd : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":5},{"id":"7ef62dfe8f816e89c6676bde6eff4806039163f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2023-04-18 16:15:50.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"94a9aa80ba0034f40c6bf46947fbafaf1ad2ea2c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2023-04-18 16:26:03.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"91968b14dea2fe9954cd5995c632cecb14e76aee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-18 16:39:37.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be4d035f4cf34688a3a95d79bb27a091\n\n- nodejs-run-lint https://zuul.opendev.org/t/openstack/build/6d9a9be67c4e4a2689bffd4b358359f8 : SUCCESS in 3m 16s\n- nodejs-run-test-browser https://zuul.opendev.org/t/openstack/build/96506f0e0065426bb57ee04398585f86 : SUCCESS in 6m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/09d184c119ff4f58a09c0d5d891da94f : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f1f9821c4e841d0a476ef7df83de273 : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/119e927cb900452182a8c8fd0625643c : SUCCESS in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6278c6d639854e3fbca9958001d570ff : SUCCESS in 3m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0cc95ac375eb47bba9416560c4139b71 : SUCCESS in 3m 37s (non-voting)\n- refstack-tox-functional https://zuul.opendev.org/t/openstack/build/b6ff6d4a09e848ceb91f4d09d0e646e4 : SUCCESS in 4m 25s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/1ffb76e383e84523b48eff9330ff2670 : SUCCESS in 2m 54s","accounts_in_message":[],"_revision_number":7},{"id":"33289a93f5e8d0fb4eea30c998b46fd93c8880dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2023-04-18 16:54:07.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"29e424ee709918460c6ca8bb400f178fd0b0036c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2023-04-18 17:08:08.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"44cf243255519e2663b2d9daa10a0576c86be30a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-18 17:22:25.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8395edd864ec4ef8993613e5f433fb7d\n\n- nodejs-run-lint https://zuul.opendev.org/t/openstack/build/c62e95d9f20948bd8719cc3fcb9e8a3f : SUCCESS in 3m 34s\n- nodejs-run-test-browser https://zuul.opendev.org/t/openstack/build/ad9dda56c6fb4d88a2015fec446d0334 : SUCCESS in 6m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b3a10c9f07c54e729daed1665f37b779 : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5623c85e7364592aab5e02daf884246 : SUCCESS in 3m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ac65d13677084c15a55e2fd0f48455cc : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/22295a4d3c3347a6ad74dcf4916f1d46 : SUCCESS in 3m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0f4364fefb544feeab0aacae10e1e5f0 : SUCCESS in 5m 08s (non-voting)\n- refstack-tox-functional https://zuul.opendev.org/t/openstack/build/331c96682e08494f8590517eeccdc70e : SUCCESS in 4m 46s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/9b1be751c3894ea39b1313f3ca1f90e2 : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":9},{"id":"5a5d98fa67e8fe6b6b9f569938b10af37a1f96f2","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-04-19 13:26:17.000000000","message":"Patch Set 9:\n\n(10 comments)","accounts_in_message":[],"_revision_number":9},{"id":"c3396bbe3bf1b913f34d5c6ff5ae3e1215693b1a","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2023-04-19 15:52:30.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"b6791df1a23d7cf7a86fc96ba154ea7e2b58d5cd","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-04-20 08:26:30.000000000","message":"Patch Set 9: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"c852a6137fc781b5fe3a3049daf80cd176a822b5","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2023-05-09 14:42:43.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"5132d12a9082d8c3f381712eb355911ffeb80700","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 14:43:04.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"09a2b84bda708b5779421cd0ada9130672f139d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 14:56:38.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/664c78110a444e7591494f67679ef721\n\n- nodejs-run-lint https://zuul.opendev.org/t/openstack/build/8b3d0dd3efa648acbfd71b5740729f82 : SUCCESS in 3m 43s\n- nodejs-run-test-browser https://zuul.opendev.org/t/openstack/build/167d2cc804104a1b9428efce517ece35 : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/879e999f4e364f769b3203c3493ffdb4 : SUCCESS in 3m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a5d956a77d343849fe62059f0263055 : SUCCESS in 3m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9130e28ff755407a947d7edfb6bdc3a6 : SUCCESS in 3m 19s\n- refstack-tox-functional https://zuul.opendev.org/t/openstack/build/b387afadea18480f8c76606f85541998 : SUCCESS in 5m 33s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/3a7000cc1fee4ca4aadfb7401204f5a6 : SUCCESS in 3m 01s","accounts_in_message":[],"_revision_number":9},{"id":"3436469082a41848445f1a1aeb817a2e33c1e180","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 14:56:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"dfa3f4810076c61a70898881454d77633b195995","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 14:57:40.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d9affbfdd1d491c8b9291832734cbb7\n\n- opendev-promote-docs https://zuul.opendev.org/t/openstack/build/0b581e65aa6c400d8d00d69f7a155a87 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"9e79b3b4ae7fb5fe810c59fb9c1ab65155cf2cf6","revisions":{"7dc6a1913311b479615d53de4207ca07455d5119":{"kind":"REWORK","_number":1,"created":"2023-04-11 10:26:18.000000000","uploader":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"ref":"refs/changes/34/880034/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack","ref":"refs/changes/34/880034/1","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/34/880034/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/34/880034/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/34/880034/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack refs/changes/34/880034/1"}}},"commit":{"parents":[{"commit":"aa239130d6a28521682b5261653917c25b7d1ad8","subject":"Merge \"Fix security vulnerabilities using Bandit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/aa239130d6a28521682b5261653917c25b7d1ad8"}]}],"author":{"name":"jskunda","email":"jskunda@redhat.com","date":"2023-04-10 21:47:21.000000000","tz":120},"committer":{"name":"jskunda","email":"jskunda@redhat.com","date":"2023-04-11 10:26:03.000000000","tz":120},"subject":"Fix bug in refstack-ui","message":"Fix bug in refstack-ui\n\nRefstack webpage is not rendered correctly. The bug is tracked in\nhere: https://issues.redhat.com/browse/OSP-22171\n\nThis patch will update documentation build process to generate html\nfiles.\n- Modify convert-docs.py to use Sphinx instead of Docutils\n- Change paths to html files in about.html\n\nChange-Id: Id296b9cecf75bb289b0da41bec880a13694298e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/7dc6a1913311b479615d53de4207ca07455d5119"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/7dc6a1913311b479615d53de4207ca07455d5119"}]},"branch":"refs/heads/master"},"f93d6f4317d86c04eba8c8271b2eed9fdd8af346":{"kind":"REWORK","_number":2,"created":"2023-04-11 12:30:54.000000000","uploader":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"ref":"refs/changes/34/880034/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack","ref":"refs/changes/34/880034/2","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/34/880034/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/34/880034/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/34/880034/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack refs/changes/34/880034/2"}}},"commit":{"parents":[{"commit":"aa239130d6a28521682b5261653917c25b7d1ad8","subject":"Merge \"Fix security vulnerabilities using Bandit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/aa239130d6a28521682b5261653917c25b7d1ad8"}]}],"author":{"name":"jskunda","email":"jskunda@redhat.com","date":"2023-04-10 21:47:21.000000000","tz":120},"committer":{"name":"jskunda","email":"jskunda@redhat.com","date":"2023-04-11 12:30:31.000000000","tz":120},"subject":"Fix bug in refstack-ui","message":"Fix bug in refstack-ui\n\nRefstack webpage is not rendered correctly. The bug is tracked in\nhere: https://issues.redhat.com/browse/OSP-22171\n\nThis patch will update documentation build process to generate html\nfiles.\n- Modify convert-docs.py to use Sphinx instead of Docutils\n- Change paths to html files in about.html\n\nChange-Id: Id296b9cecf75bb289b0da41bec880a13694298e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/f93d6f4317d86c04eba8c8271b2eed9fdd8af346"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/f93d6f4317d86c04eba8c8271b2eed9fdd8af346"}]},"branch":"refs/heads/master"},"9f58d94399d08b3420e1f59ba27394823377ac71":{"kind":"REWORK","_number":3,"created":"2023-04-16 19:32:30.000000000","uploader":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"ref":"refs/changes/34/880034/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack","ref":"refs/changes/34/880034/3","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/34/880034/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/34/880034/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/34/880034/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack refs/changes/34/880034/3"}}},"commit":{"parents":[{"commit":"aa239130d6a28521682b5261653917c25b7d1ad8","subject":"Merge \"Fix security vulnerabilities using Bandit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/aa239130d6a28521682b5261653917c25b7d1ad8"}]}],"author":{"name":"jskunda","email":"jskunda@redhat.com","date":"2023-04-10 21:47:21.000000000","tz":120},"committer":{"name":"jskunda","email":"jskunda@redhat.com","date":"2023-04-16 19:32:28.000000000","tz":120},"subject":"Fix bug in refstack-ui","message":"Fix bug in refstack-ui\n\nRefstack webpage is not rendered correctly. The bug is tracked in\nhere: https://issues.redhat.com/browse/OSP-22171\n\nThis patch will update documentation build process to generate html\nfiles.\n- Modify convert-docs.py to use Sphinx instead of Docutils\n- Change paths to html files in about.html\n\nChange-Id: Id296b9cecf75bb289b0da41bec880a13694298e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/9f58d94399d08b3420e1f59ba27394823377ac71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/9f58d94399d08b3420e1f59ba27394823377ac71"}]},"branch":"refs/heads/master"},"a609b8451e0035e1ab94da32ef6f60044793196f":{"kind":"REWORK","_number":4,"created":"2023-04-17 11:30:44.000000000","uploader":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"ref":"refs/changes/34/880034/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack","ref":"refs/changes/34/880034/4","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/34/880034/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/34/880034/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/34/880034/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack refs/changes/34/880034/4"}}},"commit":{"parents":[{"commit":"aa239130d6a28521682b5261653917c25b7d1ad8","subject":"Merge \"Fix security vulnerabilities using Bandit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/aa239130d6a28521682b5261653917c25b7d1ad8"}]}],"author":{"name":"jskunda","email":"jskunda@redhat.com","date":"2023-04-10 21:47:21.000000000","tz":120},"committer":{"name":"jskunda","email":"jskunda@redhat.com","date":"2023-04-17 11:30:26.000000000","tz":120},"subject":"Fix bug in refstack-ui","message":"Fix bug in refstack-ui\n\nRefstack webpage is not rendered correctly. The bug is tracked in\nhere: https://issues.redhat.com/browse/OSP-22171\n\nThis patch will update documentation build process to generate html\nfiles.\n- Modify convert-docs.py to use Sphinx instead of Docutils\n- Change paths to html files in about.html\n\nChange-Id: Id296b9cecf75bb289b0da41bec880a13694298e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/a609b8451e0035e1ab94da32ef6f60044793196f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/a609b8451e0035e1ab94da32ef6f60044793196f"}]},"branch":"refs/heads/master"},"362ad7e8d8a6fae047df9b43177f108c96a72f5a":{"kind":"REWORK","_number":5,"created":"2023-04-18 14:06:08.000000000","uploader":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"ref":"refs/changes/34/880034/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack","ref":"refs/changes/34/880034/5","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/34/880034/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/34/880034/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/34/880034/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack refs/changes/34/880034/5"}}},"commit":{"parents":[{"commit":"aa239130d6a28521682b5261653917c25b7d1ad8","subject":"Merge \"Fix security vulnerabilities using Bandit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/aa239130d6a28521682b5261653917c25b7d1ad8"}]}],"author":{"name":"jskunda","email":"jskunda@redhat.com","date":"2023-04-10 21:47:21.000000000","tz":120},"committer":{"name":"jskunda","email":"jskunda@redhat.com","date":"2023-04-18 14:06:06.000000000","tz":120},"subject":"Fix bug in refstack-ui","message":"Fix bug in refstack-ui\n\nRefstack webpage\u0027s about section is not rendered correctly. The bug is\ntracked in here:\nhttps://tree.taiga.io/project/openstack-interop-working-group/issue/81\n\nThis patch will update documentation build process to generate html\nfiles.\n- Modify convert-docs.py to use Sphinx instead of Docutils\n- Change paths to html files in about.html\n- Remove docutils and beautifulsoup4 dependencies as they aren\u0027t used\n  in the code anymore\n\nChange-Id: Id296b9cecf75bb289b0da41bec880a13694298e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/362ad7e8d8a6fae047df9b43177f108c96a72f5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/362ad7e8d8a6fae047df9b43177f108c96a72f5a"}]},"branch":"refs/heads/master"},"5cf1362c0b679baf2fd2af9ee71232e68211637e":{"kind":"REWORK","_number":6,"created":"2023-04-18 16:15:50.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/34/880034/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack","ref":"refs/changes/34/880034/6","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/34/880034/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/34/880034/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/34/880034/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack refs/changes/34/880034/6"}}},"commit":{"parents":[{"commit":"aa239130d6a28521682b5261653917c25b7d1ad8","subject":"Merge \"Fix security vulnerabilities using Bandit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/aa239130d6a28521682b5261653917c25b7d1ad8"}]}],"author":{"name":"jskunda","email":"jskunda@redhat.com","date":"2023-04-10 21:47:21.000000000","tz":120},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2023-04-18 16:14:56.000000000","tz":120},"subject":"Fix bug in refstack-ui","message":"Fix bug in refstack-ui\n\nRefstack webpage\u0027s about section is not rendered correctly. The bug is\ntracked in here:\nhttps://tree.taiga.io/project/openstack-interop-working-group/issue/81\n\nThis patch will update documentation build process to generate html\nfiles by changing paths to html files in about.html\n\nReplace deprecated options in doc/source/conf.py by the new ones:\n* openstackdocs_bug_project\n* openstackdocs_bug_tag\n\nChange-Id: Id296b9cecf75bb289b0da41bec880a13694298e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/5cf1362c0b679baf2fd2af9ee71232e68211637e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/5cf1362c0b679baf2fd2af9ee71232e68211637e"}]},"branch":"refs/heads/master"},"16fb9b4c79555638bcf0d001f3bdb3485213a085":{"kind":"REWORK","_number":7,"created":"2023-04-18 16:26:03.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/34/880034/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack","ref":"refs/changes/34/880034/7","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/34/880034/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/34/880034/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/34/880034/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack refs/changes/34/880034/7"}}},"commit":{"parents":[{"commit":"aa239130d6a28521682b5261653917c25b7d1ad8","subject":"Merge \"Fix security vulnerabilities using Bandit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/aa239130d6a28521682b5261653917c25b7d1ad8"}]}],"author":{"name":"jskunda","email":"jskunda@redhat.com","date":"2023-04-10 21:47:21.000000000","tz":120},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2023-04-18 16:25:47.000000000","tz":120},"subject":"Fix bug in refstack-ui","message":"Fix bug in refstack-ui\n\nRefstack webpage\u0027s about section is not rendered correctly. The bug is\ntracked in here:\nhttps://tree.taiga.io/project/openstack-interop-working-group/issue/81\n\nThis patch will update documentation build process to generate html\nfiles by changing paths to html files in about.html\n\nReplace deprecated options in doc/source/conf.py by the new ones:\n* openstackdocs_bug_project\n* openstackdocs_bug_tag\n\nChange-Id: Id296b9cecf75bb289b0da41bec880a13694298e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/16fb9b4c79555638bcf0d001f3bdb3485213a085"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/16fb9b4c79555638bcf0d001f3bdb3485213a085"}]},"branch":"refs/heads/master"},"72037ddbe15ad6b8d8d5651499be3e3a8050a409":{"kind":"REWORK","_number":8,"created":"2023-04-18 16:54:07.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/34/880034/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack","ref":"refs/changes/34/880034/8","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/34/880034/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/34/880034/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/34/880034/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack refs/changes/34/880034/8"}}},"commit":{"parents":[{"commit":"aa239130d6a28521682b5261653917c25b7d1ad8","subject":"Merge \"Fix security vulnerabilities using Bandit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/aa239130d6a28521682b5261653917c25b7d1ad8"}]}],"author":{"name":"jskunda","email":"jskunda@redhat.com","date":"2023-04-10 21:47:21.000000000","tz":120},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2023-04-18 16:51:46.000000000","tz":120},"subject":"Fix missing about content in refstack-ui","message":"Fix missing about content in refstack-ui\n\nRefstack webpage\u0027s about section is not rendered correctly. The bug is\ntracked in here:\nhttps://tree.taiga.io/project/openstack-interop-working-group/issue/81\n\nThis patch will update documentation build process to generate html\nfiles by changing paths to html files in about.html.\n\nThe patch also adds further logic to the convert-docs.py to ignore\ndirectories and non .rst files to make passing a bigger nested\nstructure as an argument easier.\n\nReplace deprecated options in doc/source/conf.py by the new ones:\n* openstackdocs_bug_project\n* openstackdocs_bug_tag\n\nChange-Id: Id296b9cecf75bb289b0da41bec880a13694298e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/72037ddbe15ad6b8d8d5651499be3e3a8050a409"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/72037ddbe15ad6b8d8d5651499be3e3a8050a409"}]},"branch":"refs/heads/master"},"9e79b3b4ae7fb5fe810c59fb9c1ab65155cf2cf6":{"kind":"REWORK","_number":9,"created":"2023-04-18 17:08:08.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/34/880034/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack","ref":"refs/changes/34/880034/9","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/34/880034/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/34/880034/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/34/880034/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack refs/changes/34/880034/9"}}},"commit":{"parents":[{"commit":"aa239130d6a28521682b5261653917c25b7d1ad8","subject":"Merge \"Fix security vulnerabilities using Bandit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/aa239130d6a28521682b5261653917c25b7d1ad8"}]}],"author":{"name":"jskunda","email":"jskunda@redhat.com","date":"2023-04-10 21:47:21.000000000","tz":120},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2023-04-18 17:07:50.000000000","tz":120},"subject":"Fix missing about content in refstack-ui","message":"Fix missing about content in refstack-ui\n\nRefstack webpage\u0027s about section is not rendered correctly. The bug is\ntracked in here:\nhttps://tree.taiga.io/project/openstack-interop-working-group/issue/81\n\nThis patch will update documentation build process to generate html\nfiles by changing paths to html files in about.html.\n\nReplace deprecated options in doc/source/conf.py by the new ones:\n* openstackdocs_bug_project\n* openstackdocs_bug_tag\n\nChange-Id: Id296b9cecf75bb289b0da41bec880a13694298e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/9e79b3b4ae7fb5fe810c59fb9c1ab65155cf2cf6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/9e79b3b4ae7fb5fe810c59fb9c1ab65155cf2cf6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
