)]}'
{"id":"openinfra%2Frefstack~884426","triplet_id":"openinfra%2Frefstack~master~I10b726353bb1011170ec0cd479f8ff21c0bfe1b8","project":"openinfra/refstack","branch":"master","hashtags":[],"change_id":"I10b726353bb1011170ec0cd479f8ff21c0bfe1b8","subject":"Merge branch \u0027master\u0027 of https://opendev.org/openinfra/refstack","status":"ABANDONED","created":"2023-05-25 17:20:32.000000000","updated":"2023-06-01 16:12:55.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8016b97ee73d4f193edb10f7e750a9de0ce5e3cb","_number":884426,"virtual_id_number":884426,"owner":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-05-25 17:30:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-25 17:30:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7a291eee0bb227922a341861c4b81b34e443e10e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2023-05-25 17:20:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bd5f629ff19f8d52a25a0cbc96a9447f16a1f2d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 17:30:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2c0e88af44941edb56f20cba4f4b05c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4c8f95517f5b429f80917459c1f2f62c : SUCCESS in 2m 24s\n- nodejs-run-lint https://zuul.opendev.org/t/openstack/build/6b1bdacbc1e44aef8fae12e63e11770b : SUCCESS in 3m 25s\n- nodejs-run-test-browser https://zuul.opendev.org/t/openstack/build/417bec20a1fe47d4bfa6bd6cf9589953 : SUCCESS in 6m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e5d5ecab074b47e88e4f6d0724308b83 : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a37844cb8e544b1a3610129591c1dca : SUCCESS in 3m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/abfe0d59f82d4c06883c01586342b2b3 : SUCCESS in 3m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cfceb0c9cd3542928bd5667c591c320d : SUCCESS in 4m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/89670a99569f4d2cb4e97cf93a033e48 : SUCCESS in 3m 25s\n- refstack-tox-functional https://zuul.opendev.org/t/openstack/build/f28dc94442764dda9d7645e2e65b5970 : SUCCESS in 4m 14s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/3ecaf7fa67b64badb345c87ea5328b92 : SUCCESS in 2m 45s","accounts_in_message":[],"_revision_number":1},{"id":"8016b97ee73d4f193edb10f7e750a9de0ce5e3cb","tag":"autogenerated:gerrit:abandon","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2023-06-01 16:12:55.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"10b6f6acbcc8cd6753d3377e1bf300ec1754cb15","revisions":{"10b6f6acbcc8cd6753d3377e1bf300ec1754cb15":{"kind":"REWORK","_number":1,"created":"2023-05-25 17:20:32.000000000","uploader":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"ref":"refs/changes/26/884426/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack","ref":"refs/changes/26/884426/1","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/26/884426/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/26/884426/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/26/884426/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack refs/changes/26/884426/1"}}},"commit":{"parents":[{"commit":"7c4a2dcc31afb6fa4d1ef6292b118b9984c385e2","subject":"Fixed unresolved link, \u0027one-click setup\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/7c4a2dcc31afb6fa4d1ef6292b118b9984c385e2"}]},{"commit":"620d54c00f4a44e950dc2478b66521e8378474d6","subject":"Merge \"Adding 2 missing programs to menu on home page\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/620d54c00f4a44e950dc2478b66521e8378474d6"}]}],"author":{"name":"Vida Haririan","email":"vhariria@redhat.com","date":"2023-05-24 11:55:51.000000000","tz":-240},"committer":{"name":"Vida Haririan","email":"vhariria@redhat.com","date":"2023-05-24 11:55:51.000000000","tz":-240},"subject":"Merge branch \u0027master\u0027 of https://opendev.org/openinfra/refstack","message":"Merge branch \u0027master\u0027 of https://opendev.org/openinfra/refstack\n\nChange-Id: I10b726353bb1011170ec0cd479f8ff21c0bfe1b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/10b6f6acbcc8cd6753d3377e1bf300ec1754cb15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/10b6f6acbcc8cd6753d3377e1bf300ec1754cb15"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
