)]}'
{"id":"openinfra%2Frefstack~884427","triplet_id":"openinfra%2Frefstack~master~I1cc309c9a1bec622aadfcf0840c546d3d243e7c5","project":"openinfra/refstack","branch":"master","attention_set":{},"removed_from_attention_set":{"30674":{"account":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"last_update":"2023-06-08 16:09:46.000000000","reason":"Change was submitted"},"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2023-06-08 16:09:46.000000000","reason":"Change was submitted"},"22873":{"account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"last_update":"2023-06-08 16:03:00.000000000","reason":"\u003cGERRIT_ACCOUNT_22873\u003e replied on the change","reason_account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}}},"hashtags":[],"change_id":"I1cc309c9a1bec622aadfcf0840c546d3d243e7c5","subject":"Update Refstack Docs","status":"MERGED","created":"2023-05-25 17:20:32.000000000","updated":"2023-06-08 16:10:38.000000000","submitted":"2023-06-08 16:09:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"884427","meta_rev_id":"5221be9605dda76bbae6292eadb40ae4e32d7c58","_number":884427,"virtual_id_number":884427,"owner":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-08 16:09:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":2,"date":"2023-06-08 16:03:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":1,"date":"2023-06-08 16:03:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-25 17:29:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-01 16:19:44.000000000","updated_by":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"reviewer":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"state":"CC"},{"updated":"2023-06-01 16:47:05.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"}],"messages":[{"id":"bbe79fb9d5f4b6f659c522067d3d45d5ac176c57","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2023-05-25 17:20:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"94c1dcc981766c8f4571ae6b340b9b4673c67665","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 17:29:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52e57887f7674f6c934c1f0a08c1cea1\n\n- nodejs-run-lint https://zuul.opendev.org/t/openstack/build/75b26f91bb6e4b93a170b264140e9331 : SUCCESS in 3m 24s\n- nodejs-run-test-browser https://zuul.opendev.org/t/openstack/build/a69a78a30c0046518cefef7e3a638e57 : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ac6f80e9c30448b996c51a145b297be : SUCCESS in 3m 54s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/a842946ad6ad4c0588f53d27a46fbf7e : SUCCESS in 2m 44s","accounts_in_message":[],"_revision_number":1},{"id":"3cded5ede82c2ce0aaedd835f16c8b0605e91fed","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-06-01 16:19:44.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ffba3e362cc5f748f4c50be3c6f61bfccd10b618","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2023-06-01 16:47:05.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"bb4c1e4871a1587f026b067b241f2ff148daca7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2023-06-02 11:43:43.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"38d8bb7bee6cc01e506e83315ed0b835f342168c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-02 11:50:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a78cd8c94694ddc8c2f8a7e38f5fd22\n\n- nodejs-run-lint https://zuul.opendev.org/t/openstack/build/a98f00fe382642bf8b70ee5614cc3435 : SUCCESS in 3m 32s\n- nodejs-run-test-browser https://zuul.opendev.org/t/openstack/build/68f883f86eff457aa84acd1baa6e88c4 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/740771360f78486cb97b797596d18ae9 : SUCCESS in 3m 32s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/23f9eb6447ff41c08ba7751ed3f5f6e2 : SUCCESS in 3m 01s","accounts_in_message":[],"_revision_number":2},{"id":"774f75765e9ee3c09d67d8cfd25f3729b47ff286","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2023-06-05 15:51:07.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"7cdd6feb9daa5d7ea636d8d7a6f2c52524bf2158","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-05 15:58:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f8951c7c1944ccd8b07af22e7e4550a\n\n- nodejs-run-lint https://zuul.opendev.org/t/openstack/build/7cbd86c9596647eab16e68ae1ccbb4ba : SUCCESS in 3m 34s\n- nodejs-run-test-browser https://zuul.opendev.org/t/openstack/build/8d693f820dcc42d79ff56aadc5bb3f22 : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9cd92236e144768912f1843252d63fd : SUCCESS in 3m 10s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/060fbdb58a76454fac9b09baa423fef5 : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":3},{"id":"f7cc9b67981bacd2800ce679f12bb93da5e12880","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2023-06-05 15:58:33.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6eed04591bd6303e7c1f3cd6c92a6cb7979ae279","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2023-06-06 14:25:21.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ca15db10912b5eeed86460f25ca453906ccd1230","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2023-06-08 16:03:00.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1968430016744bcbc0b9a156453074961c56a6cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-08 16:03:19.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"eb52cc6c265373bd7df9334771f91afdcb7c3167","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-08 16:09:46.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/565a63da159e469f8f34f50ab2a6eadb\n\n- nodejs-run-lint https://zuul.opendev.org/t/openstack/build/42790c08be9047eaa3f43fadc4f191ad : SUCCESS in 3m 39s\n- nodejs-run-test-browser https://zuul.opendev.org/t/openstack/build/5da12e92694d4dc2920d5e82f776402d : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b056269cd8b6480eb9c38ed33fe5454a : SUCCESS in 3m 30s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/88ea3567057745409a628e5b23fab391 : SUCCESS in 2m 29s","accounts_in_message":[],"_revision_number":3},{"id":"42d85c05166cd6b258345d019fd7a509a9c21f87","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-08 16:09:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"5221be9605dda76bbae6292eadb40ae4e32d7c58","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-08 16:10:38.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a12a37c0b8144242be412a4280f14df6\n\n- opendev-promote-docs https://zuul.opendev.org/t/openstack/build/43b89fc6927645269980b7eb04759518 : SUCCESS in 34s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5eac91ce6b1dcb01d38987173daaff73df7d9266","revisions":{"a7adb5a029c0147a7ff998d72cc7ab01b7252b4e":{"kind":"REWORK","_number":1,"created":"2023-05-25 17:20:32.000000000","uploader":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"ref":"refs/changes/27/884427/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack","ref":"refs/changes/27/884427/1","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/27/884427/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/27/884427/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/27/884427/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack refs/changes/27/884427/1"}}},"commit":{"parents":[{"commit":"10b6f6acbcc8cd6753d3377e1bf300ec1754cb15","subject":"Merge branch \u0027master\u0027 of https://opendev.org/openinfra/refstack","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/10b6f6acbcc8cd6753d3377e1bf300ec1754cb15"}]}],"author":{"name":"Vida Haririan","email":"vhariria@redhat.com","date":"2023-05-25 17:10:27.000000000","tz":-240},"committer":{"name":"Vida Haririan","email":"vhariria@redhat.com","date":"2023-05-25 17:18:35.000000000","tz":-240},"subject":"Update Refstack Docs","message":"Update Refstack Docs\n\nRemoved Option 2 install docs and associated files.\nRenamed Option1AnsiblePlaybook.rst to refstack_setup_via_ansible.rst\n, move to /refstack/doc/source\n\nChange-Id: I1cc309c9a1bec622aadfcf0840c546d3d243e7c5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/a7adb5a029c0147a7ff998d72cc7ab01b7252b4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/a7adb5a029c0147a7ff998d72cc7ab01b7252b4e"}]},"branch":"refs/heads/master"},"5c9f538cb10771d4d1c76c0fc16befa3159a4fd0":{"kind":"REWORK","_number":2,"created":"2023-06-02 11:43:43.000000000","uploader":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"ref":"refs/changes/27/884427/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack","ref":"refs/changes/27/884427/2","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/27/884427/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/27/884427/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/27/884427/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack refs/changes/27/884427/2"}}},"commit":{"parents":[{"commit":"620d54c00f4a44e950dc2478b66521e8378474d6","subject":"Merge \"Adding 2 missing programs to menu on home page\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/620d54c00f4a44e950dc2478b66521e8378474d6"}]}],"author":{"name":"Vida Haririan","email":"vhariria@redhat.com","date":"2023-05-25 17:10:27.000000000","tz":-240},"committer":{"name":"Vida Haririan","email":"vhariria@redhat.com","date":"2023-06-02 11:43:35.000000000","tz":-240},"subject":"Update Refstack Docs","message":"Update Refstack Docs\n\nRemoved Option 2 install docs and associated files.\nRenamed Option1AnsiblePlaybook.rst to refstack_setup_via_ansible.rst\n, move to /refstack/doc/source\n\nChange-Id: I1cc309c9a1bec622aadfcf0840c546d3d243e7c5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/5c9f538cb10771d4d1c76c0fc16befa3159a4fd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/5c9f538cb10771d4d1c76c0fc16befa3159a4fd0"}]},"branch":"refs/heads/master"},"5eac91ce6b1dcb01d38987173daaff73df7d9266":{"kind":"REWORK","_number":3,"created":"2023-06-05 15:51:07.000000000","uploader":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"ref":"refs/changes/27/884427/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack","ref":"refs/changes/27/884427/3","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/27/884427/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/27/884427/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/27/884427/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack refs/changes/27/884427/3"}}},"commit":{"parents":[{"commit":"620d54c00f4a44e950dc2478b66521e8378474d6","subject":"Merge \"Adding 2 missing programs to menu on home page\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/620d54c00f4a44e950dc2478b66521e8378474d6"}]}],"author":{"name":"Vida Haririan","email":"vhariria@redhat.com","date":"2023-05-25 17:10:27.000000000","tz":-240},"committer":{"name":"Vida Haririan","email":"vhariria@redhat.com","date":"2023-06-05 15:50:39.000000000","tz":-240},"subject":"Update Refstack Docs","message":"Update Refstack Docs\n\nRemoved Option 2 install docs and associated files.\nRenamed Option1AnsiblePlaybook.rst to refstack_setup_via_ansible.rst\n, move to /refstack/doc/source\n\nChange-Id: I1cc309c9a1bec622aadfcf0840c546d3d243e7c5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/5eac91ce6b1dcb01d38987173daaff73df7d9266"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/5eac91ce6b1dcb01d38987173daaff73df7d9266"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
