)]}'
{"id":"openinfra%2Frefstack~926628","triplet_id":"openinfra%2Frefstack~master~I39d4db0033fb0c418525c41294f3c82162a7c666","project":"openinfra/refstack","branch":"master","attention_set":{},"removed_from_attention_set":{"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2025-09-02 12:13:16.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I39d4db0033fb0c418525c41294f3c82162a7c666","subject":"Replace deprecated LegacyEngineFacade","status":"ABANDONED","created":"2024-08-20 12:31:07.000000000","updated":"2025-09-02 12:13:16.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3b2eec1ab04cd5853c3a83b45964ef68763126de","_number":926628,"virtual_id_number":926628,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-08-21 08:50:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-20 12:40:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c9e020c91c4182a4fac5d131cba89be89e373b77","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-08-20 12:31:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"32a62c8ea14a9c64f329f7ca55fd5367ec40f3b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-20 12:40:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6f39a90974e5474b8b2f0fef84f2b6f3\n\n- nodejs-run-lint https://zuul.opendev.org/t/openstack/build/26e5c5a2ae8e4e0b987c937123c2d332 : SUCCESS in 3m 28s\n- nodejs-run-test-browser https://zuul.opendev.org/t/openstack/build/3fee46c9a48e4ff0b06acaf2754b5860 : SUCCESS in 6m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3d0a6ae8a7704f13a3f85fd779b435d8 : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2ebeca63b4849268c48ddbdb5325293 : SUCCESS in 2m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c7e5294756da4335a907e97c6b5424f8 : FAILURE in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/193c067b38a54d2f8940bf91b055f582 : FAILURE in 3m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/401a799cafb8414ab24779568ab6590d : FAILURE in 3m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e4f70e0f76954fe08c7d37239111494b : FAILURE in 5m 21s\n- refstack-tox-functional https://zuul.opendev.org/t/openstack/build/c63e262739044fe98c23489908a8c2cf : FAILURE in 3m 30s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/e94e65c60b0c436481b3a3b13567b132 : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":1},{"id":"f30282a77cc6a28c781c0b7a393953f4cc159493","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-08-20 13:33:26.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"f62fc136979feb2e3f90429c5937b9f12a7111ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-20 13:44:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4b2b81ed53c549db8674215442bab317\n\n- nodejs-run-lint https://zuul.opendev.org/t/openstack/build/9ce2073ae8e447898c486d8efa2c72a5 : SUCCESS in 3m 08s\n- nodejs-run-test-browser https://zuul.opendev.org/t/openstack/build/f05c4ae29b5343c99fcdc31ce0b3ec4c : SUCCESS in 5m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ea849cf1b3ae464d8daba4fe853b74df : SUCCESS in 3m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e962383f635d405893a195987fe12c59 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/91741689eebd4629b92e4e6b380a3303 : SUCCESS in 3m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bc0a1585a8374954a276c22ad4e89de1 : SUCCESS in 5m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/646b62669fab491995ab19c0164a36d0 : SUCCESS in 3m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/39427e55f78e4f778ed0f7d2db886240 : SUCCESS in 3m 42s\n- refstack-tox-functional https://zuul.opendev.org/t/openstack/build/a5014fc21cd34e17b650767af9675695 : FAILURE in 3m 11s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/5028896420f14052adea145ed9d6a6e0 : SUCCESS in 2m 52s","accounts_in_message":[],"_revision_number":2},{"id":"338867e54778bd2560d5eb2a7f2e286c600bfdbd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-08-20 14:03:47.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"08d9d88205266d663b6d80077023a7be62f70689","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-20 14:12:01.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7fc44e1d32fd4e88ab0d0d0044633b4b\n\n- nodejs-run-lint https://zuul.opendev.org/t/openstack/build/b587da0ce7084368a3ae16331b27d2ee : SUCCESS in 2m 51s\n- nodejs-run-test-browser https://zuul.opendev.org/t/openstack/build/add1f1ee680040e88816584234b70bf2 : SUCCESS in 5m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b3e3307414e14c61bb35ad09f30e74d4 : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7002cd63e7e448f8b82fbaba425cef4 : SUCCESS in 3m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/04f5b87bf77f442a950c10facc8ee921 : SUCCESS in 3m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/60c809c0b12a4234b6ac317250327ec1 : SUCCESS in 3m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/feba2d20009c4a7383fdff69944b8bce : SUCCESS in 4m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/26c44f9eb8644a14b347f32d0b50d2df : SUCCESS in 3m 46s\n- refstack-tox-functional https://zuul.opendev.org/t/openstack/build/fecb29ef019349b1892016249a6cce40 : FAILURE in 3m 35s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/5b6200ca3ab445279c1c240e0296ab31 : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":3},{"id":"796a9fed932b9df9e9e0a626526329792e0e4e62","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-08-21 08:39:55.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"c02cb7edd2b2f36df9ab5752645e409373428980","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 08:50:11.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b728b642125447b7bc2b692fb05cd9b7\n\n- nodejs-run-lint https://zuul.opendev.org/t/openstack/build/3e6d117160b146dfaa5b97dc6dd5c9ef : SUCCESS in 3m 00s\n- nodejs-run-test-browser https://zuul.opendev.org/t/openstack/build/3f96aced79fc414a94a6ed65038505df : SUCCESS in 5m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6d55410892b2410891e0fe36567a7011 : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69415eca7dda4c6787cf1b49e8ea269b : SUCCESS in 3m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6da3d6ac13414263874a337576b6d7c4 : SUCCESS in 3m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a370998601054dc9bffd195addc3ae97 : SUCCESS in 3m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6a6fc488d52d499aa1e415f5dc32c185 : SUCCESS in 3m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/331a34c335c342d6bd8600969734eed5 : SUCCESS in 3m 07s\n- refstack-tox-functional https://zuul.opendev.org/t/openstack/build/2182e602f9c04b099a7e9b99445d82aa : SUCCESS in 3m 43s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/c193860c19ab4752a96ec4a0a36b1ebd : SUCCESS in 2m 16s","accounts_in_message":[],"_revision_number":4},{"id":"3b2eec1ab04cd5853c3a83b45964ef68763126de","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-09-02 12:13:16.000000000","message":"Abandoned\n\nlooks like this service will be discontinued","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"1825245b974594fbf8a867bc3720a9ab1a654186","revisions":{"1fef6a009e7792c1d1d5f9aee2e9d971539e09a4":{"kind":"REWORK","_number":1,"created":"2024-08-20 12:31:07.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/28/926628/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack","ref":"refs/changes/28/926628/1","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/28/926628/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/28/926628/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/28/926628/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack refs/changes/28/926628/1"}}},"commit":{"parents":[{"commit":"fa704ce067271db16dc4059cf991e48e76e9607c","subject":"Include openstack-tox-py311 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/fa704ce067271db16dc4059cf991e48e76e9607c"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-08-20 12:30:59.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-08-20 12:31:06.000000000","tz":120},"subject":"Replace deprecated LegacyEngineFacade","message":"Replace deprecated LegacyEngineFacade\n\nLegacyEngineFacade was deprecated in oslo.db 1.12.0 which was released\nin 2015.\n\nChange-Id: I39d4db0033fb0c418525c41294f3c82162a7c666\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/1fef6a009e7792c1d1d5f9aee2e9d971539e09a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/1fef6a009e7792c1d1d5f9aee2e9d971539e09a4"}]},"branch":"refs/heads/master"},"ea755bdac7d17782a692442234b1ca85b7bdad98":{"kind":"REWORK","_number":2,"created":"2024-08-20 13:33:26.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/28/926628/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack","ref":"refs/changes/28/926628/2","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/28/926628/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/28/926628/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/28/926628/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack refs/changes/28/926628/2"}}},"commit":{"parents":[{"commit":"fa704ce067271db16dc4059cf991e48e76e9607c","subject":"Include openstack-tox-py311 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/fa704ce067271db16dc4059cf991e48e76e9607c"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-08-20 12:30:59.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-08-20 13:33:03.000000000","tz":120},"subject":"Replace deprecated LegacyEngineFacade","message":"Replace deprecated LegacyEngineFacade\n\nLegacyEngineFacade was deprecated in oslo.db 1.12.0 which was released\nin 2015.\n\nChange-Id: I39d4db0033fb0c418525c41294f3c82162a7c666\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/ea755bdac7d17782a692442234b1ca85b7bdad98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/ea755bdac7d17782a692442234b1ca85b7bdad98"}]},"branch":"refs/heads/master"},"ef479156cd7d3e1df0df58676232d2cbfbabf99a":{"kind":"REWORK","_number":3,"created":"2024-08-20 14:03:47.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/28/926628/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack","ref":"refs/changes/28/926628/3","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/28/926628/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/28/926628/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/28/926628/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack refs/changes/28/926628/3"}}},"commit":{"parents":[{"commit":"fa704ce067271db16dc4059cf991e48e76e9607c","subject":"Include openstack-tox-py311 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/fa704ce067271db16dc4059cf991e48e76e9607c"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-08-20 12:30:59.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-08-20 14:03:42.000000000","tz":120},"subject":"Replace deprecated LegacyEngineFacade","message":"Replace deprecated LegacyEngineFacade\n\nLegacyEngineFacade was deprecated in oslo.db 1.12.0 which was released\nin 2015.\n\nChange-Id: I39d4db0033fb0c418525c41294f3c82162a7c666\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/ef479156cd7d3e1df0df58676232d2cbfbabf99a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/ef479156cd7d3e1df0df58676232d2cbfbabf99a"}]},"branch":"refs/heads/master"},"1825245b974594fbf8a867bc3720a9ab1a654186":{"kind":"REWORK","_number":4,"created":"2024-08-21 08:39:55.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/28/926628/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack","ref":"refs/changes/28/926628/4","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/28/926628/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/28/926628/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/28/926628/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack refs/changes/28/926628/4"}}},"commit":{"parents":[{"commit":"fa704ce067271db16dc4059cf991e48e76e9607c","subject":"Include openstack-tox-py311 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/fa704ce067271db16dc4059cf991e48e76e9607c"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-08-20 12:30:59.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-08-21 08:39:49.000000000","tz":120},"subject":"Replace deprecated LegacyEngineFacade","message":"Replace deprecated LegacyEngineFacade\n\nLegacyEngineFacade was deprecated in oslo.db 1.12.0 which was released\nin 2015.\n\nChange-Id: I39d4db0033fb0c418525c41294f3c82162a7c666\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/1825245b974594fbf8a867bc3720a9ab1a654186"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/1825245b974594fbf8a867bc3720a9ab1a654186"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
