)]}'
{"id":"openinfra%2Frefstack~941177","triplet_id":"openinfra%2Frefstack~master~I00b8d0b2e54b7c08b560633e6a342249a58c9af3","project":"openinfra/refstack","branch":"master","attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-02-11 02:55:05.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I00b8d0b2e54b7c08b560633e6a342249a58c9af3","subject":"Update supported python versions","status":"NEW","created":"2025-02-11 02:36:00.000000000","updated":"2025-02-11 02:55:05.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8835c12d42a4393b37a032a5e5c811e271e0fcff","_number":941177,"virtual_id_number":941177,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-02-11 02:55:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-11 02:55:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0b4e3395fbc84727696dba4d45092e4abd081b17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-11 02:36:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8835c12d42a4393b37a032a5e5c811e271e0fcff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 02:55:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0a16a934239c4f3698af9afaf0ec4f6a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d0272af3889440a9be24584a8d54b03f : SUCCESS in 2m 46s\n- nodejs-run-lint https://zuul.opendev.org/t/openstack/build/4778c0d5d5c0451ba7a4c78de1aec273 : SUCCESS in 4m 10s\n- nodejs-run-test-browser https://zuul.opendev.org/t/openstack/build/b35e5c29e049489cb6a6d59528a10261 : SUCCESS in 5m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c656907a736846f48862dc6f5421cfa5 : SUCCESS in 6m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff486d025a054ab98e51e3183bbc9cf9 : SUCCESS in 6m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8950d6d0577748a5b0ecf7a566f4aef7 : SUCCESS in 5m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/853000fcfd5c4ea88b36152e0fdabe18 : SUCCESS in 5m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b834a1c349ad477f86dfacbf55757bf8 : SUCCESS in 3m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/721f438d2a40458196e07f5307b34c14 : FAILURE in 3m 44s\n- refstack-tox-functional https://zuul.opendev.org/t/openstack/build/f2ee4aa1e5ba405cae607a16f5f2d9ed : FAILURE in 3m 57s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/b0ab3be1770a498190e4b9f565c308cd : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"22a5963dfc31953ee020b0f14c39bae6371b4d2e","revisions":{"22a5963dfc31953ee020b0f14c39bae6371b4d2e":{"kind":"REWORK","_number":1,"created":"2025-02-11 02:36:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/77/941177/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack","ref":"refs/changes/77/941177/1","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/77/941177/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/77/941177/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/77/941177/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack refs/changes/77/941177/1"}}},"commit":{"parents":[{"commit":"fa704ce067271db16dc4059cf991e48e76e9607c","subject":"Include openstack-tox-py311 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/fa704ce067271db16dc4059cf991e48e76e9607c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-11 02:35:31.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-11 02:35:58.000000000","tz":540},"subject":"Update supported python versions","message":"Update supported python versions\n\n... according to the latest tested runtimes.\n\nChange-Id: I00b8d0b2e54b7c08b560633e6a342249a58c9af3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/22a5963dfc31953ee020b0f14c39bae6371b4d2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/22a5963dfc31953ee020b0f14c39bae6371b4d2e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
