)]}'
{"id":"openinfra%2Frefstack~944505","triplet_id":"openinfra%2Frefstack~master~I04ebd2271fd92376ff422b9722c0c26a65c2e702","project":"openinfra/refstack","branch":"master","topic":"python3.12","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2025-04-22 16:44:39.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I04ebd2271fd92376ff422b9722c0c26a65c2e702","subject":"Update db migrations to work with sqlalchemy 2.0","status":"ABANDONED","created":"2025-03-14 19:57:16.000000000","updated":"2025-04-22 16:44:39.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"43d408470c5b1a01b754dce29c728c7c5b4cdced","_number":944505,"virtual_id_number":944505,"owner":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-03-14 20:22:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-14 20:06:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0d76be38f8ddfe3dba74e6da5c8246abfbeb559f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-03-14 19:57:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2fbfe688120d3e48331ecf3eecf75bd5fbfd165b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-14 20:06:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cb75e2706aeb418b8fde469d89c758bd\n\n- nodejs-run-lint https://zuul.opendev.org/t/openstack/build/3e0613e2f95f467aa0f56b0d24734f7e : SUCCESS in 2m 35s\n- nodejs-run-test-browser https://zuul.opendev.org/t/openstack/build/f773cffb54e74b07b6a2c099f90305d9 : SUCCESS in 4m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5cb2938668a64b5880749a747125e70f : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/baed6af8586e4dfeb47455ec486b5d59 : SUCCESS in 2m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8cdb54912f834663a75cf693e7a55760 : FAILURE in 2m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/53afed5979d84b3d916b621e38e95d02 : SUCCESS in 2m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7a623fe523054273bbe566a3bdcf6821 : SUCCESS in 2m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/65d4bd67ad0e4adf90288e059258c027 : SUCCESS in 1m 53s\n- refstack-tox-functional https://zuul.opendev.org/t/openstack/build/50172d377fc8461fbcc9935c9d56947d : SUCCESS in 3m 40s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/376a947249f94851b18af6488e5fa460 : SUCCESS in 2m 20s","accounts_in_message":[],"_revision_number":1},{"id":"7974b514ebbbe4b0d227b94c050952006aa844a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-03-14 20:09:51.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"445dc37119b8cc76bfa16fdc7cda374b1d7ceb33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-14 20:22:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96034eb2094142d3be47a7b39cede25c\n\n- nodejs-run-lint https://zuul.opendev.org/t/openstack/build/c1d7e6c7e1ab46cb8274a4afeb26b7d2 : SUCCESS in 2m 52s\n- nodejs-run-test-browser https://zuul.opendev.org/t/openstack/build/92301aed36e24029b821b832763df36a : SUCCESS in 4m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ff478efb1989492bb935ab745da9eaf3 : SUCCESS in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd9d16747f66437fadc35470ccc08a27 : SUCCESS in 1m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/70a11fced2d844868a68dc947d49a2ed : SUCCESS in 3m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/80bb7895136747a9acf4899be06e6f3e : SUCCESS in 3m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/29bf075a17e14517884a008db3e023ed : SUCCESS in 3m 21s\n- refstack-tox-functional https://zuul.opendev.org/t/openstack/build/9bc802dc30b14f0897687ac2c14ed552 : SUCCESS in 3m 48s\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/da96615ac80f49d79bd3c8ed97368db3 : SUCCESS in 2m 25s","accounts_in_message":[],"_revision_number":2},{"id":"43d408470c5b1a01b754dce29c728c7c5b4cdced","tag":"autogenerated:gerrit:abandon","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-04-22 16:44:39.000000000","message":"Abandoned\n\nOpenDev is looking at winding down this service so sqlalchemy 2.0 updates are not necessary.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"673c6b2cb61153b9f82c5fdd948f02c05acbc7a7","revisions":{"7515967316d84d754d5e9ffead7bf4dd4ae88945":{"kind":"REWORK","_number":1,"created":"2025-03-14 19:57:16.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/05/944505/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack","ref":"refs/changes/05/944505/1","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/05/944505/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/05/944505/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/05/944505/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack refs/changes/05/944505/1"}}},"commit":{"parents":[{"commit":"fa704ce067271db16dc4059cf991e48e76e9607c","subject":"Include openstack-tox-py311 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/fa704ce067271db16dc4059cf991e48e76e9607c"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2025-03-14 19:55:53.000000000","tz":-420},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2025-03-14 19:56:54.000000000","tz":-420},"subject":"Update db migrations to work with sqlalchemy 2.0","message":"Update db migrations to work with sqlalchemy 2.0\n\nSqlaclhemy 2.0 requires that executed statemenst be of type\nsqlalchemy.text and not plain strings. Update the use of cursor.execute\nwithin db migrations to respect this requirement.\n\nChange-Id: I04ebd2271fd92376ff422b9722c0c26a65c2e702\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/7515967316d84d754d5e9ffead7bf4dd4ae88945"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/7515967316d84d754d5e9ffead7bf4dd4ae88945"}]},"branch":"refs/heads/master"},"673c6b2cb61153b9f82c5fdd948f02c05acbc7a7":{"kind":"REWORK","_number":2,"created":"2025-03-14 20:09:51.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/05/944505/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openinfra/refstack","ref":"refs/changes/05/944505/2","commands":{"Checkout":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/05/944505/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/05/944505/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openinfra/refstack refs/changes/05/944505/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openinfra/refstack refs/changes/05/944505/2"}}},"commit":{"parents":[{"commit":"fa704ce067271db16dc4059cf991e48e76e9607c","subject":"Include openstack-tox-py311 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/fa704ce067271db16dc4059cf991e48e76e9607c"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2025-03-14 19:55:53.000000000","tz":-420},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2025-03-14 20:04:45.000000000","tz":-420},"subject":"Update db migrations to work with sqlalchemy 2.0","message":"Update db migrations to work with sqlalchemy 2.0\n\nSqlaclhemy 2.0 requires that executed statemenst be of type\nsqlalchemy.text and not plain strings. Update the use of cursor.execute\nwithin db migrations to respect this requirement.\n\nNote we drop python3.8 testing as part of this change as the openstack\nconstraint for docutils is no longer python3.8 compatible. The\nproduction deployment currently runs python3.11 and is lookingat\npython3.12 so this should be fine.\n\nChange-Id: I04ebd2271fd92376ff422b9722c0c26a65c2e702\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/673c6b2cb61153b9f82c5fdd948f02c05acbc7a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openinfra/refstack/commit/673c6b2cb61153b9f82c5fdd948f02c05acbc7a7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
