)]}'
{"id":"openstack%2Fadjutant~885382","triplet_id":"openstack%2Fadjutant~stable%2Fxena~I1420f59cc5a315a47f7a4bfe0ac2a9761d06e11d","project":"openstack/adjutant","branch":"stable/xena","topic":"fix-zuul-errors","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-06-09 17:26:54.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_13252\u003e by clicking the attention icon","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}},"hashtags":[],"change_id":"I1420f59cc5a315a47f7a4bfe0ac2a9761d06e11d","subject":"Fix deprecated zuul queue syntax","status":"MERGED","created":"2023-06-06 18:50:32.000000000","updated":"2023-06-09 17:26:54.000000000","submitted":"2023-06-07 06:55:12.000000000","submitter":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"885382-fix-zuul-errors","meta_rev_id":"9e40c44729a7eaead4e732786661e0eab3bbb06c","_number":885382,"virtual_id_number":885382,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-06-07 06:55:01.000000000","_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-06-06 19:02:57.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-06-07 06:55:01.000000000","_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2023-06-07 06:55:01.000000000","_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-06 19:02:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5402206d312e2254da6ac13d2d99102e0852ccb5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-06-06 18:50:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8b68ab14cf203cbec2339c1f92e38cb8874c2bdc","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-06-06 18:51:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"51b6e25dadf5c2bc8c90ff2abe46a62964eea9d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-06 19:02:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8380a599097d4b38b58b55c4dc72568a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1b6f9bbcd2341c4b7a7d54355f3464f : SUCCESS in 3m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dda58fdaa82b4e9597a51954fc844925 : SUCCESS in 3m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a8f128cca71c416d8458393ada6456c2 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7336cdc91c5c4ffead53440d771c41c3 : FAILURE in 3m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f899913a4b0d4f028be053b98cc784ac : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b60c8b3f665e4c1e85652524f1f1eb5e : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9c542725d8854508b0673e4e91746829 : SUCCESS in 4m 13s (non-voting)\n- adjutant-black-style-check https://zuul.opendev.org/t/openstack/build/06a0d2c0d32349918237273cfed9ed2c : FAILURE in 6m 00s\n\nWarning:\n  Comments left for invalid file adjutant/actions/v1/users.py","accounts_in_message":[],"_revision_number":1},{"id":"6bb8887595ad08616331d9370fb724442d960a24","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-06-07 06:55:01.000000000","message":"Patch Set 1: Verified+2 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3a87ee5526c1950305d65295177064a08e43004c","tag":"autogenerated:gerrit:merged","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-06-07 06:55:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1db1d691edb9e0fe57407d1a00d3640c586c1db0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-07 06:58:23.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fdc1d8dc017440bb9a8aa2cb1a802040\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c0b440f3875415faef0b398634bab43 : FAILURE in 22s","accounts_in_message":[],"_revision_number":1},{"id":"4f418ba47bcd8b56ef6d8c76ddae0b271bfc34bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-07 07:06:40.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/eed736e8879e4b0d808f2c752874c164\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/659a30e457a04d9883f367fc51d030a4 : SUCCESS in 4m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/94a1c7c1d5a04c6987c8ad49d0f01ab6 : SUCCESS in 3m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8b1478a3d70f4c57ad4acb2e98d6b18c : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/861a11336b4346d09f1616f477e6e597 : FAILURE in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d85fb17fa6264d56a99e56abddca1b2c : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b159621f9af4d7982021611d4e52af5 : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1f09489b54f14d08a211cd15418cc6f1 : SUCCESS in 4m 03s (non-voting)\n- adjutant-black-style-check https://zuul.opendev.org/t/openstack/build/3cea711092c6492d8e693d0be6b7158a : FAILURE in 4m 16s\n\nWarning:\n  Comments left for invalid file adjutant/actions/v1/users.py","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"af4cf0debaf8eb7d98a20e267e475430ed7bc215","revisions":{"af4cf0debaf8eb7d98a20e267e475430ed7bc215":{"kind":"REWORK","_number":1,"created":"2023-06-06 18:50:32.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/82/885382/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/adjutant","ref":"refs/changes/82/885382/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/adjutant refs/changes/82/885382/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/adjutant refs/changes/82/885382/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/adjutant refs/changes/82/885382/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/adjutant refs/changes/82/885382/1"}}},"commit":{"parents":[{"commit":"f8cd1d888fc2335dbc9edcd63e4aa5ff54fac8b7","subject":"Fix an issue with Adjutant\u0027s Invite process not checking inherited roles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/adjutant/commit/f8cd1d888fc2335dbc9edcd63e4aa5ff54fac8b7"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2023-06-06 18:48:42.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2023-06-06 18:49:25.000000000","tz":120},"subject":"Fix deprecated zuul queue syntax","message":"Fix deprecated zuul queue syntax\n\nZuul deprecated (and now removed) declaring shared queues at a pipeline\nlevel [1]. This patch fixes the syntax to make the gate work for\nadjutant.\n\n[1] https://lists.openstack.org/pipermail/openstack-discuss/2022-May/028603.html\n\nChange-Id: I1420f59cc5a315a47f7a4bfe0ac2a9761d06e11d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/adjutant/commit/af4cf0debaf8eb7d98a20e267e475430ed7bc215"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/adjutant/commit/af4cf0debaf8eb7d98a20e267e475430ed7bc215"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
