)]}'
{"id":"openstack%2Faetos~957562","triplet_id":"openstack%2Faetos~master~I5f5c82ce25033748c321dd836256094fbcf305ef","project":"openstack/aetos","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-09-03 08:03:04.000000000","reason":"Change was submitted"},"32968":{"account":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"last_update":"2025-09-03 07:15:21.000000000","reason":"\u003cGERRIT_ACCOUNT_32968\u003e replied on the change","reason_account":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}},"4264":{"account":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"last_update":"2025-09-03 07:05:38.000000000","reason":"\u003cGERRIT_ACCOUNT_4264\u003e replied on the change","reason_account":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},"34975":{"account":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"last_update":"2025-09-03 08:03:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5f5c82ce25033748c321dd836256094fbcf305ef","subject":"Reduce Prometheus requests for modify_query calls","status":"MERGED","created":"2025-08-15 17:46:23.000000000","updated":"2025-09-03 08:04:20.000000000","submitted":"2025-09-03 08:03:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"957562","meta_rev_id":"6953908de6f9f63409bb364c716678e4a776df5b","_number":957562,"virtual_id_number":957562,"owner":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-03 08:03:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":2,"date":"2025-09-03 07:05:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":2,"date":"2025-09-03 07:15:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":1,"date":"2025-09-03 07:15:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-15 18:35:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-02 07:05:32.000000000","updated_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"reviewer":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"state":"REVIEWER"},{"updated":"2025-09-02 07:05:32.000000000","updated_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2025-09-02 07:05:32.000000000","updated_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"a471e75f93aef6fb8c6cc2eb0491c1d5d0112aae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2025-08-15 17:46:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d5757a13241c7567dd1bfee04d53480b3c981b59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-15 18:35:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf715165abda4ea89e517d14d2990b5e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3dd258d405d74200b8c002868235aaac : SUCCESS in 2m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c0689cc19b6149158788d1ff658dc574 : SUCCESS in 2m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/be334c785b8d44929af253c238a3eba6 : SUCCESS in 5m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5cde4ad3844749ebb5cb3230a96288f9 : SUCCESS in 6m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47d40f3339b74283af7a86f769a3b009 : SUCCESS in 4m 13s\n- requirements-check https://zuul.opendev.org/t/openstack/build/364e622f890643698bb539357681ac8e : SUCCESS in 1m 34s\n- telemetry-dsvm-integration-aetos https://zuul.opendev.org/t/openstack/build/5fbf512d63d6422a8e33fd120abc96d4 : SUCCESS in 44m 48s\n- telemetry-dsvm-integration-aetos-ipv6-only https://zuul.opendev.org/t/openstack/build/a126aca06dff40f79e9043ff5d1dd4c1 : SUCCESS in 44m 41s","accounts_in_message":[],"_revision_number":1},{"id":"ecd3f05a9bcfcebeb92ec15f24abde9fdbc0ce15","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2025-09-02 07:05:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0031efb0134d3077fe4a6e132a640606e4a4f154","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2025-09-03 07:05:38.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1209d0ef3994bd85629adc82651c6d864d69fc17","author":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"date":"2025-09-03 07:15:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c7ae11a55276131bc4117fd16dd62dadc9b53db2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 07:15:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d59c5ed7e2ea2e4d0bfdcc49e8d201104517a85c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 08:03:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb40d85c2ff849d9a37967d5af062130\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1eb56bbad1e4ebf80b044f5c8ae58b1 : SUCCESS in 2m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f164b462f13544f8a24f885e11d24b6b : SUCCESS in 2m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c27c3a628acd483f9fd28d58c356c173 : SUCCESS in 2m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f136336a6794b93af41a8ceec69be8f : SUCCESS in 3m 26s\n- requirements-check https://zuul.opendev.org/t/openstack/build/eedcb1d0405046c78b012e504eaeb657 : SUCCESS in 1m 57s\n- telemetry-dsvm-integration-aetos https://zuul.opendev.org/t/openstack/build/da9e9294e5744b91afb1ab6abd90f5d2 : SUCCESS in 42m 30s\n- telemetry-dsvm-integration-aetos-ipv6-only https://zuul.opendev.org/t/openstack/build/b91d08b566fd4f339d33230f8ace79a1 : SUCCESS in 30m 53s","accounts_in_message":[],"_revision_number":1},{"id":"e936f1325e80782c7ba23c2fd40cfb075fe84a5b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 08:03:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6953908de6f9f63409bb364c716678e4a776df5b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 08:04:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f90e847e9fc40e2b10216caabe9ffea\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9c7b8a3e95549249680ac57b8dd7205 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e563b9e0dc62e7779ec20b4455617cda113a3a11","revisions":{"e563b9e0dc62e7779ec20b4455617cda113a3a11":{"kind":"REWORK","_number":1,"created":"2025-08-15 17:46:23.000000000","uploader":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"ref":"refs/changes/62/957562/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aetos","ref":"refs/changes/62/957562/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aetos refs/changes/62/957562/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aetos refs/changes/62/957562/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aetos refs/changes/62/957562/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aetos refs/changes/62/957562/1"}}},"commit":{"parents":[{"commit":"918dcf4d148ae3a8d382271baeb42332b7ea2259","subject":"Validate input of [DEFAULT] auth_mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aetos/commit/918dcf4d148ae3a8d382271baeb42332b7ea2259"}]}],"author":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2025-08-13 19:33:50.000000000","tz":-240},"committer":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2025-08-13 19:42:21.000000000","tz":-240},"subject":"Reduce Prometheus requests for modify_query calls","message":"Reduce Prometheus requests for modify_query calls\n\nEach call to modify_query will send a request to Prometheus to\ndetermine all metric names saved in Prometheus unless we\nprovide the list with the modify_query call. We use the modify_query\nin two places in Aetos, one is the query controller, the other one\nis the process_matches function. In the query controller, we call\nmodify_query just once, so the code can stay as is. In process_matches\nwe call modify_query in a loop, so we send a request to Prometheus\nto retrieve a list of metrics before the loop and provide the list\nto modify_query to to avoid sending a request to Prometheus with each\niteration of the loop.\n\nChange-Id: I5f5c82ce25033748c321dd836256094fbcf305ef\nSigned-off-by: Jaromir Wysoglad \u003cjwysogla@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aetos/commit/e563b9e0dc62e7779ec20b4455617cda113a3a11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aetos/commit/e563b9e0dc62e7779ec20b4455617cda113a3a11"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
