)]}'
{"id":"openstack%2Faetos~977709","triplet_id":"openstack%2Faetos~master~Ifd03ac940c8541c45353bc5ec1444cc1bb62bb58","project":"openstack/aetos","branch":"master","attention_set":{},"removed_from_attention_set":{"32968":{"account":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"last_update":"2026-03-02 14:26:01.000000000","reason":"\u003cGERRIT_ACCOUNT_32968\u003e replied on the change","reason_account":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}},"34975":{"account":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"last_update":"2026-03-02 16:15:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifd03ac940c8541c45353bc5ec1444cc1bb62bb58","subject":"Add time and timeout parameters to query endpoint","status":"MERGED","created":"2026-02-23 18:58:21.000000000","updated":"2026-03-02 16:34:24.000000000","submitted":"2026-03-02 16:15:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"977709","meta_rev_id":"78f2ad2c3b0e03bd78dd1df9b74fb42186188c4c","_number":977709,"virtual_id_number":977709,"owner":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-02 16:15:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":2,"date":"2026-02-24 06:42:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":2,"date":"2026-03-02 14:26:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":1,"date":"2026-03-02 14:26:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-23 20:14:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-24 06:42:38.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2026-03-02 13:30:03.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"state":"REVIEWER"}],"messages":[{"id":"7c6f79bed6108b4e8213477a374cb84dc33917da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2026-02-23 18:58:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"69c1ba7ff1191bf6844e2bfa3544bc4929f80684","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 20:14:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab2c7bff764c417886c7979e57ba1d9d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fde24cea03e49afa319ece74a87ffb9 : SUCCESS in 2m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9f17aef8085f4c168d5ac5f604fc93b8 : SUCCESS in 3m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b583d923d90144b1b598d7760adbae5b : SUCCESS in 3m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1910c24c6f0e4b97b30454c8e5cb43cb : SUCCESS in 4m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84b8a79d3862465f80a97b08b8a6d69d : SUCCESS in 3m 52s\n- watcher-tempest-aetos https://zuul.opendev.org/t/openstack/build/16b4592fb5944c6f8e3c8f2ee1956bdc : SUCCESS in 1h 10m 41s\n- telemetry-dsvm-integration-aetos https://zuul.opendev.org/t/openstack/build/aa45456f502b466b822250389f2606b1 : SUCCESS in 25m 30s\n- telemetry-dsvm-integration-aetos-ipv6-only https://zuul.opendev.org/t/openstack/build/1c023c89ea2c406593e733a25027c8f1 : SUCCESS in 31m 53s","accounts_in_message":[],"_revision_number":1},{"id":"ed886275e0fd0ec0fa9ee3574cd70ff2ca34c761","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2026-02-24 06:42:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ac925b3be91d375154a9e141f3b6db66884e3e0d","author":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"date":"2026-03-02 14:26:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"adfbab9979b5205d19f4091afe3b1262f77c441d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 14:26:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8af819c93671e29a8153122dfd266eeba240a59b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 16:15:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9d29af69bab49a3b2b0c049019d861c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72fa2d7037d24528b3b2ef4dc7b961af : SUCCESS in 3m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/00fb6968c9f34017bde14b4f3e04577c : SUCCESS in 4m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8c77380d71d449c1873342d2d09d26ba : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1300dd8365da4c61ad37fa17b701e3a5 : SUCCESS in 3m 42s\n- watcher-tempest-aetos https://zuul.opendev.org/t/openstack/build/8ec5ac3d6ab844b2a890770f2a40ae6a : SUCCESS in 1h 22m 16s\n- telemetry-dsvm-integration-aetos https://zuul.opendev.org/t/openstack/build/4cf9348807b84425915fabcd83f786b2 : SUCCESS in 31m 22s\n- telemetry-dsvm-integration-aetos-ipv6-only https://zuul.opendev.org/t/openstack/build/5de21b8e734948468e3d9775a2cc30e5 : SUCCESS in 41m 37s","accounts_in_message":[],"_revision_number":1},{"id":"bac400eb48a05ff9aed5ea988b8eac16202ddc37","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 16:15:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"78f2ad2c3b0e03bd78dd1df9b74fb42186188c4c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 16:34:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49a893aa42fa42ba8ef985d6da1d5da8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76c97bacccbf41be817928454b61abba : SUCCESS in 2m 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6e2eb84bfbc0dde8f3c639463ceb56b74cb29f2e","revisions":{"6e2eb84bfbc0dde8f3c639463ceb56b74cb29f2e":{"kind":"REWORK","_number":1,"created":"2026-02-23 18:58:21.000000000","uploader":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"ref":"refs/changes/09/977709/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aetos","ref":"refs/changes/09/977709/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aetos refs/changes/09/977709/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aetos refs/changes/09/977709/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aetos refs/changes/09/977709/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aetos refs/changes/09/977709/1"}}},"commit":{"parents":[{"commit":"d370b9bd2afc22a3f5fea492ec8d07909d80c7ce","subject":"Update watcher aetos job name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aetos/commit/d370b9bd2afc22a3f5fea492ec8d07909d80c7ce"}]}],"author":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2026-02-23 18:52:05.000000000","tz":0},"committer":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2026-02-23 18:57:55.000000000","tz":0},"subject":"Add time and timeout parameters to query endpoint","message":"Add time and timeout parameters to query endpoint\n\nChange-Id: Ifd03ac940c8541c45353bc5ec1444cc1bb62bb58\nSigned-off-by: Jaromir Wysoglad \u003cjwysogla@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aetos/commit/6e2eb84bfbc0dde8f3c639463ceb56b74cb29f2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aetos/commit/6e2eb84bfbc0dde8f3c639463ceb56b74cb29f2e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
