)]}'
{"id":"openstack%2Faetos~977895","triplet_id":"openstack%2Faetos~master~I31331f3392dc28d4900e8521e074b8ae52176bbe","project":"openstack/aetos","branch":"master","attention_set":{},"removed_from_attention_set":{"32968":{"account":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"last_update":"2026-03-03 15:11:51.000000000","reason":"\u003cGERRIT_ACCOUNT_32968\u003e replied on the change","reason_account":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-03-03 14:32:58.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"34975":{"account":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"last_update":"2026-03-03 16:48:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I31331f3392dc28d4900e8521e074b8ae52176bbe","subject":"Remove global APPCONFIGS from app.py","status":"MERGED","created":"2026-02-24 22:15:08.000000000","updated":"2026-03-03 16:49:37.000000000","submitted":"2026-03-03 16:48:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"977895","meta_rev_id":"727561a5f7b55ddbc8063de33a94bb6e75bcac4c","_number":977895,"virtual_id_number":977895,"owner":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-03 16:48:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2026-03-03 14:32:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2026-03-03 15:11:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2026-03-03 15:11:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-24 23:13:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-03 14:00:31.000000000","updated_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"reviewer":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"state":"REVIEWER"},{"updated":"2026-03-03 14:00:31.000000000","updated_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"9e185bbb6d155cf9557761af1a73d200cdc0fa6b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2026-02-24 22:15:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a7057405bcd852169f73e7213d39fcc3c1006e9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 23:13:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f4385f43f19540e4a90fae6ffc84a6a0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2be121ae7adc4acd9c84e65ed26855de : SUCCESS in 4m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0c1f7df8d7984b7b90580c4b611b2fe1 : SUCCESS in 3m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/afd6bcbf0e3644229f7ef406a21b6294 : SUCCESS in 3m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2a7bdadc361f4fda804e42667c7f2aac : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/609daa62473d44868037475bb25c9e85 : SUCCESS in 4m 44s\n- watcher-tempest-aetos https://zuul.opendev.org/t/openstack/build/c14c98a03b3b42789ef6ce93bcf7d0b7 : FAILURE in 57m 27s\n- telemetry-dsvm-integration-aetos https://zuul.opendev.org/t/openstack/build/e06599ff07bf420981e25cd634bff740 : SUCCESS in 44m 41s\n- telemetry-dsvm-integration-aetos-ipv6-only https://zuul.opendev.org/t/openstack/build/d866a5053daf4bbb9cb4f1a573258912 : SUCCESS in 44m 41s","accounts_in_message":[],"_revision_number":1},{"id":"442865beb9eeed03b78c7c41feff755157878674","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2026-02-25 07:56:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ccca190ec963763b5a4022686a53382d085f20ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 08:52:38.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e1a99887f1e241308671b219ad8cd16a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f81ae79dd3e4cf3812e93701072973e : SUCCESS in 2m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/04b00af54f9d4bf5824baadc51d219a9 : SUCCESS in 3m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5bb285d986d94a6f9e355f92d84bfd61 : SUCCESS in 2m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2deaacad8d2b4e848931c8ec508744e2 : SUCCESS in 3m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2754e69e645b40edb02091b648702cc0 : SUCCESS in 1m 49s\n- watcher-tempest-aetos https://zuul.opendev.org/t/openstack/build/afb8860f3a6d41b3ac59ddf8f0c418cc : SUCCESS in 51m 31s\n- telemetry-dsvm-integration-aetos https://zuul.opendev.org/t/openstack/build/c343e7529b7a4486aca4e0296eea6f00 : FAILURE in 50m 36s\n- telemetry-dsvm-integration-aetos-ipv6-only https://zuul.opendev.org/t/openstack/build/a31cc52d00f3427fa8eb3a45d922b5fb : FAILURE in 40m 01s","accounts_in_message":[],"_revision_number":1},{"id":"478a86ef8ed0ee0c794505868527813940229373","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2026-02-25 09:35:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"51d59ed26ee53e874c360c6e5f4f44167c211e16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 11:16:03.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/94a9836802f945b1be07b9d64b51830d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e970f807f514a748a921b333b464efe : SUCCESS in 3m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0ddbb34e99d045dc9de4e8d9d8d5d4ee : SUCCESS in 3m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b3ce406992d14012bb363be34b379140 : SUCCESS in 4m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2e317f8dd3cb4c98a3cae213cf93ca5a : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1a1799f3cc74efb8020952823cf8d28 : SUCCESS in 4m 43s\n- watcher-tempest-aetos https://zuul.opendev.org/t/openstack/build/cd246d15200c4f80b09aae7c33d07014 : SUCCESS in 1h 28m 59s\n- telemetry-dsvm-integration-aetos https://zuul.opendev.org/t/openstack/build/c8f0a96090b34f5695d61710c83c85c9 : FAILURE in 50m 25s\n- telemetry-dsvm-integration-aetos-ipv6-only https://zuul.opendev.org/t/openstack/build/0334a80deb2d43569088f079d297cdfd : FAILURE in 54m 00s","accounts_in_message":[],"_revision_number":1},{"id":"7d4196ed02938d1918f4b15b447a111f15d815eb","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2026-02-27 08:04:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9e21d8007a67b93a0d286aed692ab6c55a8d4b49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 09:40:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ccf6223e9845406ab3f9d2bda9e0296a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d7dc050729647a6ac6f32759451a0fb : SUCCESS in 2m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/26f0a8b500764291ae3f2d8a988b8108 : SUCCESS in 3m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3537d4fcfaa7412d8b2b3d7d640fc8f1 : SUCCESS in 3m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d11359f2c1f34f30bdb6e807631b508c : SUCCESS in 6m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/653a3234c5cc44dda2dd932b739b2ef9 : SUCCESS in 3m 42s\n- watcher-tempest-aetos https://zuul.opendev.org/t/openstack/build/ce504f922e624ab7addd94d120798901 : SUCCESS in 1h 30m 43s\n- telemetry-dsvm-integration-aetos https://zuul.opendev.org/t/openstack/build/16c9c622a5de4b3ebf10f59d339761ac : SUCCESS in 44m 06s\n- telemetry-dsvm-integration-aetos-ipv6-only https://zuul.opendev.org/t/openstack/build/ce259fd1918c40cd875a6cf2c2ed1f88 : SUCCESS in 44m 09s","accounts_in_message":[],"_revision_number":1},{"id":"b73ce42ee8f3f06739993aedc3efbd31a6400907","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-03 14:32:58.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b75cfe064c7325d4439ac4edb248f3766f18e919","author":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"date":"2026-03-03 15:11:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fde2c8fa1472d6551c163e7204972f1539fd017e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 15:12:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"151cf6380988eb561ef31b0a0991e7428476a938","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 16:48:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/debcc5bfe8c042da9c773bc766171d97\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/699b659f7bc04016bcb3bc0e1514dc28 : SUCCESS in 2m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/91d6ce1704724d178b866b83c5601f54 : SUCCESS in 1m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/25345a6360324dd9b94801eede9c53a2 : SUCCESS in 2m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a3edd474d344bc78128fa9f47e2a9c0 : SUCCESS in 2m 02s\n- watcher-tempest-aetos https://zuul.opendev.org/t/openstack/build/1a5987d87c704526ab391bcbae614c27 : SUCCESS in 1h 30m 36s\n- telemetry-dsvm-integration-aetos https://zuul.opendev.org/t/openstack/build/71a6472bac594a159b76d2bc7398343d : SUCCESS in 34m 17s\n- telemetry-dsvm-integration-aetos-ipv6-only https://zuul.opendev.org/t/openstack/build/b7b58be5ff394e898977298d6a096754 : SUCCESS in 50m 00s","accounts_in_message":[],"_revision_number":1},{"id":"fa114b81910312c5c9f972014815f683879f164a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 16:48:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"727561a5f7b55ddbc8063de33a94bb6e75bcac4c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 16:49:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db664a2f1f1542a0864d851c2e097c9a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6ad9514ce854b119166acc1d7e06918 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8649ccfba47155dc98251240cfcf81d01f8fbc4f","revisions":{"8649ccfba47155dc98251240cfcf81d01f8fbc4f":{"kind":"REWORK","_number":1,"created":"2026-02-24 22:15:08.000000000","uploader":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"ref":"refs/changes/95/977895/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aetos","ref":"refs/changes/95/977895/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aetos refs/changes/95/977895/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aetos refs/changes/95/977895/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aetos refs/changes/95/977895/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aetos refs/changes/95/977895/1"}}},"commit":{"parents":[{"commit":"d370b9bd2afc22a3f5fea492ec8d07909d80c7ce","subject":"Update watcher aetos job name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aetos/commit/d370b9bd2afc22a3f5fea492ec8d07909d80c7ce"}]}],"author":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2026-02-24 22:13:02.000000000","tz":0},"committer":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2026-02-24 22:14:35.000000000","tz":0},"subject":"Remove global APPCONFIGS from app.py","message":"Remove global APPCONFIGS from app.py\n\nThere is no need for the global keyword with the way we\u0027re\naccessing the variable in the functions.\n\nChange-Id: I31331f3392dc28d4900e8521e074b8ae52176bbe\nSigned-off-by: Jaromir Wysoglad \u003cjwysogla@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aetos/commit/8649ccfba47155dc98251240cfcf81d01f8fbc4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aetos/commit/8649ccfba47155dc98251240cfcf81d01f8fbc4f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
