)]}'
{"id":"openstack%2Faetos~980505","triplet_id":"openstack%2Faetos~master~I8b577d72abe02ae39194839838e4d2bca9c01521","project":"openstack/aetos","branch":"master","hashtags":[],"change_id":"I8b577d72abe02ae39194839838e4d2bca9c01521","subject":"Optinally hide some labels from unprivileged users","status":"NEW","created":"2026-03-13 18:32:47.000000000","updated":"2026-03-13 19:43:35.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"16ac28feaaaf5563eed437d1117817105858205a","_number":980505,"virtual_id_number":980505,"owner":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-03-13 19:43:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"all":[{"value":-1,"date":"2026-03-13 18:33:26.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-13 19:43:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e05ea2cc62f3f3a3179639e75fbb22165ce00e04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2026-03-13 18:32:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f8cc4e30bdbbb6ed2396daa87d30b246989501e1","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2026-03-13 18:33:26.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4c3f8af831c1bbb1bd187c3898944ecd7f1f6e69","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2026-03-13 18:33:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"16ac28feaaaf5563eed437d1117817105858205a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-13 19:43:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40e6ecebe51e460cabf8001d5d797ef0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7116314801ce451ba483aaa1eb473fdf : SUCCESS in 2m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2e3b8ebef94f48c7a428e48a00c0258f : SUCCESS in 2m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/98187e035f444b56bad67deff8bb46b8 : SUCCESS in 2m 49s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a587aeb708e847408a681e81e1554925 : SUCCESS in 3m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fb5a97da8be427b8b7a8b5f091ab689 : SUCCESS in 3m 25s\n- watcher-tempest-aetos https://zuul.opendev.org/t/openstack/build/a4f8e6abdec749578f06426e5fa6c687 : SUCCESS in 1h 09m 53s\n- telemetry-dsvm-integration-aetos https://zuul.opendev.org/t/openstack/build/cad29bec7f1b44cd969c40ef170c710c : SUCCESS in 43m 52s\n- telemetry-dsvm-integration-aetos-ipv6-only https://zuul.opendev.org/t/openstack/build/a098ce673000441680c6761deaee4102 : SUCCESS in 48m 26s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d86376b35547586641bc31ca1133b5d2081e55e0","revisions":{"d86376b35547586641bc31ca1133b5d2081e55e0":{"kind":"REWORK","_number":1,"created":"2026-03-13 18:32:47.000000000","uploader":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"ref":"refs/changes/05/980505/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aetos","ref":"refs/changes/05/980505/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aetos refs/changes/05/980505/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aetos refs/changes/05/980505/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aetos refs/changes/05/980505/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aetos refs/changes/05/980505/1"}}},"commit":{"parents":[{"commit":"a803debfa4a5db981073b1674912b46b238daada","subject":"Merge \"tox: Drop redundant injection of VIRTUAL_ENV variable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aetos/commit/a803debfa4a5db981073b1674912b46b238daada"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2026-03-13 18:26:29.000000000","tz":0},"committer":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2026-03-13 18:32:42.000000000","tz":0},"subject":"Optinally hide some labels from unprivileged users","message":"Optinally hide some labels from unprivileged users\n\ndepending on deployment mode of metric collectors etc,\nsome prometheus labels might expose internal cluster data\nlike physical host names.\n\nThis patch allows operators to chose which labels to hide\nfrom unprivileged users.\nCurrently this affects query, series and label(s) API endpoints.\n\nChange-Id: I8b577d72abe02ae39194839838e4d2bca9c01521\nSigned-off-by: Pavlo Shchelokovskyy \u003cshchelokovskyy@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aetos/commit/d86376b35547586641bc31ca1133b5d2081e55e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aetos/commit/d86376b35547586641bc31ca1133b5d2081e55e0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
