)]}'
{"id":"openstack%2Fansible-collections-openstack~1005017","triplet_id":"openstack%2Fansible-collections-openstack~master~Iade373f6ab999684c932a5228d879a573f3250ad","project":"openstack/ansible-collections-openstack","branch":"master","attention_set":{"38666":{"account":{"_account_id":38666,"name":"Patrick Pfurtscheller","display_name":"Patrick Pfurtscheller","email":"patrick@pfurtscheller.org","username":"PfurtschellerP"},"last_update":"2026-09-11 07:54:09.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Iade373f6ab999684c932a5228d879a573f3250ad","subject":"feat: add option to specify the availability zone for a newly created loadbalancer","status":"NEW","created":"2026-09-10 14:07:05.000000000","updated":"2026-09-17 15:01:38.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f7e3637098e6acceef5bfaccc511699680ddf23d","_number":1005017,"virtual_id_number":1005017,"owner":{"_account_id":38666,"name":"Patrick Pfurtscheller","display_name":"Patrick Pfurtscheller","email":"patrick@pfurtscheller.org","username":"PfurtschellerP"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-17 15:01:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-10 15:31:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-17 13:38:58.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"}],"messages":[{"id":"a27ae7728ce624c3011d2461e2cc6a49dc084d90","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38666,"name":"Patrick Pfurtscheller","display_name":"Patrick Pfurtscheller","email":"patrick@pfurtscheller.org","username":"PfurtschellerP"},"date":"2026-09-10 14:07:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1af3d05b00d3a53b0e7b23b2d20e1c8843c20350","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 15:31:01.000000000","message":"Patch Set 1: Verified-1\n\n(10 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/99d2a24f315f467cbf99f337aa6fcd50\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/ba112738b7574935bfa971ff7900888f : FAILURE in 2m 43s\n- openstack-tox-linters-ansible-devel https://zuul.opendev.org/t/openstack/build/e9e8b4f437074c28b0f224247d9f1f9d : FAILURE in 3m 36s (non-voting)\n- openstack-tox-linters-ansible-2.18 https://zuul.opendev.org/t/openstack/build/40e04f71f33746c08f3e4dbe93ea2e82 : FAILURE in 2m 44s\n- openstack-tox-linters-ansible-2.19 https://zuul.opendev.org/t/openstack/build/c8ea6176ff9e4e8d91541d59d9f17ab6 : FAILURE in 4m 51s\n- openstack-tox-linters-ansible-2.20 https://zuul.opendev.org/t/openstack/build/13093b25019d4399ab1577883c025fc4 : FAILURE in 3m 26s\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/4447b7d4301744388ab22ba140ebcb4b : SUCCESS in 36m 48s\n- ansible-collections-openstack-functional-devstack-releases https://zuul.opendev.org/t/openstack/build/74badc2f06e9490da10f71f74d18cf97 : SUCCESS in 1h 12m 52s\n- ansible-collections-openstack-functional-devstack-ansible-2.18 https://zuul.opendev.org/t/openstack/build/9eb44e0412c24b6cb31f865bc5aa06bb : SUCCESS in 1h 05m 53s\n- ansible-collections-openstack-functional-devstack-ansible-2.19 https://zuul.opendev.org/t/openstack/build/10d4ae3f02a04bfc8353d9c35931a187 : SUCCESS in 1h 15m 12s\n- ansible-collections-openstack-functional-devstack-ansible-2.20 https://zuul.opendev.org/t/openstack/build/09d973879a914dd7bc6f25c31544f990 : SUCCESS in 1h 14m 13s\n- ansible-collections-openstack-functional-devstack-ansible-devel https://zuul.opendev.org/t/openstack/build/5fbf0b9fc6d347d591c95dd5dabd1491 : POST_FAILURE in 8m 55s (non-voting)\n- ansible-collections-openstack-functional-devstack-octavia https://zuul.opendev.org/t/openstack/build/e1c69999fb544cfe835068d4aeb817a7 : SUCCESS in 38m 34s\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/9a12579e144145d4b1e183a7d2abf23b : FAILURE in 9m 22s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/24596934f45841ce9950cb34c6c6e01e : FAILURE in 11m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"70388232b0d5bdfe6c39f496d530d890dfe5b968","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38666,"name":"Patrick Pfurtscheller","display_name":"Patrick Pfurtscheller","email":"patrick@pfurtscheller.org","username":"PfurtschellerP"},"date":"2026-09-11 06:25:08.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"c8d948f2f746d641063743b5491982a2990aa41e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 07:54:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a3959824b2194ce1b4ce2cbb272a5de9\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/e35239d12fb14cd88707911db86bfa25 : SUCCESS in 2m 27s\n- openstack-tox-linters-ansible-devel https://zuul.opendev.org/t/openstack/build/b8d35bdbd8a9428bb6a407a67a24fe8f : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-linters-ansible-2.18 https://zuul.opendev.org/t/openstack/build/a39bd55626b243a882200670cbc563d7 : SUCCESS in 5m 46s\n- openstack-tox-linters-ansible-2.19 https://zuul.opendev.org/t/openstack/build/14e4114ee4154f8689690d36b4585a89 : SUCCESS in 7m 05s\n- openstack-tox-linters-ansible-2.20 https://zuul.opendev.org/t/openstack/build/41b584acea2e4df586b5ee1ad21a5df9 : SUCCESS in 3m 24s\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/cd910bd87607466db8beced5281a33a3 : SUCCESS in 1h 13m 44s\n- ansible-collections-openstack-functional-devstack-releases https://zuul.opendev.org/t/openstack/build/8628813837944d9d851d32a5e07a0964 : SUCCESS in 1h 18m 36s\n- ansible-collections-openstack-functional-devstack-ansible-2.18 https://zuul.opendev.org/t/openstack/build/37dd7a51ec9140d7a725c4f819d013a6 : SUCCESS in 1h 08m 49s\n- ansible-collections-openstack-functional-devstack-ansible-2.19 https://zuul.opendev.org/t/openstack/build/dd235153df7e4114bfb157490519250b : SUCCESS in 1h 21m 43s\n- ansible-collections-openstack-functional-devstack-ansible-2.20 https://zuul.opendev.org/t/openstack/build/74b627b34f00451c886052041443a7b0 : SUCCESS in 1h 20m 52s\n- ansible-collections-openstack-functional-devstack-ansible-devel https://zuul.opendev.org/t/openstack/build/02bbff0274644fa5924d70484928d057 : POST_FAILURE in 11m 50s (non-voting)\n- ansible-collections-openstack-functional-devstack-octavia https://zuul.opendev.org/t/openstack/build/df7f5bd9055a4fc78e90f058dceb7077 : FAILURE in 36m 36s\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/aef0626735de42deada06dff6fe0a19e : FAILURE in 5m 35s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/9bfd8114640e4e36bbe88f9b68d4f034 : FAILURE in 11m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f1873b2b30c0e854c1ef9e0bf9b1dd0f2bd3255f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-17 13:38:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f7e3637098e6acceef5bfaccc511699680ddf23d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 15:01:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43214572405041e794ea0755765e04cb\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/397cefd590cc40708ee53d952fbd1beb : SUCCESS in 3m 15s\n- openstack-tox-linters-ansible-devel https://zuul.opendev.org/t/openstack/build/81abb006689e4aa79a203fbc297a1cdc : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-linters-ansible-2.18 https://zuul.opendev.org/t/openstack/build/28658bd0332c4663b03e0825446f261a : SUCCESS in 7m 24s\n- openstack-tox-linters-ansible-2.19 https://zuul.opendev.org/t/openstack/build/85184cf66bbb45a5b7f33acac437ab35 : SUCCESS in 7m 58s\n- openstack-tox-linters-ansible-2.20 https://zuul.opendev.org/t/openstack/build/9f8fe6516cff479aa9d7bd9db0536ffd : SUCCESS in 7m 06s\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/84c15003fd7647ada56b9002d127bccf : SUCCESS in 1h 19m 07s\n- ansible-collections-openstack-functional-devstack-releases https://zuul.opendev.org/t/openstack/build/9aee3532129d4bfcace7a6d77c09f508 : SUCCESS in 1h 19m 38s\n- ansible-collections-openstack-functional-devstack-ansible-2.18 https://zuul.opendev.org/t/openstack/build/2e5f2b2d58cf4811887eefad03bc84a2 : SUCCESS in 1h 15m 49s\n- ansible-collections-openstack-functional-devstack-ansible-2.19 https://zuul.opendev.org/t/openstack/build/09e39f6548144b0ab1f4af25887365dc : SUCCESS in 53m 19s\n- ansible-collections-openstack-functional-devstack-ansible-2.20 https://zuul.opendev.org/t/openstack/build/a7f18844aca04ec9a765350654248946 : SUCCESS in 55m 50s\n- ansible-collections-openstack-functional-devstack-ansible-devel https://zuul.opendev.org/t/openstack/build/593c3c4df120455f88cec53f1d6fbb84 : POST_FAILURE in 8m 22s (non-voting)\n- ansible-collections-openstack-functional-devstack-octavia https://zuul.opendev.org/t/openstack/build/4be565809fa540febce67cd297077b7b : SUCCESS in 42m 23s\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/f2d19c8a79ce4ae085eb472625c0d8a5 : FAILURE in 10m 18s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/025da40f7d1541e79c90c6b4f294ea72 : FAILURE in 11m 21s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"74caf969b319c1bf1bd7dea1e742834f3ccab597","revisions":{"26f2f584bf7d90ccd32e1b542ff9ff1f9d83513a":{"kind":"REWORK","_number":1,"created":"2026-09-10 14:07:05.000000000","uploader":{"_account_id":38666,"name":"Patrick Pfurtscheller","display_name":"Patrick Pfurtscheller","email":"patrick@pfurtscheller.org","username":"PfurtschellerP"},"ref":"refs/changes/17/1005017/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-collections-openstack","ref":"refs/changes/17/1005017/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/17/1005017/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/17/1005017/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/17/1005017/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/17/1005017/1"}}},"commit":{"parents":[{"commit":"19a8bc6e05d628668d9cce34ee5c3730e595859b","subject":"Merge \"Search for the router network in specified project\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/19a8bc6e05d628668d9cce34ee5c3730e595859b"}]}],"author":{"name":"Patrick Pfurtscheller","email":"patrick@pfurtscheller.org","date":"2026-09-10 14:06:42.000000000","tz":120},"committer":{"name":"Patrick Pfurtscheller","email":"patrick@pfurtscheller.org","date":"2026-09-10 14:06:42.000000000","tz":120},"subject":"feat: add option to specify the availability zone for a newly created loadbalancer","message":"feat: add option to specify the availability zone for a newly created loadbalancer\n\nChange-Id: Iade373f6ab999684c932a5228d879a573f3250ad\nSigned-off-by: Patrick Pfurtscheller \u003cpatrick@pfurtscheller.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/26f2f584bf7d90ccd32e1b542ff9ff1f9d83513a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/26f2f584bf7d90ccd32e1b542ff9ff1f9d83513a"}]},"branch":"refs/heads/master"},"74caf969b319c1bf1bd7dea1e742834f3ccab597":{"kind":"REWORK","_number":2,"created":"2026-09-11 06:25:08.000000000","uploader":{"_account_id":38666,"name":"Patrick Pfurtscheller","display_name":"Patrick Pfurtscheller","email":"patrick@pfurtscheller.org","username":"PfurtschellerP"},"ref":"refs/changes/17/1005017/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-collections-openstack","ref":"refs/changes/17/1005017/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/17/1005017/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/17/1005017/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/17/1005017/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/17/1005017/2"}}},"commit":{"parents":[{"commit":"19a8bc6e05d628668d9cce34ee5c3730e595859b","subject":"Merge \"Search for the router network in specified project\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/19a8bc6e05d628668d9cce34ee5c3730e595859b"}]}],"author":{"name":"Patrick Pfurtscheller","email":"patrick@pfurtscheller.org","date":"2026-09-10 14:06:42.000000000","tz":120},"committer":{"name":"Patrick Pfurtscheller","email":"patrick@pfurtscheller.org","date":"2026-09-11 06:24:46.000000000","tz":120},"subject":"feat: add option to specify the availability zone for a newly created loadbalancer","message":"feat: add option to specify the availability zone for a newly created loadbalancer\n\nChange-Id: Iade373f6ab999684c932a5228d879a573f3250ad\nSigned-off-by: Patrick Pfurtscheller \u003cpatrick@pfurtscheller.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/74caf969b319c1bf1bd7dea1e742834f3ccab597"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/74caf969b319c1bf1bd7dea1e742834f3ccab597"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
