)]}'
{"id":"openstack%2Fansible-collections-openstack~1006027","triplet_id":"openstack%2Fansible-collections-openstack~master~I8b5fa92bb30e1b4e0a9860274e071547d65b7bc8","project":"openstack/ansible-collections-openstack","branch":"master","topic":"fix_project_domain_in_volume_type_access","attention_set":{"30911":{"account":{"_account_id":30911,"name":"Jan Horstmann","email":"horstmann@osism.tech","username":"jhorstmann"},"last_update":"2026-09-17 13:38:29.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I8b5fa92bb30e1b4e0a9860274e071547d65b7bc8","subject":"Use **-operator to unpack dict in volume_type_access","status":"NEW","created":"2026-09-17 12:47:53.000000000","updated":"2026-09-17 14:20:08.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c65e7666a58610130222182592967ab946858ba5","_number":1006027,"virtual_id_number":1006027,"owner":{"_account_id":30911,"name":"Jan Horstmann","email":"horstmann@osism.tech","username":"jhorstmann"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-17 14:20:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2026-09-17 13:38:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-17 13:38:29.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2026-09-17 14:20:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d1800953298affbcac10a3519ac5c4f491bf530b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30911,"name":"Jan Horstmann","email":"horstmann@osism.tech","username":"jhorstmann"},"date":"2026-09-17 12:47:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b140ff3b350c2e365e0c5dd94dcfba3843e91806","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-17 13:38:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c65e7666a58610130222182592967ab946858ba5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 14:20:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ad08a78c7b347339c76d398dad33d94\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/33183a1719754546bc9f61d70741bb5e : SUCCESS in 3m 27s\n- openstack-tox-linters-ansible-devel https://zuul.opendev.org/t/openstack/build/dada00fb6e294a529e8e4340c06abfe1 : SUCCESS in 7m 43s (non-voting)\n- openstack-tox-linters-ansible-2.18 https://zuul.opendev.org/t/openstack/build/1cd02034055d484499cf2f5ed5f987d0 : SUCCESS in 6m 06s\n- openstack-tox-linters-ansible-2.19 https://zuul.opendev.org/t/openstack/build/397ffb2e881249d4a95472d15bf12c98 : SUCCESS in 6m 33s\n- openstack-tox-linters-ansible-2.20 https://zuul.opendev.org/t/openstack/build/8b6bc8f309894e0a91955f1954e6022f : SUCCESS in 8m 45s\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/6704dbc08477472da59bd72ea7027e1b : SUCCESS in 1h 30m 54s\n- ansible-collections-openstack-functional-devstack-releases https://zuul.opendev.org/t/openstack/build/1153b90d039a4ffba057be3bedabec66 : SUCCESS in 1h 12m 48s\n- ansible-collections-openstack-functional-devstack-ansible-2.18 https://zuul.opendev.org/t/openstack/build/da633e46a0ab4b28b009382ba4213214 : SUCCESS in 57m 01s\n- ansible-collections-openstack-functional-devstack-ansible-2.19 https://zuul.opendev.org/t/openstack/build/d386f3b4ce284e5d9ad6d13751435f66 : SUCCESS in 1h 13m 44s\n- ansible-collections-openstack-functional-devstack-ansible-2.20 https://zuul.opendev.org/t/openstack/build/8865f281c06b4e7b91c5b3b7f2e1953f : SUCCESS in 1h 04m 13s\n- ansible-collections-openstack-functional-devstack-ansible-devel https://zuul.opendev.org/t/openstack/build/32aedc56decb48beb1895ee7bd454b85 : POST_FAILURE in 8m 00s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/2b85304b91a240c5a1ebbaa5ab0d5efd : FAILURE in 9m 11s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/418bf857899540c69d93a761ca2336ae : FAILURE in 11m 12s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e433ee401bdfde61865babfd49d931b217dc8871","revisions":{"e433ee401bdfde61865babfd49d931b217dc8871":{"kind":"REWORK","_number":1,"created":"2026-09-17 12:47:53.000000000","uploader":{"_account_id":30911,"name":"Jan Horstmann","email":"horstmann@osism.tech","username":"jhorstmann"},"ref":"refs/changes/27/1006027/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-collections-openstack","ref":"refs/changes/27/1006027/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/27/1006027/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/27/1006027/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/27/1006027/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/27/1006027/1"}}},"commit":{"parents":[{"commit":"6fccec65bc32b6c39776c612705bd482691d9b3f","subject":"Merge \"fix using undefined variable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/6fccec65bc32b6c39776c612705bd482691d9b3f"}]}],"author":{"name":"Jan Horstmann","email":"horstmann@osism.tech","date":"2026-09-17 12:15:37.000000000","tz":120},"committer":{"name":"Jan Horstmann","email":"horstmann@osism.tech","date":"2026-09-17 12:47:51.000000000","tz":120},"subject":"Use **-operator to unpack dict in volume_type_access","message":"Use **-operator to unpack dict in volume_type_access\n\nCloses-Bug: #2167592\n\nChange-Id: I8b5fa92bb30e1b4e0a9860274e071547d65b7bc8\nSigned-off-by: Jan Horstmann \u003chorstmann@osism.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/e433ee401bdfde61865babfd49d931b217dc8871"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/e433ee401bdfde61865babfd49d931b217dc8871"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
