)]}'
{"id":"openstack%2Fansible-collections-openstack~1006688","triplet_id":"openstack%2Fansible-collections-openstack~master~I16cff99e86f0676a5c16121e8eca2b7a13ea9d6e","project":"openstack/ansible-collections-openstack","branch":"master","attention_set":{"34380":{"account":{"_account_id":34380,"name":"Maximilian Stinsky","email":"maximilian@stinsky.com","username":"mstinsky"},"last_update":"2026-09-23 07:37:26.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I16cff99e86f0676a5c16121e8eca2b7a13ea9d6e","subject":"identity_user: support Keystone user resource options","status":"NEW","created":"2026-09-22 08:25:47.000000000","updated":"2026-09-23 07:37:26.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9ee86235e052936f2746844484b1e859338e0c55","_number":1006688,"virtual_id_number":1006688,"owner":{"_account_id":34380,"name":"Maximilian Stinsky","email":"maximilian@stinsky.com","username":"mstinsky"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-09-22 09:18:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-22 09:18:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8167a4bc64685d64fdcca6915c92915d3cb21ce6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34380,"name":"Maximilian Stinsky","email":"maximilian@stinsky.com","username":"mstinsky"},"date":"2026-09-22 08:25:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e41c6de3dd0db96dbcd3a87e96a95be4df73a7c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 09:18:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3c388a00c025464eb8a355edb5d01fda\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/b186cc236d8d448cb23e9a805e188794 : SUCCESS in 3m 08s\n- openstack-tox-linters-ansible-devel https://zuul.opendev.org/t/openstack/build/7f230942d53f418faff5ccfd1906448d : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-linters-ansible-2.18 https://zuul.opendev.org/t/openstack/build/bc503ff7a70a44e9af1f3b7b0ed74390 : SUCCESS in 6m 28s\n- openstack-tox-linters-ansible-2.19 https://zuul.opendev.org/t/openstack/build/65b70788b50f4b2399da288f45704fcc : SUCCESS in 6m 28s\n- openstack-tox-linters-ansible-2.20 https://zuul.opendev.org/t/openstack/build/48447e8375ad4983976ae5046e6490f2 : SUCCESS in 6m 26s\n- openstack-tox-linters-ansible-2.21 https://zuul.opendev.org/t/openstack/build/51bfddb3bd1d4512bd0c854b9e94b7b9 : SUCCESS in 4m 07s\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/5431eef2fb664cd3b8d46dfa5b025376 : FAILURE in 17m 45s\n- ansible-collections-openstack-functional-devstack-releases https://zuul.opendev.org/t/openstack/build/b23d1fee45a94b90a469937090037781 : FAILURE in 36m 37s\n- ansible-collections-openstack-functional-devstack-ansible-2.18 https://zuul.opendev.org/t/openstack/build/1895f53b93ce4adc90fcbe16bb8f5c6d : FAILURE in 38m 17s\n- ansible-collections-openstack-functional-devstack-ansible-2.19 https://zuul.opendev.org/t/openstack/build/7e4ed0a3d6a24066a924add5bf96181c : FAILURE in 22m 07s\n- ansible-collections-openstack-functional-devstack-ansible-2.20 https://zuul.opendev.org/t/openstack/build/200823663ce8440c9e8261b7a28cebe3 : FAILURE in 18m 35s\n- ansible-collections-openstack-functional-devstack-ansible-2.21 https://zuul.opendev.org/t/openstack/build/653559ce6508449fa3cd2fcdca847539 : FAILURE in 37m 40s\n- ansible-collections-openstack-functional-devstack-ansible-devel https://zuul.opendev.org/t/openstack/build/4ec08ab40678436b9725ecfab9bc7344 : POST_FAILURE in 8m 09s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/d8aad523b13041adaeb6e2f6dac2144f : FAILURE in 5m 51s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/c3757a658ff64bd8b1637f419c53cc46 : FAILURE in 13m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"033e4fe80eaaff65f8f29f72aa51d288eeebb18f","author":{"_account_id":34380,"name":"Maximilian Stinsky","email":"maximilian@stinsky.com","username":"mstinsky"},"date":"2026-09-23 06:44:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9ee86235e052936f2746844484b1e859338e0c55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 07:37:26.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8a984ec716c54d5c9e67c0d254f7ec5b\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/b028bee49729450ba8c1dddb97a2f03d : SUCCESS in 2m 27s\n- openstack-tox-linters-ansible-devel https://zuul.opendev.org/t/openstack/build/c1fd8edc27d34aee841fb9155bb9c590 : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-linters-ansible-2.18 https://zuul.opendev.org/t/openstack/build/05989d58a3f24c68b087127081b5edce : SUCCESS in 5m 56s\n- openstack-tox-linters-ansible-2.19 https://zuul.opendev.org/t/openstack/build/eb22288c7e054dbb9116f9d2403d154c : SUCCESS in 4m 12s\n- openstack-tox-linters-ansible-2.20 https://zuul.opendev.org/t/openstack/build/f60edc19d3e349cdb00908345ffe95b9 : SUCCESS in 6m 21s\n- openstack-tox-linters-ansible-2.21 https://zuul.opendev.org/t/openstack/build/d6e3d4034ed143589899318b1e7cded6 : SUCCESS in 6m 25s\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/b4583b598b7644dcac0b07d4e16090f7 : FAILURE in 37m 02s\n- ansible-collections-openstack-functional-devstack-releases https://zuul.opendev.org/t/openstack/build/f828d157480e478b98eb41cd1af6cf4c : FAILURE in 35m 09s\n- ansible-collections-openstack-functional-devstack-ansible-2.18 https://zuul.opendev.org/t/openstack/build/0cbdf1fc594e426a869b9f649b27a980 : FAILURE in 35m 33s\n- ansible-collections-openstack-functional-devstack-ansible-2.19 https://zuul.opendev.org/t/openstack/build/08dd7f2ef6f74f3a9903dc58f571e025 : FAILURE in 39m 29s\n- ansible-collections-openstack-functional-devstack-ansible-2.20 https://zuul.opendev.org/t/openstack/build/7a6a262220094cd0a94e7978569579eb : FAILURE in 23m 17s\n- ansible-collections-openstack-functional-devstack-ansible-2.21 https://zuul.opendev.org/t/openstack/build/343cb6a6680c45db9fd284c603ca3d5e : FAILURE in 36m 31s\n- ansible-collections-openstack-functional-devstack-ansible-devel https://zuul.opendev.org/t/openstack/build/ad12a2d315774d4283551e9308861906 : RETRY_LIMIT in 2m 07s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/429528fd320f4670be6d3c1485569a15 : RETRY_LIMIT in 1m 40s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/e18b2e971eeb404d9e02620e52c6cb11 : FAILURE in 12m 00s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cfb7a39221d5ae633d5edba506b9b65d81b3e863","revisions":{"cfb7a39221d5ae633d5edba506b9b65d81b3e863":{"kind":"REWORK","_number":1,"created":"2026-09-22 08:25:47.000000000","uploader":{"_account_id":34380,"name":"Maximilian Stinsky","email":"maximilian@stinsky.com","username":"mstinsky"},"ref":"refs/changes/88/1006688/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-collections-openstack","ref":"refs/changes/88/1006688/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/88/1006688/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/88/1006688/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/88/1006688/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/88/1006688/1"}}},"commit":{"parents":[{"commit":"da4a211b2d1dd147dead0e3db63a373b43b3c09e","subject":"CI: Test with ansible-core 2.21","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/da4a211b2d1dd147dead0e3db63a373b43b3c09e"}]}],"author":{"name":"Maximilian Stinsky","email":"maximilian.stinsky-damke@wiit.cloud","date":"2026-09-22 08:21:29.000000000","tz":120},"committer":{"name":"Maximilian Stinsky","email":"maximilian.stinsky-damke@wiit.cloud","date":"2026-09-22 08:24:31.000000000","tz":120},"subject":"identity_user: support Keystone user resource options","message":"identity_user: support Keystone user resource options\n\nAdd an `options` parameter to the identity_user module, mapping 1:1 to\nKeystone\u0027s user resource `options` field exposed by openstacksdk\n(`User.options`). This allows setting options such as\n`ignore_password_expiry` and `ignore_change_password_upon_first_use`.\n\nOn update, desired options are merged with the user\u0027s existing options\nso that options not managed by the module are preserved, and the\noperation is idempotent when the desired values already match.\n\nChange-Id: I16cff99e86f0676a5c16121e8eca2b7a13ea9d6e\nSigned-off-by: Maximilian Stinsky \u003cmaximilian.stinsky-damke@wiit.cloud\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/cfb7a39221d5ae633d5edba506b9b65d81b3e863"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/cfb7a39221d5ae633d5edba506b9b65d81b3e863"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
