)]}'
{"id":"openstack%2Fansible-collections-openstack~710641","triplet_id":"openstack%2Fansible-collections-openstack~master~I34d544ce48f25bffde8e6e0cf82cdf9a85e681c3","project":"openstack/ansible-collections-openstack","branch":"master","hashtags":[],"change_id":"I34d544ce48f25bffde8e6e0cf82cdf9a85e681c3","subject":"Test with and assert support for python3","status":"MERGED","created":"2020-03-01 12:31:37.000000000","updated":"2020-03-04 22:27:00.000000000","submitted":"2020-03-04 22:27:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"710641-1583360820422-a42b5d9c","meta_rev_id":"5149bb5d619cf771f222a4b566c39f9de1c2b794","_number":710641,"virtual_id_number":710641,"owner":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-04 22:27:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-03 16:40:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":1,"date":"2020-03-04 21:39:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-03 16:40:25.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2020-03-04 22:27:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ae878540fb8fe86bc8f3534b4261f523371d2dea","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-03-01 12:31:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1e8ad8d6ef814b29b13d4d5555f3235ef10e6356","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-01 13:01:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/fc07f7f6c3b84c63b9faaffa94cff678 : SUCCESS in 4m 13s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c6fb7f2fd9b143368f2958f49380146c : SUCCESS in 4m 46s\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/af9f073031df496a97fddec7cd0a06f8 : SUCCESS in 29m 32s","accounts_in_message":[],"_revision_number":1},{"id":"9e094e759fec96c02b9ae67d1c94f824758b9e1f","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-03-02 12:43:17.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"274237ed78a09655b75b651106f6d72afd382900","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-03-03 14:20:48.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"a07d6ab8e6e026daab4615d0fa3b92d330855c8c","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-03-03 14:23:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7385f5ceb1390d788f74c99bafe259318335be21","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-03-03 15:12:06.000000000","message":"Patch Set 1: Workflow-1\n\nwill update this to say 3.6","accounts_in_message":[],"_revision_number":1},{"id":"7f4092f74b627969b03c1256d25ec49981d2d480","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-03-03 15:29:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"12725bfd66ceb95ba185e5f7806df499330cec0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 16:33:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/ae666b3b58754c3d89148e646920fb0f : SUCCESS in 6m 41s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/961363156a6045bea5e42401b0634c5c : SUCCESS in 6m 34s\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/2444d39d89cd4600ae358dae06cc7d75 : SUCCESS in 49m 21s","accounts_in_message":[],"_revision_number":2},{"id":"c8e1ca1df1d0719133353fafb57af1c073a10d0b","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-03-03 16:40:25.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"55cdbd919a4f1949edd0f19361455c9769b8f5d7","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-03-04 21:39:56.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7884654e18dc2d4fd36e911ed718e55de14dec77","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 21:40:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"315e93f1f706817d41707db83a48d8ae524411e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 22:27:00.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/ee880245dd004dc58533e46ac833064a : SUCCESS in 3m 55s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/16a2dab6e8d74a0eb0956c8bcc2c8b48 : SUCCESS in 3m 50s\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/b031b3b1c8984b7aa29ed7ac65b4f1a2 : SUCCESS in 41m 34s","accounts_in_message":[],"_revision_number":2},{"id":"5149bb5d619cf771f222a4b566c39f9de1c2b794","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 22:27:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"53affc3220362435906909a0caa2233e1878802f","revisions":{"24bdbc784bafde099202bf1dbc26ab3cbd36c934":{"kind":"REWORK","_number":1,"created":"2020-03-01 12:31:37.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/41/710641/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-collections-openstack","ref":"refs/changes/41/710641/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/41/710641/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/41/710641/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/41/710641/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/41/710641/1"}}},"commit":{"parents":[{"commit":"96a5914eb6806da70a27391ff449813f88c4a458","subject":"Remove from __future__ lines","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/96a5914eb6806da70a27391ff449813f88c4a458"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-03-01 12:29:59.000000000","tz":-360},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-03-01 12:29:59.000000000","tz":-360},"subject":"Test with and assert support for python3","message":"Test with and assert support for python3\n\nCollections are new. openstacksdk as of now doesn\u0027t support python2.\nWe shouldn\u0027t pretend to support python2 either. If you\u0027re using\nansible new enough to use collections, you can use python3.\n\nChange-Id: I34d544ce48f25bffde8e6e0cf82cdf9a85e681c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/24bdbc784bafde099202bf1dbc26ab3cbd36c934"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/24bdbc784bafde099202bf1dbc26ab3cbd36c934"}]},"branch":"refs/heads/master"},"53affc3220362435906909a0caa2233e1878802f":{"kind":"REWORK","_number":2,"created":"2020-03-03 15:29:10.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/41/710641/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-collections-openstack","ref":"refs/changes/41/710641/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/41/710641/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/41/710641/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/41/710641/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/41/710641/2"}}},"commit":{"parents":[{"commit":"96a5914eb6806da70a27391ff449813f88c4a458","subject":"Remove from __future__ lines","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/96a5914eb6806da70a27391ff449813f88c4a458"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-03-01 12:29:59.000000000","tz":-360},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-03-03 15:28:43.000000000","tz":-360},"subject":"Test with and assert support for python3","message":"Test with and assert support for python3\n\nCollections are new. openstacksdk as of now doesn\u0027t support python2.\nWe shouldn\u0027t pretend to support python2 either. If you\u0027re using\nansible new enough to use collections, you can use python3.\n\nThis release of openstacksdk only supports 3.6 and onwards, so\nset that as our min.\n\nChange-Id: I34d544ce48f25bffde8e6e0cf82cdf9a85e681c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/53affc3220362435906909a0caa2233e1878802f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/53affc3220362435906909a0caa2233e1878802f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
