)]}'
{"id":"openstack%2Fansible-collections-openstack~727226","triplet_id":"openstack%2Fansible-collections-openstack~master~I509ae5d398ff8a93d3cbcd09edb21118a1ff03c3","project":"openstack/ansible-collections-openstack","branch":"master","hashtags":[],"change_id":"I509ae5d398ff8a93d3cbcd09edb21118a1ff03c3","subject":"Remove incorrect os_ prefix in documentation","status":"MERGED","created":"2020-05-12 15:20:18.000000000","updated":"2020-05-12 18:16:45.000000000","submitted":"2020-05-12 18:16:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"727226-1589307405066-838f9c63","meta_rev_id":"bcd909b54ea43c50a518925de38267d57dd95b89","_number":727226,"virtual_id_number":727226,"owner":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-12 18:16:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"all":[{"value":2,"date":"2020-05-12 17:19:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"all":[{"value":1,"date":"2020-05-12 17:19:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-12 17:19:13.000000000","updated_by":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"reviewer":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"state":"REVIEWER"},{"updated":"2020-05-12 18:16:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"af608e3928e797a23f5ae7758fcb83eae42580a2","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-05-12 15:20:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d7eff6c750f732e2713e935e15b64b4213fcb912","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 16:35:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/91629218d0d54be79579aca4728aef97 : SUCCESS in 4m 31s\n- openstack-tox-linters-ansible-devel https://zuul.opendev.org/t/openstack/build/1ccc5feec576407f864edf0e208fdc3d : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-linters-ansible-2.9 https://zuul.opendev.org/t/openstack/build/08bf8922a70e41f681ed4ef217c4d239 : SUCCESS in 6m 57s\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/c238dcbd30ce490fbd42a49a86fa3cee : SUCCESS in 38m 11s\n- ansible-collections-openstack-functional-devstack-releases https://zuul.opendev.org/t/openstack/build/997d5b7f320844e3bfe90519c196cbdc : SUCCESS in 37m 55s\n- ansible-collections-openstack-functional-devstack-ansible-devel https://zuul.opendev.org/t/openstack/build/300f76bbc39548c1b4a493d8a170a494 : SUCCESS in 38m 09s (non-voting)\n- ansible-collections-openstack-functional-devstack-ansible-2.9 https://zuul.opendev.org/t/openstack/build/2f553189800c45e59f4cdad38332f113 : SUCCESS in 40m 16s\n- ansible-collections-openstack-functional-devstack-train-ansible-devel https://zuul.opendev.org/t/openstack/build/c71ae0d646e146bf8f862bea0fb65e7a : SUCCESS in 39m 14s (non-voting)\n- ansible-collections-openstack-functional-devstack-train-ansible-2.9 https://zuul.opendev.org/t/openstack/build/2232666c34f14c1585cc620ce142eac2 : SUCCESS in 44m 28s\n- ansible-collections-openstack-functional-devstack-stein-ansible-devel https://zuul.opendev.org/t/openstack/build/5e683071863c4b9c889c030bf8cce03a : SUCCESS in 46m 19s (non-voting)\n- ansible-collections-openstack-functional-devstack-stein-ansible-2.9 https://zuul.opendev.org/t/openstack/build/343bc0594fd04116b7bd52bd81e2347e : SUCCESS in 47m 43s\n- ansible-collections-openstack-functional-devstack-rocky-ansible-devel https://zuul.opendev.org/t/openstack/build/fcc2c31eee384de5a69903abf68542f8 : SUCCESS in 46m 59s (non-voting)\n- ansible-collections-openstack-functional-devstack-rocky-ansible-2.9 https://zuul.opendev.org/t/openstack/build/0e02dd3415d84d4d9e754822030c38a5 : SUCCESS in 47m 07s\n- ansible-collections-openstack-functional-devstack-queens-ansible-2.9 https://zuul.opendev.org/t/openstack/build/cfaf9a5381204bb290293d65a56e4eab : SUCCESS in 51m 38s","accounts_in_message":[],"_revision_number":1},{"id":"eef7fc15460aa6bfdf9f91008db2b42a1ed7875a","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2020-05-12 17:19:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c016c6f0af93f15faca13af427b3ecd11ac600ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 17:19:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"005ae6479275afd39768c6dccdcce6250a68f0c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 18:16:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/2a0533e586404895a4ff9bde8590b77f : SUCCESS in 7m 48s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0f74352decdf4137bfd5309e4cf4fc04 : SUCCESS in 5m 33s\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/85a88941652346c484310a1b9452f675 : SUCCESS in 49m 45s\n- ansible-collections-openstack-functional-devstack-releases https://zuul.opendev.org/t/openstack/build/be947f5ab1ad434690b16f094c12aa40 : SUCCESS in 46m 49s\n- ansible-collections-openstack-functional-devstack-ansible-2.9 https://zuul.opendev.org/t/openstack/build/0793512e0d0d41aabd0b975d859f983c : SUCCESS in 41m 34s\n- ansible-collections-openstack-functional-devstack-train-ansible-2.9 https://zuul.opendev.org/t/openstack/build/e315787c96a6431b9409389d365ddc7d : SUCCESS in 40m 59s\n- ansible-collections-openstack-functional-devstack-stein-ansible-2.9 https://zuul.opendev.org/t/openstack/build/1abb8abf56334b98a528d5e54af0b5b7 : SUCCESS in 39m 53s\n- ansible-collections-openstack-functional-devstack-rocky-ansible-2.9 https://zuul.opendev.org/t/openstack/build/3992bc7770ca43a79101479cd761fe02 : SUCCESS in 50m 52s","accounts_in_message":[],"_revision_number":1},{"id":"bcd909b54ea43c50a518925de38267d57dd95b89","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 18:16:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"238387f2f8b5287cd9e1fb9ff1ebf4566314547a","revisions":{"238387f2f8b5287cd9e1fb9ff1ebf4566314547a":{"kind":"REWORK","_number":1,"created":"2020-05-12 15:20:18.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/26/727226/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-collections-openstack","ref":"refs/changes/26/727226/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/26/727226/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/26/727226/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/26/727226/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/26/727226/1"}}},"commit":{"parents":[{"commit":"6271268567fd6744141e2c7d3857b2982f694015","subject":"Run linters in a temp dir instead of an install","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/6271268567fd6744141e2c7d3857b2982f694015"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-05-12 15:10:57.000000000","tz":-300},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-05-12 15:10:57.000000000","tz":-300},"subject":"Remove incorrect os_ prefix in documentation","message":"Remove incorrect os_ prefix in documentation\n\nThe parameteres are project_domain_name and user_domain_name.\n\nChange-Id: I509ae5d398ff8a93d3cbcd09edb21118a1ff03c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/238387f2f8b5287cd9e1fb9ff1ebf4566314547a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/238387f2f8b5287cd9e1fb9ff1ebf4566314547a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
