)]}'
{"id":"openstack%2Fansible-collections-openstack~737226","triplet_id":"openstack%2Fansible-collections-openstack~master~If15fedb121e76c661774d1c7717c3e300f49618e","project":"openstack/ansible-collections-openstack","branch":"master","topic":"test1","hashtags":[],"change_id":"If15fedb121e76c661774d1c7717c3e300f49618e","subject":"DNM test new CI","status":"ABANDONED","created":"2020-06-22 07:46:10.000000000","updated":"2020-06-23 13:15:41.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"307c42250599823b927b59ef78430a74bde2beef","_number":737226,"virtual_id_number":737226,"owner":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-06-22 22:35:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-22 22:35:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0ff1488ce8e03e01da6677862cdd03cfee28c9e0","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2020-06-22 07:46:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"beae52c63fec2d9b00fe98c10c55e316c9c597cc","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2020-06-22 08:44:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"04a6042dd32609c8bef29fa300e025cb9ddfe981","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 09:52:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/0f2d45b9de9b40d0b2a250da5f4ff5b4 : SUCCESS in 4m 25s\n- openstack-tox-linters-ansible-devel https://zuul.opendev.org/t/openstack/build/828325b3be4a4b03af4008d56bbadd87 : SUCCESS in 7m 09s (non-voting)\n- openstack-tox-linters-ansible-2.10 https://zuul.opendev.org/t/openstack/build/d79c45f7c2a34a87af1b66f34f6c1b75 : SUCCESS in 5m 48s\n- openstack-tox-linters-ansible-2.9 https://zuul.opendev.org/t/openstack/build/40e643400f104d599bd962330b2550b6 : SUCCESS in 7m 00s\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/5b7e5053840e446a9fa9ff651a0454d0 : SUCCESS in 29m 18s\n- ansible-collections-openstack-functional-devstack-releases https://zuul.opendev.org/t/openstack/build/5e2f69aad7ea43479cbd44e10c681f45 : SUCCESS in 38m 17s\n- ansible-collections-openstack-functional-devstack-ansible-2.9 https://zuul.opendev.org/t/openstack/build/b93f34b791e6455cbc61e2a07f80988f : SUCCESS in 31m 44s\n- ansible-collections-openstack-functional-devstack-ansible-2.10 https://zuul.opendev.org/t/openstack/build/c5c10ac04833492a9937dde19a93008a : SUCCESS in 52m 39s\n- ansible-collections-openstack-functional-devstack-ansible-devel https://zuul.opendev.org/t/openstack/build/6a79caaa35894317bbec728743aadb97 : SUCCESS in 50m 52s (non-voting)\n- ansible-collections-openstack-functional-devstack-ansible-pip https://zuul.opendev.org/t/openstack/build/e15a9f03a71143459726724b1b79e6a6 : SUCCESS in 39m 19s\n- ansible-collections-openstack-functional-devstack-ussuri-ansible-2.9 https://zuul.opendev.org/t/openstack/build/4bd27893eecf4a559dfb4a0b8cdf8814 : SUCCESS in 49m 34s\n- ansible-collections-openstack-functional-devstack-train-ansible-2.9 https://zuul.opendev.org/t/openstack/build/25442f7c6ff24b14b2d10eb97a1e2ad6 : SUCCESS in 53m 40s\n- ansible-collections-openstack-functional-devstack-stein-ansible-2.9 https://zuul.opendev.org/t/openstack/build/5557b31f3b624f24929fa071d6fed716 : SUCCESS in 51m 54s\n- ansible-collections-openstack-functional-devstack-rocky-ansible-2.9 https://zuul.opendev.org/t/openstack/build/7fa796bc89d64a429aac00c1f706a2c0 : SUCCESS in 49m 21s\n- ansible-collections-openstack-functional-devstack-queens-ansible-2.9 https://zuul.opendev.org/t/openstack/build/f0f332ddbf6a4ae2827acdda3abd2fba : SUCCESS in 37m 26s","accounts_in_message":[],"_revision_number":2},{"id":"449dc5d85c7b2421aba7821603aa522bc85f9e17","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2020-06-22 16:43:47.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"afb9126d1fa7c57418a1e029e089914325d3b535","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 17:55:09.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/90689ceed8e1485a8e6c16150cbb181d : SUCCESS in 4m 36s\n- openstack-tox-linters-ansible-devel https://zuul.opendev.org/t/openstack/build/adb482a6755e4c528f9defd05a9a0e99 : SUCCESS in 7m 07s (non-voting)\n- openstack-tox-linters-ansible-2.10 https://zuul.opendev.org/t/openstack/build/358a177cd3d74135ab72e0841c073263 : SUCCESS in 6m 23s\n- openstack-tox-linters-ansible-2.9 https://zuul.opendev.org/t/openstack/build/e7ad3406fd184d0ea8461208e12c1a09 : SUCCESS in 7m 18s\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/4230bd056a7241bb8343b352d45472f1 : SUCCESS in 38m 40s\n- ansible-collections-openstack-functional-devstack-releases https://zuul.opendev.org/t/openstack/build/848e499687454e02afca708c3598c7b3 : SUCCESS in 39m 41s\n- ansible-collections-openstack-functional-devstack-ansible-2.9 https://zuul.opendev.org/t/openstack/build/1d8b0213288549dcbea78f9e30d4b0e6 : SUCCESS in 47m 13s\n- ansible-collections-openstack-functional-devstack-ansible-2.10 https://zuul.opendev.org/t/openstack/build/9ad2fe69d3d4406cb190948ee1b96ac8 : SUCCESS in 46m 31s\n- ansible-collections-openstack-functional-devstack-ansible-devel https://zuul.opendev.org/t/openstack/build/fc5f67018e3f479da61dcbf9a59f82a6 : SUCCESS in 41m 37s (non-voting)\n- ansible-collections-openstack-functional-devstack-ansible-pip https://zuul.opendev.org/t/openstack/build/ccdca69d7bdf4a31aa76569b6414fbb6 : SUCCESS in 40m 22s\n- ansible-collections-openstack-functional-devstack-ussuri-ansible-2.9 https://zuul.opendev.org/t/openstack/build/2e67777927844f78b58d89f367fa4741 : SUCCESS in 44m 19s\n- ansible-collections-openstack-functional-devstack-train-ansible-2.9 https://zuul.opendev.org/t/openstack/build/621028f4cdc84f9e9e395a2875b38fc6 : SUCCESS in 42m 34s\n- ansible-collections-openstack-functional-devstack-stein-ansible-2.9 https://zuul.opendev.org/t/openstack/build/c93e5f41f42a4c89ae38933db0444a90 : SUCCESS in 50m 48s\n- ansible-collections-openstack-functional-devstack-rocky-ansible-2.9 https://zuul.opendev.org/t/openstack/build/c3d524b586c2474d9fd1536c3381e9d8 : SUCCESS in 45m 43s\n- ansible-collections-openstack-functional-devstack-queens-ansible-2.9 https://zuul.opendev.org/t/openstack/build/b9351b6e745546e7b5233bf11ff3e125 : SUCCESS in 32m 33s","accounts_in_message":[],"_revision_number":3},{"id":"85460fa0483984f46f4b4a891ec11ccc9c951d72","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2020-06-22 21:30:23.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"24850c3325a79509a809f18a708ca0b1505f5ec0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 22:35:37.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/f2d8cd377e514cc2b6770444a213b4c2 : SUCCESS in 5m 33s\n- openstack-tox-linters-ansible-devel https://zuul.opendev.org/t/openstack/build/c9ca3a31accf48b39ace381f9239a93a : SUCCESS in 6m 52s (non-voting)\n- openstack-tox-linters-ansible-2.10 https://zuul.opendev.org/t/openstack/build/7b1ffb57c7c64af2bce9ec5545274f80 : SUCCESS in 7m 33s\n- openstack-tox-linters-ansible-2.9 https://zuul.opendev.org/t/openstack/build/1f93e6c851524ebda74df4d977bb3839 : SUCCESS in 7m 05s\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/13e9c44538d944cea7114e7bfa89a30b : SUCCESS in 41m 10s\n- ansible-collections-openstack-functional-devstack-releases https://zuul.opendev.org/t/openstack/build/73dde48b20404a859bf892f85a4de848 : SUCCESS in 48m 26s\n- ansible-collections-openstack-functional-devstack-ansible-2.9 https://zuul.opendev.org/t/openstack/build/df4fb67b2eec44a2abf302dee11e38ab : SUCCESS in 45m 03s\n- ansible-collections-openstack-functional-devstack-ansible-2.10 https://zuul.opendev.org/t/openstack/build/3747a5e874dc438a8e3545afcafbc840 : SUCCESS in 50m 20s\n- ansible-collections-openstack-functional-devstack-ansible-devel https://zuul.opendev.org/t/openstack/build/dbf61c86fefe4fa5b38f5b6bea978215 : SUCCESS in 49m 07s (non-voting)\n- ansible-collections-openstack-functional-devstack-ansible-pip https://zuul.opendev.org/t/openstack/build/29cefe472d834f8f879cbaf503421e4b : SUCCESS in 46m 19s\n- ansible-collections-openstack-functional-devstack-ussuri-ansible-2.9 https://zuul.opendev.org/t/openstack/build/647fc512664146f2b34bceb6dacee56b : SUCCESS in 49m 34s\n- ansible-collections-openstack-functional-devstack-train-ansible-2.9 https://zuul.opendev.org/t/openstack/build/de812938a6224e2084c6b6e9c9ca91ed : SUCCESS in 48m 46s\n- ansible-collections-openstack-functional-devstack-stein-ansible-2.9 https://zuul.opendev.org/t/openstack/build/b2dece3de8f24b7383a10a562cdd9a68 : SUCCESS in 49m 06s\n- ansible-collections-openstack-functional-devstack-rocky-ansible-2.9 https://zuul.opendev.org/t/openstack/build/0ade0228cf3047ee994c42ffd78dd17a : SUCCESS in 46m 45s\n- ansible-collections-openstack-functional-devstack-queens-ansible-2.9 https://zuul.opendev.org/t/openstack/build/5795e4ef733249589886fe9efab7f8cf : SUCCESS in 55m 57s","accounts_in_message":[],"_revision_number":4},{"id":"2841d5eb1ea146a38d7945c87b9f83fb7fe94a74","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2020-06-23 13:15:41.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"c16810621f44c468a24f2dd7d06805d59395d5f9","revisions":{"ce59124496f20b8a025683f25c0b9cc66a685c73":{"kind":"REWORK","_number":1,"created":"2020-06-22 07:46:10.000000000","uploader":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"ref":"refs/changes/26/737226/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-collections-openstack","ref":"refs/changes/26/737226/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/26/737226/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/26/737226/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/26/737226/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/26/737226/1"}}},"commit":{"parents":[{"commit":"5e3a91a7c63355949a8b7c4cc664fab66a853f7c","subject":"Add OpenstackModule to os_server_action","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/5e3a91a7c63355949a8b7c4cc664fab66a853f7c"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2020-06-19 07:41:51.000000000","tz":180},"committer":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2020-06-22 07:44:46.000000000","tz":180},"subject":"DNM test new CI","message":"DNM test new CI\n\nChange-Id: If15fedb121e76c661774d1c7717c3e300f49618e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/ce59124496f20b8a025683f25c0b9cc66a685c73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/ce59124496f20b8a025683f25c0b9cc66a685c73"}]},"branch":"refs/heads/master"},"2b4720a8537f6ef91b9a84a36dba98fcdf9bc51e":{"kind":"REWORK","_number":2,"created":"2020-06-22 08:44:54.000000000","uploader":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"ref":"refs/changes/26/737226/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-collections-openstack","ref":"refs/changes/26/737226/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/26/737226/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/26/737226/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/26/737226/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/26/737226/2"}}},"commit":{"parents":[{"commit":"5e3a91a7c63355949a8b7c4cc664fab66a853f7c","subject":"Add OpenstackModule to os_server_action","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/5e3a91a7c63355949a8b7c4cc664fab66a853f7c"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2020-06-19 07:41:51.000000000","tz":180},"committer":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2020-06-22 08:44:45.000000000","tz":180},"subject":"DNM test new CI","message":"DNM test new CI\n\nChange-Id: If15fedb121e76c661774d1c7717c3e300f49618e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/2b4720a8537f6ef91b9a84a36dba98fcdf9bc51e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/2b4720a8537f6ef91b9a84a36dba98fcdf9bc51e"}]},"branch":"refs/heads/master"},"a3ed2222fa41de05f332938d6838eae649abdba8":{"kind":"REWORK","_number":3,"created":"2020-06-22 16:43:47.000000000","uploader":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"ref":"refs/changes/26/737226/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-collections-openstack","ref":"refs/changes/26/737226/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/26/737226/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/26/737226/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/26/737226/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/26/737226/3"}}},"commit":{"parents":[{"commit":"5e3a91a7c63355949a8b7c4cc664fab66a853f7c","subject":"Add OpenstackModule to os_server_action","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/5e3a91a7c63355949a8b7c4cc664fab66a853f7c"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2020-06-19 07:41:51.000000000","tz":180},"committer":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2020-06-22 16:43:37.000000000","tz":180},"subject":"DNM test new CI","message":"DNM test new CI\n\nChange-Id: If15fedb121e76c661774d1c7717c3e300f49618e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/a3ed2222fa41de05f332938d6838eae649abdba8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/a3ed2222fa41de05f332938d6838eae649abdba8"}]},"branch":"refs/heads/master"},"c16810621f44c468a24f2dd7d06805d59395d5f9":{"kind":"REWORK","_number":4,"created":"2020-06-22 21:30:23.000000000","uploader":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"ref":"refs/changes/26/737226/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-collections-openstack","ref":"refs/changes/26/737226/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/26/737226/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/26/737226/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/26/737226/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/26/737226/4"}}},"commit":{"parents":[{"commit":"5e3a91a7c63355949a8b7c4cc664fab66a853f7c","subject":"Add OpenstackModule to os_server_action","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/5e3a91a7c63355949a8b7c4cc664fab66a853f7c"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2020-06-19 07:41:51.000000000","tz":180},"committer":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2020-06-22 21:30:12.000000000","tz":180},"subject":"DNM test new CI","message":"DNM test new CI\n\nChange-Id: If15fedb121e76c661774d1c7717c3e300f49618e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/c16810621f44c468a24f2dd7d06805d59395d5f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/c16810621f44c468a24f2dd7d06805d59395d5f9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
