)]}'
{"id":"openstack%2Fansible-collections-openstack~757081","triplet_id":"openstack%2Fansible-collections-openstack~master~I4fd36bf3ed347e020151721a6b56d1cac0a8fd23","project":"openstack/ansible-collections-openstack","branch":"master","hashtags":[],"change_id":"I4fd36bf3ed347e020151721a6b56d1cac0a8fd23","subject":"Refactor loadbalancer module","status":"MERGED","created":"2020-10-09 11:29:57.000000000","updated":"2020-11-30 20:20:54.000000000","submitted":"2020-11-30 20:20:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"757081","meta_rev_id":"8173bdddbdb610077439e02ed7006c09824cb2ec","_number":757081,"virtual_id_number":757081,"owner":{"_account_id":29260,"name":"Jesper Schmitz Mouridsen","email":"jesper@schmitz.computer","username":"jsmdk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-30 20:20:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"date":"2020-11-26 15:34:51.000000000","_account_id":29260,"name":"Jesper Schmitz Mouridsen","email":"jesper@schmitz.computer","username":"jsmdk"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"all":[{"value":2,"date":"2020-11-27 11:18:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29260,"name":"Jesper Schmitz Mouridsen","email":"jesper@schmitz.computer","username":"jsmdk"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"all":[{"value":1,"date":"2020-11-30 19:10:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29260,"name":"Jesper Schmitz Mouridsen","email":"jesper@schmitz.computer","username":"jsmdk"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29260,"name":"Jesper Schmitz Mouridsen","email":"jesper@schmitz.computer","username":"jsmdk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-12 16:08:21.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-10-12 18:39:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-23 21:05:57.000000000","updated_by":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"reviewer":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"state":"REVIEWER"}],"messages":[{"id":"028ccc032873d7f103c2f021927931004ef2134c","author":{"_account_id":29260,"name":"Jesper Schmitz Mouridsen","email":"jesper@schmitz.computer","username":"jsmdk"},"date":"2020-10-09 11:29:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"60e8b2283840ea202fee3245274bceddfe37beb9","author":{"_account_id":29260,"name":"Jesper Schmitz Mouridsen","email":"jesper@schmitz.computer","username":"jsmdk"},"date":"2020-10-09 11:32:26.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"fe93741dc4ec2f7923a983892deecb8a28ead6b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 13:00:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/296451fc318441998b108998f2401885 : SUCCESS in 3m 15s\n- openstack-tox-linters-ansible-devel https://zuul.opendev.org/t/openstack/build/1d3e1328314f4f218b967bc29c1373b7 : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-linters-ansible-2.10 https://zuul.opendev.org/t/openstack/build/c037bf0bc0d242568923fbffbdfcb6dd : SUCCESS in 5m 50s\n- openstack-tox-linters-ansible-2.9 https://zuul.opendev.org/t/openstack/build/f12730460043404b91398059902cb5c4 : SUCCESS in 6m 58s\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/935c8a602c9d453d940a9fa402ce81e0 : SUCCESS in 38m 24s\n- ansible-collections-openstack-functional-devstack-releases https://zuul.opendev.org/t/openstack/build/73abdbf35c734477aa131abb811f6950 : SUCCESS in 40m 50s\n- ansible-collections-openstack-functional-devstack-ansible-2.9 https://zuul.opendev.org/t/openstack/build/7d57428f12a94506aff1ac3c177c1d7b : SUCCESS in 46m 58s\n- ansible-collections-openstack-functional-devstack-ansible-2.10 https://zuul.opendev.org/t/openstack/build/2f5640eccbcb4ff69bb1adf66c318919 : SUCCESS in 51m 20s\n- ansible-collections-openstack-functional-devstack-ansible-devel https://zuul.opendev.org/t/openstack/build/c5327843b7da42b69dca4dd916224b36 : SUCCESS in 45m 41s (non-voting)\n- ansible-collections-openstack-functional-devstack-ansible-pip https://zuul.opendev.org/t/openstack/build/902c60df31d64903b9db030a98190682 : SUCCESS in 44m 51s\n- ansible-collections-openstack-functional-devstack-ussuri-ansible-2.9 https://zuul.opendev.org/t/openstack/build/f20fe5ce839f43b8807a893c170f7cd3 : SUCCESS in 56m 47s\n- ansible-collections-openstack-functional-devstack-train-ansible-2.9 https://zuul.opendev.org/t/openstack/build/79205cd7e087495f8331f539fc4d0887 : SUCCESS in 49m 31s\n- ansible-collections-openstack-functional-devstack-stein-ansible-2.9 https://zuul.opendev.org/t/openstack/build/b81961cd1fbd40cf92fd67bd6c1013a0 : SUCCESS in 48m 14s\n- ansible-collections-openstack-functional-devstack-rocky-ansible-2.9 https://zuul.opendev.org/t/openstack/build/8372ebb767d3462c8c3a5348b620bbd0 : SUCCESS in 54m 05s\n- ansible-collections-openstack-functional-devstack-queens-ansible-2.9 https://zuul.opendev.org/t/openstack/build/525cf17d9c5d470d9e130253f2040358 : SUCCESS in 41m 39s\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/550659792e1841e6a97a706a915fe214 : SUCCESS in 26m 12s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/4870b5293e7f418ea26952f404d8c896 : SUCCESS in 25m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1367ab278460af820ceed408f19b4eec3a70180e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-10-12 16:08:21.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)\n\nJust minor typo in the docs, otherwise looks ok via review.","accounts_in_message":[],"_revision_number":2},{"id":"143f64f76bc930086b17dcda7547b2d7a55e458d","author":{"_account_id":29260,"name":"Jesper Schmitz Mouridsen","email":"jesper@schmitz.computer","username":"jsmdk"},"date":"2020-10-12 17:14:27.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9f1e134ef0cfbd5e37b7a81e681674a610c4f116","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 18:39:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/e74ee804bc7c4d16a934c04ea2be9238 : SUCCESS in 3m 40s\n- openstack-tox-linters-ansible-devel https://zuul.opendev.org/t/openstack/build/549664346ca54ac68cc3abf17af2edf7 : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-linters-ansible-2.10 https://zuul.opendev.org/t/openstack/build/a33f075be9a447a9994ceb2c630baebc : SUCCESS in 4m 55s\n- openstack-tox-linters-ansible-2.9 https://zuul.opendev.org/t/openstack/build/d494f7e540fa4e38bfa9e2c3e4a953a8 : SUCCESS in 7m 05s\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/1926c9bd364243028c981c009c83350d : SUCCESS in 52m 58s\n- ansible-collections-openstack-functional-devstack-releases https://zuul.opendev.org/t/openstack/build/3c87fb1670674043b523d0e79d07a600 : SUCCESS in 47m 13s\n- ansible-collections-openstack-functional-devstack-ansible-2.9 https://zuul.opendev.org/t/openstack/build/3acea7979ea5415388cdbebfe1790d6f : SUCCESS in 33m 21s\n- ansible-collections-openstack-functional-devstack-ansible-2.10 https://zuul.opendev.org/t/openstack/build/492257be259842e8b5b3fd6658b6a069 : SUCCESS in 52m 47s\n- ansible-collections-openstack-functional-devstack-ansible-devel https://zuul.opendev.org/t/openstack/build/0697dfa3c481424bb6bddec7b68ca689 : SUCCESS in 53m 34s (non-voting)\n- ansible-collections-openstack-functional-devstack-ansible-pip https://zuul.opendev.org/t/openstack/build/75035ab2900a4065a791178e5a417041 : SUCCESS in 51m 45s\n- ansible-collections-openstack-functional-devstack-ussuri-ansible-2.9 https://zuul.opendev.org/t/openstack/build/60d9195b73184dc99472b3cbb401a198 : SUCCESS in 36m 57s\n- ansible-collections-openstack-functional-devstack-train-ansible-2.9 https://zuul.opendev.org/t/openstack/build/b0e953eead7e45fa87d4914efc79ac4e : SUCCESS in 45m 43s\n- ansible-collections-openstack-functional-devstack-stein-ansible-2.9 https://zuul.opendev.org/t/openstack/build/9df86976e3184f4db5b66102b40f2303 : SUCCESS in 54m 34s\n- ansible-collections-openstack-functional-devstack-rocky-ansible-2.9 https://zuul.opendev.org/t/openstack/build/27d19f00819f4bc5b1f334aa6d24676b : SUCCESS in 54m 57s\n- ansible-collections-openstack-functional-devstack-queens-ansible-2.9 https://zuul.opendev.org/t/openstack/build/c0eef797046b461d86c69875d238d52c : SUCCESS in 52m 27s\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/fed422142d794e4a939efca4c1d8deef : SUCCESS in 23m 02s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/dc758e6b69654f9d85b37fe2b7212fa1 : SUCCESS in 23m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b4cf7c0412d3a07547b28e63dcf4dd6080cf4e93","author":{"_account_id":29260,"name":"Jesper Schmitz Mouridsen","email":"jesper@schmitz.computer","username":"jsmdk"},"date":"2020-11-14 16:41:37.000000000","message":"Patch Set 3:\n\n\u003e (2 comments)\n \u003e \n \u003e Just minor typo in the docs, otherwise looks ok via review.\n\n\nHi, thanks for reviewing, I changed the docs in patch set 3","accounts_in_message":[],"_revision_number":3},{"id":"57e8cfd0fdd2fac1aa355666fc6c07656776cde0","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2020-11-23 21:05:57.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)\n\nnit inside\nCan we add tests for it?","accounts_in_message":[],"_revision_number":3},{"id":"f8be3a4bfe1f7c22cd8dd55b86b07f143abfc124","author":{"_account_id":29260,"name":"Jesper Schmitz Mouridsen","email":"jesper@schmitz.computer","username":"jsmdk"},"date":"2020-11-26 15:34:51.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 3: Code-Review+1\n\u003e \n\u003e (1 comment)\n\u003e \n\u003e nit inside\n\u003e Can we add tests for it?\n\n\nWe discussed tests briefly in the lb_health_monitor change https://review.opendev.org/c/openstack/ansible-collections-openstack/+/753306. My best suggestion for a test is still at https://gist.github.com/jsm222/352ea6fe7ae4446bed39ed788d738827, that is tested on an octavia enabled devstack (Not in zuul). Otherwise I could use some pointers for writing unit tests for this project.\nThanks.","accounts_in_message":[],"_revision_number":3},{"id":"7372a46f00ea06cf03731ada7da83a31f0018885","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2020-11-27 11:18:39.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"f38c504f05b2b3d51167e6b7d78199b77b181889","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2020-11-30 19:10:03.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"4135fc756e1c43bfeb8bd8bdd1f2af95e9708ca2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-30 19:10:19.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"974541283d1305e1d6eee5ab66dc08038603841c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-30 20:20:53.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/233cd3eaafdc4b6782958409744fa316 : SUCCESS in 3m 46s\n- openstack-tox-linters-ansible-2.10 https://zuul.opendev.org/t/openstack/build/04abff03ef8d4d3b8edae417f1f04044 : SUCCESS in 8m 00s\n- openstack-tox-linters-ansible-2.9 https://zuul.opendev.org/t/openstack/build/a69d3046a10b4a7bb58535e4091fc544 : SUCCESS in 7m 01s\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/726d56e2be554528a326d4e6838aee28 : SUCCESS in 1h 05m 50s\n- ansible-collections-openstack-functional-devstack-releases https://zuul.opendev.org/t/openstack/build/227569253681454f85c24f05b824ccff : SUCCESS in 36m 46s\n- ansible-collections-openstack-functional-devstack-ansible-2.9 https://zuul.opendev.org/t/openstack/build/11c67b69e7e344a6bdb31b34dafa1ecc : SUCCESS in 54m 33s\n- ansible-collections-openstack-functional-devstack-ansible-2.10 https://zuul.opendev.org/t/openstack/build/1154456053e44524b02ac5d5bddceafa : SUCCESS in 38m 58s\n- ansible-collections-openstack-functional-devstack-ansible-pip https://zuul.opendev.org/t/openstack/build/f889408761be4aa6abc3e45d8b406ab3 : SUCCESS in 49m 05s\n- ansible-collections-openstack-functional-devstack-victoria-ansible-2.10 https://zuul.opendev.org/t/openstack/build/a16fed737b63410385683abb6f954b4f : SUCCESS in 38m 54s\n- ansible-collections-openstack-functional-devstack-ussuri-ansible-2.10 https://zuul.opendev.org/t/openstack/build/c78a7613301b4b659204d8136f84683b : SUCCESS in 51m 04s\n- ansible-collections-openstack-functional-devstack-train-ansible-2.10 https://zuul.opendev.org/t/openstack/build/e66cb25cb36c43b5b67e9f87bbff7ec7 : SUCCESS in 51m 03s\n- ansible-collections-openstack-functional-devstack-stein-ansible-2.10 https://zuul.opendev.org/t/openstack/build/a70b219695a3453092dd588880aa1e76 : SUCCESS in 47m 58s\n- ansible-collections-openstack-functional-devstack-rocky-ansible-2.10 https://zuul.opendev.org/t/openstack/build/1fabecc85d724ce7a6bf39b3f78af58f : SUCCESS in 50m 35s\n- ansible-collections-openstack-functional-devstack-queens-ansible-2.10 https://zuul.opendev.org/t/openstack/build/c61c258c52604f5b9c3ce1d9d5971d27 : SUCCESS in 49m 05s","accounts_in_message":[],"_revision_number":3},{"id":"8173bdddbdb610077439e02ed7006c09824cb2ec","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-30 20:20:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ab96eb6a11d0f876b00a87a3efbfaaf4a884a6ad","revisions":{"8b887ca0b9adb8ae14464c5e16657ce0b740dda3":{"kind":"REWORK","_number":1,"created":"2020-10-09 11:29:57.000000000","uploader":{"_account_id":29260,"name":"Jesper Schmitz Mouridsen","email":"jesper@schmitz.computer","username":"jsmdk"},"ref":"refs/changes/81/757081/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-collections-openstack","ref":"refs/changes/81/757081/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/81/757081/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/81/757081/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/81/757081/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/81/757081/1"}}},"commit":{"parents":[{"commit":"38e61994c74027745165ae1d79b08526f371dfe1","subject":"Merge \"Add volume_snapshot_info module\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/38e61994c74027745165ae1d79b08526f371dfe1"}]}],"author":{"name":"Jesper Schmitz Mouridsen","email":"jesper@schmitz.computer","date":"2020-10-09 11:29:33.000000000","tz":120},"committer":{"name":"Jesper Schmitz Mouridsen","email":"jesper@schmitz.computer","date":"2020-10-09 11:29:33.000000000","tz":120},"subject":"Refactor loadbalancer module  * enable check_mode  * enable allowed_cidrs on listener if octavia version is \u003e\u003d 2.12  * Only send flavor_id if it is not None","message":"Refactor loadbalancer module\n * enable check_mode\n * enable allowed_cidrs on listener if octavia version is \u003e\u003d 2.12\n * Only send flavor_id if it is not None\n\nChange-Id: I4fd36bf3ed347e020151721a6b56d1cac0a8fd23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/8b887ca0b9adb8ae14464c5e16657ce0b740dda3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/8b887ca0b9adb8ae14464c5e16657ce0b740dda3"}]},"branch":"refs/heads/master"},"ff57e43fa4116af616e97dfb0139a0140f932f77":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-10-09 11:32:26.000000000","uploader":{"_account_id":29260,"name":"Jesper Schmitz Mouridsen","email":"jesper@schmitz.computer","username":"jsmdk"},"ref":"refs/changes/81/757081/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-collections-openstack","ref":"refs/changes/81/757081/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/81/757081/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/81/757081/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/81/757081/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/81/757081/2"}}},"commit":{"parents":[{"commit":"38e61994c74027745165ae1d79b08526f371dfe1","subject":"Merge \"Add volume_snapshot_info module\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/38e61994c74027745165ae1d79b08526f371dfe1"}]}],"author":{"name":"Jesper Schmitz Mouridsen","email":"jesper@schmitz.computer","date":"2020-10-09 11:29:33.000000000","tz":120},"committer":{"name":"Jesper Schmitz Mouridsen","email":"jesper@schmitz.computer","date":"2020-10-09 11:31:25.000000000","tz":0},"subject":"Refactor loadbalancer module","message":"Refactor loadbalancer module\n\n * enable check_mode\n * enable allowed_cidrs on listener if octavia version is \u003e\u003d 2.12\n * Only send flavor_id if it is not None\n\nChange-Id: I4fd36bf3ed347e020151721a6b56d1cac0a8fd23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/ff57e43fa4116af616e97dfb0139a0140f932f77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/ff57e43fa4116af616e97dfb0139a0140f932f77"}]},"branch":"refs/heads/master"},"ab96eb6a11d0f876b00a87a3efbfaaf4a884a6ad":{"kind":"REWORK","_number":3,"created":"2020-10-12 17:14:27.000000000","uploader":{"_account_id":29260,"name":"Jesper Schmitz Mouridsen","email":"jesper@schmitz.computer","username":"jsmdk"},"ref":"refs/changes/81/757081/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-collections-openstack","ref":"refs/changes/81/757081/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/81/757081/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/81/757081/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/81/757081/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/81/757081/3"}}},"commit":{"parents":[{"commit":"38e61994c74027745165ae1d79b08526f371dfe1","subject":"Merge \"Add volume_snapshot_info module\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/38e61994c74027745165ae1d79b08526f371dfe1"}]}],"author":{"name":"Jesper Schmitz Mouridsen","email":"jesper@schmitz.computer","date":"2020-10-09 11:29:33.000000000","tz":120},"committer":{"name":"Jesper Schmitz Mouridsen","email":"jesper@schmitz.computer","date":"2020-10-12 17:08:45.000000000","tz":120},"subject":"Refactor loadbalancer module","message":"Refactor loadbalancer module\n\n * enable check_mode\n * enable allowed_cidrs on listener if octavia version is \u003e\u003d 2.12\n * Only send flavor_id if it is not None\n\nChange-Id: I4fd36bf3ed347e020151721a6b56d1cac0a8fd23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/ab96eb6a11d0f876b00a87a3efbfaaf4a884a6ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/ab96eb6a11d0f876b00a87a3efbfaaf4a884a6ad"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
