)]}'
{"id":"openstack%2Fansible-collections-openstack~767719","triplet_id":"openstack%2Fansible-collections-openstack~master~I0fd13cc98c161067b614dddde29a01372672db01","project":"openstack/ansible-collections-openstack","branch":"master","topic":"remove-update-nova-flavor","hashtags":[],"change_id":"I0fd13cc98c161067b614dddde29a01372672db01","subject":"[WIP] Add parameter to flavor recreation on update","status":"ABANDONED","created":"2020-12-18 13:38:33.000000000","updated":"2021-08-05 11:40:13.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1a6ab3e44474dba389f71eaea469be24f2f868ac","_number":767719,"virtual_id_number":767719,"owner":{"_account_id":30158,"name":"Cyril Grosjean","email":"cyril.grosjean@wifirst.fr","username":"cgrosjean"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-12-24 11:01:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30158,"name":"Cyril Grosjean","email":"cyril.grosjean@wifirst.fr","username":"cgrosjean"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"all":[{"value":2,"date":"2020-12-28 14:39:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30158,"name":"Cyril Grosjean","email":"cyril.grosjean@wifirst.fr","username":"cgrosjean"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":30158,"name":"Cyril Grosjean","email":"cyril.grosjean@wifirst.fr","username":"cgrosjean"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2020-12-24 09:41:52.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":30158,"name":"Cyril Grosjean","email":"cyril.grosjean@wifirst.fr","username":"cgrosjean"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30158,"name":"Cyril Grosjean","email":"cyril.grosjean@wifirst.fr","username":"cgrosjean"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-18 13:51:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-21 11:30:00.000000000","updated_by":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"reviewer":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"state":"REVIEWER"}],"messages":[{"id":"f145258b0272f1e1c087694e912e2bfcf81da266","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30158,"name":"Cyril Grosjean","email":"cyril.grosjean@wifirst.fr","username":"cgrosjean"},"date":"2020-12-18 13:38:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"acb3145ea6f451ba3ef351f6f79029d7c31522b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 13:51:34.000000000","message":"Patch Set 1: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/f9a25e6292684f9a9953dcb800cc31ce : FAILURE in 2m 58s\n- openstack-tox-linters-ansible-devel https://zuul.opendev.org/t/openstack/build/0a2bac2a2c3a496c9d39a193e53559ca : FAILURE in 4m 39s (non-voting)\n- openstack-tox-linters-ansible-2.10 https://zuul.opendev.org/t/openstack/build/9932bf95ef964f6aa2990289b6da4198 : FAILURE in 5m 07s\n- openstack-tox-linters-ansible-2.9 https://zuul.opendev.org/t/openstack/build/682d5d6d2af74aa0b9b61c3527e204a5 : FAILURE in 5m 53s\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- ansible-collections-openstack-functional-devstack-releases https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- ansible-collections-openstack-functional-devstack-ansible-2.9 https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- ansible-collections-openstack-functional-devstack-ansible-2.10 https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- ansible-collections-openstack-functional-devstack-ansible-devel https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- ansible-collections-openstack-functional-devstack-ansible-pip https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- ansible-collections-openstack-functional-devstack-victoria-ansible-2.10 https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- ansible-collections-openstack-functional-devstack-ussuri-ansible-2.10 https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- ansible-collections-openstack-functional-devstack-train-ansible-2.10 https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- ansible-collections-openstack-functional-devstack-stein-ansible-2.10 https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- ansible-collections-openstack-functional-devstack-rocky-ansible-2.10 https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- ansible-collections-openstack-functional-devstack-queens-ansible-2.10 https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n\nWarning:\n  Job tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: file plugins/modules/os_nova_flavor.py has only 1 lines\u0027\n  Job openstack-tox-linters-ansible-devel: unable to map line for file comments: \n  stderr: \u0027fatal: file plugins/modules/os_nova_flavor.py has only 1 lines\u0027\n  Job openstack-tox-linters-ansible-2.9: unable to map line for file comments: \n  stderr: \u0027fatal: file plugins/modules/os_nova_flavor.py has only 1 lines\u0027\n  Job openstack-tox-linters-ansible-2.10: unable to map line for file comments: \n  stderr: \u0027fatal: file plugins/modules/os_nova_flavor.py has only 1 lines\u0027\n  Comments left for invalid file plugins/modules/os_nova_flavor.py","accounts_in_message":[],"_revision_number":1},{"id":"650b2aeacc442e51f096ac5bf7a81710f27b5b64","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30158,"name":"Cyril Grosjean","email":"cyril.grosjean@wifirst.fr","username":"cgrosjean"},"date":"2020-12-18 14:17:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"300c1063fdbd0d31c7a984e588b966991e743f76","author":{"_account_id":30158,"name":"Cyril Grosjean","email":"cyril.grosjean@wifirst.fr","username":"cgrosjean"},"date":"2020-12-18 15:13:38.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"9d22a6adfb14df0483438988179c93f85091544b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 15:48:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/4987e1c352ee474eb52e2fd6215de237 : SUCCESS in 3m 11s\n- openstack-tox-linters-ansible-devel https://zuul.opendev.org/t/openstack/build/dad50654a4684fc5af1333e198bc831c : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-linters-ansible-2.10 https://zuul.opendev.org/t/openstack/build/513fa1a8d13d4d558deab80c1ff4bf3f : SUCCESS in 7m 09s\n- openstack-tox-linters-ansible-2.9 https://zuul.opendev.org/t/openstack/build/a9ddceab0df54e71ad787170e5af8868 : SUCCESS in 8m 54s\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/356eba0d1d794610b7deb6029f64e1de : SUCCESS in 54m 02s\n- ansible-collections-openstack-functional-devstack-releases https://zuul.opendev.org/t/openstack/build/f3df33a5e07645a98c3c0d4b20f62983 : SUCCESS in 54m 05s\n- ansible-collections-openstack-functional-devstack-ansible-2.9 https://zuul.opendev.org/t/openstack/build/bfba81395a004893808b86cb20d82985 : SUCCESS in 36m 23s\n- ansible-collections-openstack-functional-devstack-ansible-2.10 https://zuul.opendev.org/t/openstack/build/739342447b9d4947837835e627948b20 : SUCCESS in 45m 37s\n- ansible-collections-openstack-functional-devstack-ansible-devel https://zuul.opendev.org/t/openstack/build/697c22c275f9450e95170f68d7ab9bde : SUCCESS in 47m 21s (non-voting)\n- ansible-collections-openstack-functional-devstack-ansible-pip https://zuul.opendev.org/t/openstack/build/436ff95e8671449ba2e05128fa28a60a : SUCCESS in 1h 02m 06s\n- ansible-collections-openstack-functional-devstack-victoria-ansible-2.10 https://zuul.opendev.org/t/openstack/build/e29901e86abd42678d79fe59f1a9f26c : SUCCESS in 33m 07s\n- ansible-collections-openstack-functional-devstack-ussuri-ansible-2.10 https://zuul.opendev.org/t/openstack/build/dc00cb2267eb407ba7715597074cbb89 : SUCCESS in 54m 55s\n- ansible-collections-openstack-functional-devstack-train-ansible-2.10 https://zuul.opendev.org/t/openstack/build/33df9b0fdccf42729e00e1c209800d90 : POST_FAILURE in 16m 37s (non-voting)\n- ansible-collections-openstack-functional-devstack-stein-ansible-2.10 https://zuul.opendev.org/t/openstack/build/744c36787e44441982d719b43323310c : SUCCESS in 41m 49s\n- ansible-collections-openstack-functional-devstack-rocky-ansible-2.10 https://zuul.opendev.org/t/openstack/build/e638dfbce53a41429bdb8809bc3586bd : SUCCESS in 40m 52s\n- ansible-collections-openstack-functional-devstack-queens-ansible-2.10 https://zuul.opendev.org/t/openstack/build/9addf0d658814e5dbc0fe8171bb76519 : SUCCESS in 44m 45s\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/f1fb38fc5187422f9856cd927319eca7 : SUCCESS in 26m 51s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/152e50279903454fa529672272f810c8 : SUCCESS in 25m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c4dbeef2d3c267c5466ce2ad680c1dcc6a91b1ef","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2020-12-21 11:30:00.000000000","message":"Patch Set 2: Code-Review-1\n\nUnfortunately it breaks whole idea of ansible task describing the desired state. By default flavor should be (re)created as it\u0027s configured so. What you need is maybe \"dry-run\" aka \"check_mode\"? I see this module supports it.","accounts_in_message":[],"_revision_number":2},{"id":"79e1e1d16998e5e709b08d75aa1a6a4dd7e19b31","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30158,"name":"Cyril Grosjean","email":"cyril.grosjean@wifirst.fr","username":"cgrosjean"},"date":"2020-12-24 09:30:37.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2f9a48abd59a32ab302af4d20ed55a4cba8fb7b5","author":{"_account_id":30158,"name":"Cyril Grosjean","email":"cyril.grosjean@wifirst.fr","username":"cgrosjean"},"date":"2020-12-24 09:33:48.000000000","message":"Patch Set 3: Workflow-1\n\nAfter discussion, I have inverted the parameter default, to keep the current behavior without the parameter, and having the possibility to fail if wanted so.\n\nI will need to add some tests for this, but if someone has an input on this, would be great.","accounts_in_message":[],"_revision_number":3},{"id":"5593f77dede47f7b707281ccbd13544d5e0bc133","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2020-12-24 09:36:54.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"15fae6ec40f7ddbca3f8ce446de02173118d094c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30158,"name":"Cyril Grosjean","email":"cyril.grosjean@wifirst.fr","username":"cgrosjean"},"date":"2020-12-24 09:40:02.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e4d96c584746db7792022922d249c03b3e7965e1","author":{"_account_id":30158,"name":"Cyril Grosjean","email":"cyril.grosjean@wifirst.fr","username":"cgrosjean"},"date":"2020-12-24 09:41:52.000000000","message":"Patch Set 4: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2e8a0ebd9b014f7b0a043658094e67cbdf830f95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-24 11:01:38.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/ea77d056c46e4b4492830eb18ce21e84 : SUCCESS in 2m 50s\n- openstack-tox-linters-ansible-devel https://zuul.opendev.org/t/openstack/build/991aeecce7d74c14be04d15469544ee6 : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-linters-ansible-2.10 https://zuul.opendev.org/t/openstack/build/97a4cb870a994fa3a1d69842f922fde3 : SUCCESS in 5m 36s\n- openstack-tox-linters-ansible-2.9 https://zuul.opendev.org/t/openstack/build/31cff1938af143f7a25430a6b99065fa : SUCCESS in 6m 12s\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/62f153d70b354c19bc736ac80835f1d9 : SUCCESS in 1h 04m 56s\n- ansible-collections-openstack-functional-devstack-releases https://zuul.opendev.org/t/openstack/build/43797d1e31654cd4ac9ac8609bd21a24 : SUCCESS in 41m 29s\n- ansible-collections-openstack-functional-devstack-ansible-2.9 https://zuul.opendev.org/t/openstack/build/84cf87ba4eb740d1bcb48574607e6947 : SUCCESS in 34m 24s\n- ansible-collections-openstack-functional-devstack-ansible-2.10 https://zuul.opendev.org/t/openstack/build/dd3b556c7c284dec8cc6f5296f62e79d : SUCCESS in 1h 08m 41s\n- ansible-collections-openstack-functional-devstack-ansible-devel https://zuul.opendev.org/t/openstack/build/240c5f3310a14986a90433512094c3dc : SUCCESS in 40m 21s (non-voting)\n- ansible-collections-openstack-functional-devstack-ansible-pip https://zuul.opendev.org/t/openstack/build/3eeb876785644a40a39e47b7f67eb22d : SUCCESS in 1h 08m 22s\n- ansible-collections-openstack-functional-devstack-victoria-ansible-2.10 https://zuul.opendev.org/t/openstack/build/fe98c3afa20f49f1ac35f4cd45b29860 : SUCCESS in 33m 57s\n- ansible-collections-openstack-functional-devstack-ussuri-ansible-2.10 https://zuul.opendev.org/t/openstack/build/48c417fa08fd403dad6a9f1f5dc70cf2 : SUCCESS in 44m 30s\n- ansible-collections-openstack-functional-devstack-train-ansible-2.10 https://zuul.opendev.org/t/openstack/build/5c84cb9873f147cea0d6027b4e7b4e4d : SUCCESS in 43m 55s (non-voting)\n- ansible-collections-openstack-functional-devstack-stein-ansible-2.10 https://zuul.opendev.org/t/openstack/build/f73b68f0c4ce4c6ea69be63299cb11b2 : SUCCESS in 49m 32s\n- ansible-collections-openstack-functional-devstack-rocky-ansible-2.10 https://zuul.opendev.org/t/openstack/build/509f9fe254ed43d79ff0c53ee48ccaa5 : SUCCESS in 44m 36s\n- ansible-collections-openstack-functional-devstack-queens-ansible-2.10 https://zuul.opendev.org/t/openstack/build/bc751649684c494b8fc56fc821eb023f : SUCCESS in 39m 35s\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/4c864d748121480f811d1d8cc5032d89 : SUCCESS in 24m 06s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/3108a519caac49e3a3f1de9d0937ad65 : SUCCESS in 19m 12s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4894316912f94cd2e885819465a8d509d6d4fc56","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2020-12-28 14:39:37.000000000","message":"Patch Set 4: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":4},{"id":"1a6ab3e44474dba389f71eaea469be24f2f868ac","tag":"autogenerated:gerrit:abandon","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2021-08-05 11:40:13.000000000","message":"Abandoned\n\nThis patch will be abandoned due to the long inactivity, please feel free to restore it in case you would like to continue to work on it. Thanks","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"5dd93e63f85632a3f04f7475e7673ea1f5cd47bc","revisions":{"f905397bd2a0a974a0d41573db6fe173141ad5db":{"kind":"REWORK","_number":1,"created":"2020-12-18 13:38:33.000000000","uploader":{"_account_id":30158,"name":"Cyril Grosjean","email":"cyril.grosjean@wifirst.fr","username":"cgrosjean"},"ref":"refs/changes/19/767719/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-collections-openstack","ref":"refs/changes/19/767719/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/19/767719/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/19/767719/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/19/767719/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/19/767719/1"}}},"commit":{"parents":[{"commit":"40a32c1e8badd2a0b0c1482eb9a9e541a430719b","subject":"Merge \"Аdd designate to install with devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/40a32c1e8badd2a0b0c1482eb9a9e541a430719b"}]}],"author":{"name":"Cyril Grosjean","email":"cyril.grosjean@wifirst.fr","date":"2020-12-18 13:27:39.000000000","tz":60},"committer":{"name":"Cyril Grosjean","email":"cyril.grosjean@wifirst.fr","date":"2020-12-18 13:36:02.000000000","tz":60},"subject":"Remove flavor update and add param for it","message":"Remove flavor update and add param for it\n\nThis commit remove the default flavor update when a flavor exists\nand a change is required, which consist of deleting and recreating\nthe flavor with the same name.\n\nIt could be hurtful on some operations as deleting a flavor with\nvirtual machines still spawned on it would remove the information.\n\nThis change alse add a param `update_flavor` (default to no/False)\nwhich if set to yes/True restore the previous behavior.\n\nSign-off-by: Cyril Grosjean \u003ccyril.grosjean@wifirst.fr\u003e\nChange-Id: I0fd13cc98c161067b614dddde29a01372672db01\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/f905397bd2a0a974a0d41573db6fe173141ad5db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/f905397bd2a0a974a0d41573db6fe173141ad5db"}]},"branch":"refs/heads/master"},"d9848499a44470decbe9c1aecaaec5246f723679":{"kind":"REWORK","_number":2,"created":"2020-12-18 14:17:28.000000000","uploader":{"_account_id":30158,"name":"Cyril Grosjean","email":"cyril.grosjean@wifirst.fr","username":"cgrosjean"},"ref":"refs/changes/19/767719/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-collections-openstack","ref":"refs/changes/19/767719/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/19/767719/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/19/767719/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/19/767719/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/19/767719/2"}}},"commit":{"parents":[{"commit":"40a32c1e8badd2a0b0c1482eb9a9e541a430719b","subject":"Merge \"Аdd designate to install with devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/40a32c1e8badd2a0b0c1482eb9a9e541a430719b"}]}],"author":{"name":"Cyril Grosjean","email":"cyril.grosjean@wifirst.fr","date":"2020-12-18 13:27:39.000000000","tz":60},"committer":{"name":"Cyril Grosjean","email":"cyril.grosjean@wifirst.fr","date":"2020-12-18 14:17:13.000000000","tz":60},"subject":"Remove flavor update and add param for it","message":"Remove flavor update and add param for it\n\nThis commit remove the default flavor update when a flavor exists\nand a change is required, which consist of deleting and recreating\nthe flavor with the same name.\n\nIt could be hurtful on some operations as deleting a flavor with\nvirtual machines still spawned on it would remove the information.\n\nThis change alse add a param `update_flavor` (default to no/False)\nwhich if set to yes/True restore the previous behavior.\n\nSign-off-by: Cyril Grosjean \u003ccyril.grosjean@wifirst.fr\u003e\nChange-Id: I0fd13cc98c161067b614dddde29a01372672db01\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/d9848499a44470decbe9c1aecaaec5246f723679"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/d9848499a44470decbe9c1aecaaec5246f723679"}]},"branch":"refs/heads/master"},"8241eb21a500421c1dcd531364d484afe1b3059e":{"kind":"REWORK","_number":3,"created":"2020-12-24 09:30:37.000000000","uploader":{"_account_id":30158,"name":"Cyril Grosjean","email":"cyril.grosjean@wifirst.fr","username":"cgrosjean"},"ref":"refs/changes/19/767719/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-collections-openstack","ref":"refs/changes/19/767719/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/19/767719/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/19/767719/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/19/767719/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/19/767719/3"}}},"commit":{"parents":[{"commit":"40a32c1e8badd2a0b0c1482eb9a9e541a430719b","subject":"Merge \"Аdd designate to install with devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/40a32c1e8badd2a0b0c1482eb9a9e541a430719b"}]}],"author":{"name":"Cyril Grosjean","email":"cyril.grosjean@wifirst.fr","date":"2020-12-18 13:27:39.000000000","tz":60},"committer":{"name":"Cyril Grosjean","email":"cyril.grosjean@wifirst.fr","date":"2020-12-24 09:30:04.000000000","tz":60},"subject":"[WIP] Add parameter to flavor recreation on update","message":"[WIP] Add parameter to flavor recreation on update\n\nThis commit add a parameter `recreate_flavor` that if set to false\nwill make the module fail if a flavor already exist and needs to be\nupdated\n\nThis recreation process could be hurtful on some production envs,\nas deleting a flavor with virtual machines still spawned on it\nwould remove the flavor information on the VM.\n\nSign-off-by: Cyril Grosjean \u003ccyril.grosjean@wifirst.fr\u003e\nChange-Id: I0fd13cc98c161067b614dddde29a01372672db01\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/8241eb21a500421c1dcd531364d484afe1b3059e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/8241eb21a500421c1dcd531364d484afe1b3059e"}]},"branch":"refs/heads/master"},"5dd93e63f85632a3f04f7475e7673ea1f5cd47bc":{"kind":"REWORK","_number":4,"created":"2020-12-24 09:40:02.000000000","uploader":{"_account_id":30158,"name":"Cyril Grosjean","email":"cyril.grosjean@wifirst.fr","username":"cgrosjean"},"ref":"refs/changes/19/767719/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-collections-openstack","ref":"refs/changes/19/767719/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/19/767719/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/19/767719/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/19/767719/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/19/767719/4"}}},"commit":{"parents":[{"commit":"40a32c1e8badd2a0b0c1482eb9a9e541a430719b","subject":"Merge \"Аdd designate to install with devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/40a32c1e8badd2a0b0c1482eb9a9e541a430719b"}]}],"author":{"name":"Cyril Grosjean","email":"cyril.grosjean@wifirst.fr","date":"2020-12-18 13:27:39.000000000","tz":60},"committer":{"name":"Cyril Grosjean","email":"cyril.grosjean@wifirst.fr","date":"2020-12-24 09:39:55.000000000","tz":60},"subject":"[WIP] Add parameter to flavor recreation on update","message":"[WIP] Add parameter to flavor recreation on update\n\nThis commit add a parameter `recreate_flavor` that if set to false\nwill make the module fail if a flavor already exist and needs to be\nupdated\n\nThis recreation process could be hurtful on some production envs,\nas deleting a flavor with virtual machines still spawned on it\nwould remove the flavor information on the VM.\n\nSign-off-by: Cyril Grosjean \u003ccyril.grosjean@wifirst.fr\u003e\nChange-Id: I0fd13cc98c161067b614dddde29a01372672db01\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/5dd93e63f85632a3f04f7475e7673ea1f5cd47bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/5dd93e63f85632a3f04f7475e7673ea1f5cd47bc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
