)]}'
{"id":"openstack%2Fansible-collections-openstack~849466","triplet_id":"openstack%2Fansible-collections-openstack~master~If8befd62ac59ef19debf0d63abe76f11b47c2da3","project":"openstack/ansible-collections-openstack","branch":"master","topic":"endpoint","attention_set":{},"removed_from_attention_set":{"32962":{"account":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},"last_update":"2022-07-12 12:21:02.000000000","reason":"\u003cGERRIT_ACCOUNT_32962\u003e replied on the change","reason_account":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"}}},"hashtags":[],"change_id":"If8befd62ac59ef19debf0d63abe76f11b47c2da3","subject":"Added assertions on endpoint module results","status":"MERGED","created":"2022-07-12 08:43:32.000000000","updated":"2022-07-12 13:08:23.000000000","submitted":"2022-07-12 13:08:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"849466-endpoint","meta_rev_id":"0ba4abc79a1188d61f44118313abfa58afea3738","_number":849466,"virtual_id_number":849466,"owner":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-12 13:08:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},"all":[{"value":2,"date":"2022-07-12 12:21:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},{"value":2,"date":"2022-07-12 12:15:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},"all":[{"value":1,"date":"2022-07-12 12:21:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-12 09:41:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-12 12:15:44.000000000","updated_by":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"reviewer":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"state":"REVIEWER"}],"messages":[{"id":"ab3f9d66ffce391d3b33cda372fcb6175c080493","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},"date":"2022-07-12 08:43:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"87469f88287dbc47b0a10e47d8bae6df870fcb45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 09:41:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/748942ff546146af9fd7552c093e65ad\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/7882c84906624fc8a005e4a023b833fb : SUCCESS in 8m 22s\n- openstack-tox-linters-ansible-devel https://zuul.opendev.org/t/openstack/build/8f6657b8b0c8428bb6f462efdb3e22ee : SUCCESS in 11m 32s (non-voting)\n- openstack-tox-linters-ansible-2.12 https://zuul.opendev.org/t/openstack/build/ce09570c9bd04bea834e5572f60fe706 : SUCCESS in 10m 08s\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/b91ce5f15578440b87398467907eb083 : SUCCESS in 43m 24s\n- ansible-collections-openstack-functional-devstack-releases https://zuul.opendev.org/t/openstack/build/79c60d108e074af6af0b72f5821fef17 : SUCCESS in 41m 25s\n- ansible-collections-openstack-functional-devstack-ansible-2.9 https://zuul.opendev.org/t/openstack/build/55a6ebcbd5834099857a21d912dd0822 : SUCCESS in 43m 22s\n- ansible-collections-openstack-functional-devstack-ansible-2.12 https://zuul.opendev.org/t/openstack/build/661f17a9cf944772ad38639657f7eecb : SUCCESS in 42m 16s\n- ansible-collections-openstack-functional-devstack-ansible-devel https://zuul.opendev.org/t/openstack/build/9948214294c447f89cc88cbb6552ecbf : SUCCESS in 44m 10s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/1ceb5538b1ef4897812e8b11bd687466 : SUCCESS in 26m 23s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/94dd1c97ed4246a081d40e8affb7e4b8 : SUCCESS in 26m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7afea71eaf3ac27d00b02307317e7408da7cbbb3","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2022-07-12 12:15:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3d68c7088c21b80fe35c7886748b053e6f525569","author":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},"date":"2022-07-12 12:21:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0a01e3126dfaf4178f3bc5677f236bfcaa4fd5ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 12:21:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f1645f418e1856a9988f9c54f4b953d016d422f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 13:08:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d96d5d08dad14e08bb7b67b4f77f2172\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/92235498e24e4518836340a6294bba3f : SUCCESS in 6m 15s\n- openstack-tox-linters-ansible-2.12 https://zuul.opendev.org/t/openstack/build/a84b21f2f62d42bcb49573c7608c2694 : SUCCESS in 9m 17s\n- ansible-collections-openstack-functional-devstack-releases https://zuul.opendev.org/t/openstack/build/58824bc218e147099fc8855ba478238b : SUCCESS in 46m 00s","accounts_in_message":[],"_revision_number":1},{"id":"0ba4abc79a1188d61f44118313abfa58afea3738","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 13:08:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"070e77feca9a148191d48475bd74d691f4d02b54","revisions":{"070e77feca9a148191d48475bd74d691f4d02b54":{"kind":"REWORK","_number":1,"created":"2022-07-12 08:43:32.000000000","uploader":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},"ref":"refs/changes/66/849466/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-collections-openstack","ref":"refs/changes/66/849466/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/66/849466/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/66/849466/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/66/849466/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/66/849466/1"}}},"commit":{"parents":[{"commit":"a120db2d60a09a359c57cf4e9ddfbaaa60006408","subject":"Update routers_info for the new SDK","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/a120db2d60a09a359c57cf4e9ddfbaaa60006408"}]}],"author":{"name":"Jakob Meng","email":"code@jakobmeng.de","date":"2022-07-12 08:39:33.000000000","tz":120},"committer":{"name":"Jakob Meng","email":"code@jakobmeng.de","date":"2022-07-12 08:43:22.000000000","tz":120},"subject":"Added assertions on endpoint module results","message":"Added assertions on endpoint module results\n\nSigned-off-by: Jakob Meng \u003ccode@jakobmeng.de\u003e\nChange-Id: If8befd62ac59ef19debf0d63abe76f11b47c2da3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/070e77feca9a148191d48475bd74d691f4d02b54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/070e77feca9a148191d48475bd74d691f4d02b54"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"}}]}],"submit_requirements":[]}
