)]}'
{"id":"openstack%2Fansible-collections-openstack~860223","triplet_id":"openstack%2Fansible-collections-openstack~stable%2F1.0.0~Ia5c6809d820865fd477d6956707a195ab672f069","project":"openstack/ansible-collections-openstack","branch":"stable/1.0.0","topic":"stable-release","attention_set":{},"removed_from_attention_set":{"32962":{"account":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},"last_update":"2022-10-04 13:45:41.000000000","reason":"\u003cGERRIT_ACCOUNT_32962\u003e replied on the change","reason_account":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"}}},"hashtags":[],"change_id":"Ia5c6809d820865fd477d6956707a195ab672f069","subject":"Release 1.10.0 version","status":"MERGED","created":"2022-10-04 09:07:38.000000000","updated":"2022-10-04 13:56:47.000000000","submitted":"2022-10-04 13:56:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"860223-stable-release","meta_rev_id":"49bb165f3b3eeeacdc4bbdafadbd7c1cc7770163","_number":860223,"virtual_id_number":860223,"owner":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":34208,"name":"Rafael Castillo","email":"rcastill@redhat.com","username":"rcastill"},{"value":0,"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-04 13:56:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":34208,"name":"Rafael Castillo","email":"rcastill@redhat.com","username":"rcastill"},"all":[{"value":2,"date":"2022-10-04 13:41:08.000000000","_account_id":34208,"name":"Rafael Castillo","email":"rcastill@redhat.com","username":"rcastill"},{"value":2,"date":"2022-10-04 13:45:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},"all":[{"value":0,"_account_id":34208,"name":"Rafael Castillo","email":"rcastill@redhat.com","username":"rcastill"},{"value":1,"date":"2022-10-04 13:45:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},{"_account_id":34208,"name":"Rafael Castillo","email":"rcastill@redhat.com","username":"rcastill"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-04 09:18:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-04 13:41:08.000000000","updated_by":{"_account_id":34208,"name":"Rafael Castillo","email":"rcastill@redhat.com","username":"rcastill"},"reviewer":{"_account_id":34208,"name":"Rafael Castillo","email":"rcastill@redhat.com","username":"rcastill"},"state":"REVIEWER"}],"messages":[{"id":"b05cd90f0c31055b9ef7e5d02bae074bd25f4616","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},"date":"2022-10-04 09:07:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ee7d1070a3fe74054e3113ce5f3a36bebd86c290","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-04 09:18:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4af0bf49bdf5485e9f620eb92e62d29d\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/fd4df7d2bd0e4076ac2e2f575f613726 : SUCCESS in 3m 59s\n- openstack-tox-linters-ansible-devel https://zuul.opendev.org/t/openstack/build/458c438e6d494b71a8b3f76f9d620e6c : SUCCESS in 6m 59s (non-voting)\n- openstack-tox-linters-ansible-2.12 https://zuul.opendev.org/t/openstack/build/c9bf5eaa97fc457e83642c2214da1a74 : SUCCESS in 8m 51s","accounts_in_message":[],"_revision_number":1},{"id":"371e03afe90ae429d1b686a836ef276ae53830a1","author":{"_account_id":34208,"name":"Rafael Castillo","email":"rcastill@redhat.com","username":"rcastill"},"date":"2022-10-04 13:41:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1664c92f5e3fec37d7ee4edef04243cd75379e32","author":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},"date":"2022-10-04 13:45:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fd809ba2e4a4c393b5f463e82f34dc8777acfeb2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-04 13:46:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"65a5cab308eb29cf3c010e3401ab4d858c729c9c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-04 13:56:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6c3782cd2654f9db4849db070be53b3\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/f5201491ecf148028a427f597e67886d : SUCCESS in 4m 32s\n- openstack-tox-linters-ansible-2.12 https://zuul.opendev.org/t/openstack/build/58de28b133824f12ab4808a00c34a1fb : SUCCESS in 9m 18s","accounts_in_message":[],"_revision_number":1},{"id":"49bb165f3b3eeeacdc4bbdafadbd7c1cc7770163","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-04 13:56:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1b03f918ac42315f2fdea6ca572ed6ea3e80494c","revisions":{"1b03f918ac42315f2fdea6ca572ed6ea3e80494c":{"kind":"REWORK","_number":1,"created":"2022-10-04 09:07:38.000000000","uploader":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},"ref":"refs/changes/23/860223/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-collections-openstack","ref":"refs/changes/23/860223/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/23/860223/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/23/860223/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/23/860223/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/23/860223/1"}}},"commit":{"parents":[{"commit":"c4350027349fd2199c8b623d9d87868de84450a3","subject":"Add SDK logging option for openstack ansible collections.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/c4350027349fd2199c8b623d9d87868de84450a3"}]}],"author":{"name":"Jakob Meng","email":"code@jakobmeng.de","date":"2022-10-04 09:07:23.000000000","tz":120},"committer":{"name":"Jakob Meng","email":"code@jakobmeng.de","date":"2022-10-04 09:07:23.000000000","tz":120},"subject":"Release 1.10.0 version","message":"Release 1.10.0 version\n\nChange-Id: Ia5c6809d820865fd477d6956707a195ab672f069\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/1b03f918ac42315f2fdea6ca572ed6ea3e80494c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/1b03f918ac42315f2fdea6ca572ed6ea3e80494c"}]},"branch":"refs/heads/stable/1.0.0"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"}}]}],"submit_requirements":[]}
