)]}'
{"id":"openstack%2Fansible-collections-openstack~922634","triplet_id":"openstack%2Fansible-collections-openstack~master~Ie2dde98ae47dd7108d554495d5025df175647d5c","project":"openstack/ansible-collections-openstack","branch":"master","topic":"fix/deleted","attention_set":{},"removed_from_attention_set":{"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2024-06-25 06:33:38.000000000","reason":"\u003cGERRIT_ACCOUNT_27900\u003e replied on the change","reason_account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}}},"hashtags":[],"change_id":"Ie2dde98ae47dd7108d554495d5025df175647d5c","subject":"Wait for deleted server to disappear from results","status":"MERGED","created":"2024-06-24 15:25:22.000000000","updated":"2024-06-25 07:40:43.000000000","submitted":"2024-06-25 07:40:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"922634-fix/deleted","meta_rev_id":"0ca6283c1d3e1c684797dd3ab325cb0fe263ee81","_number":922634,"virtual_id_number":922634,"owner":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-25 07:40:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-06-25 06:33:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-06-25 06:33:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-24 15:43:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"78aa7d0bc72722d5b5620c92e37ba778b14b0af7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-06-24 15:25:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9874e6306f606436b64ce08a35ae97bc2f0296bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-24 15:43:04.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/536ccd06545c4ea080095ba8a1ad771e\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/c17c423e26c84d779f12b9d3fbce0b91 : FAILURE in 2m 53s\n- openstack-tox-linters-ansible-devel https://zuul.opendev.org/t/openstack/build/c2522d107ed44b51bdbb66b32e657821 : FAILURE in 3m 19s (non-voting)\n- openstack-tox-linters-ansible-2.12 https://zuul.opendev.org/t/openstack/build/3b407200ff2e44fb9947871e2045fac8 : FAILURE in 4m 13s\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/403647ce616c40b6baf6161896367496 : SKIPPED Skipped due to failed job tox-pep8\n- ansible-collections-openstack-functional-devstack-releases https://zuul.opendev.org/t/openstack/build/c5c59c383b3b4071bf5e691c53795bd2 : SKIPPED Skipped due to failed job tox-pep8\n- ansible-collections-openstack-functional-devstack-ansible-2.9 https://zuul.opendev.org/t/openstack/build/7921bf53bbc949bebb81efd5615758b7 : SKIPPED Skipped due to failed job tox-pep8\n- ansible-collections-openstack-functional-devstack-ansible-2.12 https://zuul.opendev.org/t/openstack/build/5ab7ebbdcdc44a2dbbd98d9e68802337 : SKIPPED Skipped due to failed job tox-pep8\n- ansible-collections-openstack-functional-devstack-ansible-devel https://zuul.opendev.org/t/openstack/build/6a14d7a75de14057a2fcdf963a9f82f3 : SKIPPED Skipped due to failed job tox-pep8 (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/381f0b3474ab420ca769c21d84f0d60a : SKIPPED Skipped due to failed job tox-pep8 (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/c0fcc66c91534b558130bfa91771ff8f : SKIPPED Skipped due to failed job tox-pep8 (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"269b252ade10ed63505310f9507684f28bcc27fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-06-24 15:56:34.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"5704f4d73c7fb2d782542ed6d79eb87c9de1e07e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-24 17:32:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f12e7338f3594bb8b75fc2a1811fb03a\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/7fe8543e356f4ab08b8c851a93bb0187 : SUCCESS in 3m 30s\n- openstack-tox-linters-ansible-devel https://zuul.opendev.org/t/openstack/build/d680c2550753484cb3dc22dcc4394129 : FAILURE in 4m 36s (non-voting)\n- openstack-tox-linters-ansible-2.12 https://zuul.opendev.org/t/openstack/build/5ddda7e1f1eb4018bd7087745db5d620 : SUCCESS in 7m 25s\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/9d53f4371e134a63a3204c79785d8f9f : SUCCESS in 1h 15m 19s\n- ansible-collections-openstack-functional-devstack-releases https://zuul.opendev.org/t/openstack/build/757ab87aa793462c9e6cab60ea420515 : SUCCESS in 1h 03m 06s\n- ansible-collections-openstack-functional-devstack-ansible-2.9 https://zuul.opendev.org/t/openstack/build/29bc269746384b39b8088ef1f5cc2627 : SUCCESS in 1h 14m 03s\n- ansible-collections-openstack-functional-devstack-ansible-2.12 https://zuul.opendev.org/t/openstack/build/fae833fc00654ba4a0e15722db756237 : SUCCESS in 1h 17m 45s\n- ansible-collections-openstack-functional-devstack-ansible-devel https://zuul.opendev.org/t/openstack/build/cc8cbc4270204ccfb0abc13a9a8fb9c7 : FAILURE in 30m 35s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/2693f58c587b44509936bf2b8984844b : SUCCESS in 38m 59s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/b1d304a5afbc438e97857e0629a8ea82 : SUCCESS in 49m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1b9f8b17b563ca3816d0eeb21727a07ceafb1193","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-06-25 06:33:38.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f25530307deeea2e51bb8c9f0e500f564d8caad3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 06:34:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7d6815ae8619e6704fc2b490db860f6c13dfc311","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 07:40:43.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ea4be48ff544390b16043507d848ded\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/75fd4044dd2c4055afd68ce3caa1db98 : SUCCESS in 2m 42s\n- openstack-tox-linters-ansible-2.12 https://zuul.opendev.org/t/openstack/build/d411dcf29f9248d9b012c62b53d5ba25 : SUCCESS in 5m 52s\n- ansible-collections-openstack-functional-devstack-releases https://zuul.opendev.org/t/openstack/build/5bd50474b4d74bd3b6d5c5e0a2d601d9 : SUCCESS in 1h 06m 31s","accounts_in_message":[],"_revision_number":2},{"id":"0ca6283c1d3e1c684797dd3ab325cb0fe263ee81","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 07:40:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"69801f268f279ed02bdd1ca2134d24d045e51539","revisions":{"2c43670e37b9f0cec10074f73499e7d75247c2b5":{"kind":"REWORK","_number":1,"created":"2024-06-24 15:25:22.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/34/922634/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-collections-openstack","ref":"refs/changes/34/922634/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/34/922634/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/34/922634/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/34/922634/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/34/922634/1"}}},"commit":{"parents":[{"commit":"4c186a2ae6bfdc96b62de9acd0e32ce5df27ac68","subject":"Add support for creation of the default external networks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/4c186a2ae6bfdc96b62de9acd0e32ce5df27ac68"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-06-24 15:23:37.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-06-24 15:23:37.000000000","tz":120},"subject":"Wait for deleted server to disappear from results","message":"Wait for deleted server to disappear from results\n\nWhen we delete server wait for it to completely disappear from the\nresults (Nova returns it for some time with the \u0027DELETED\u0027 state). Since\ntests (and actually also users) not able to really cope with this wait\nfor server to be gone completely.\n\nChange-Id: Ie2dde98ae47dd7108d554495d5025df175647d5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/2c43670e37b9f0cec10074f73499e7d75247c2b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/2c43670e37b9f0cec10074f73499e7d75247c2b5"}]},"branch":"refs/heads/master"},"69801f268f279ed02bdd1ca2134d24d045e51539":{"kind":"REWORK","_number":2,"created":"2024-06-24 15:56:34.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/34/922634/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-collections-openstack","ref":"refs/changes/34/922634/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/34/922634/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/34/922634/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/34/922634/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/34/922634/2"}}},"commit":{"parents":[{"commit":"4c186a2ae6bfdc96b62de9acd0e32ce5df27ac68","subject":"Add support for creation of the default external networks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/4c186a2ae6bfdc96b62de9acd0e32ce5df27ac68"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-06-24 15:23:37.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-06-24 15:56:27.000000000","tz":120},"subject":"Wait for deleted server to disappear from results","message":"Wait for deleted server to disappear from results\n\nWhen we delete server wait for it to completely disappear from the\nresults (Nova returns it for some time with the \u0027DELETED\u0027 state). Since\ntests (and actually also users) not able to really cope with this wait\nfor server to be gone completely.\n\nChange-Id: Ie2dde98ae47dd7108d554495d5025df175647d5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/69801f268f279ed02bdd1ca2134d24d045e51539"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/69801f268f279ed02bdd1ca2134d24d045e51539"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
