)]}'
{"id":"openstack%2Fansible-collections-openstack~952752","triplet_id":"openstack%2Fansible-collections-openstack~master~I9b1636d778e7477191bfbe6094bd6bc70a4d384e","project":"openstack/ansible-collections-openstack","branch":"master","topic":"remove-mac-requirement","attention_set":{},"removed_from_attention_set":{"5890":{"account":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"last_update":"2026-07-24 01:11:22.000000000","reason":"\u003cGERRIT_ACCOUNT_5890\u003e replied on the change","reason_account":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"}},"17669":{"account":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"last_update":"2026-07-24 08:20:19.000000000","reason":"Change was submitted"},"24245":{"account":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"last_update":"2026-07-23 17:23:50.000000000","reason":"\u003cGERRIT_ACCOUNT_24245\u003e replied on the change","reason_account":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"}},"36238":{"account":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"last_update":"2026-07-24 08:20:19.000000000","reason":"Change was submitted"},"782":{"account":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"last_update":"2026-07-24 08:20:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9b1636d778e7477191bfbe6094bd6bc70a4d384e","subject":"Remove MAC address requirement for node registration","status":"MERGED","created":"2025-06-17 11:58:14.000000000","updated":"2026-07-24 08:20:19.000000000","submitted":"2026-07-24 08:20:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":18,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"952752-remove-mac-requirement","meta_rev_id":"38500cf6f825e0485a0cd23dfcea99fc74343509","_number":952752,"virtual_id_number":952752,"owner":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-24 08:20:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"all":[{"value":2,"date":"2026-07-24 01:11:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":1,"date":"2026-07-23 17:23:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2026-07-24 07:24:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2026-07-24 07:24:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"REVIEWER":[{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-17 12:58:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-20 15:51:25.000000000","updated_by":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"CC"},{"updated":"2025-09-08 08:54:42.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2025-11-10 22:15:01.000000000","updated_by":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"reviewer":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"state":"REVIEWER"},{"updated":"2025-11-12 06:18:12.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"state":"REVIEWER"},{"updated":"2026-07-17 15:36:42.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2026-07-24 07:24:34.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"bc6629e83ebb6e65613d95f19a4a54bd4c09d81a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"date":"2025-06-17 11:58:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bc3290b014b4537be7deeb25411aada7ac91fe44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-17 12:58:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/48b6ca23a8094cc2aaba921ae9530d06\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/2011d873e01d440d86f9bfbbd6e970ff : SUCCESS in 2m 23s\n- openstack-tox-linters-ansible-devel https://zuul.opendev.org/t/openstack/build/e5a9e2dccaf8446c94485b4af79fb882 : FAILURE in 4m 22s (non-voting)\n- openstack-tox-linters-ansible-2.18 https://zuul.opendev.org/t/openstack/build/c50774d6a423404b8eab74487f81f890 : FAILURE in 3m 59s\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/0cb1886f3d2c471eae4a9fac53a74088 : SUCCESS in 56m 25s\n- ansible-collections-openstack-functional-devstack-releases https://zuul.opendev.org/t/openstack/build/0ffa96cbc5414d6e913513da8a8ef931 : SUCCESS in 56m 25s\n- ansible-collections-openstack-functional-devstack-ansible-2.18 https://zuul.opendev.org/t/openstack/build/ec878435c7734128ac131e5e884fd435 : SUCCESS in 59m 34s\n- ansible-collections-openstack-functional-devstack-ansible-devel https://zuul.opendev.org/t/openstack/build/345408e498574db792256adf81963cc7 : SUCCESS in 56m 48s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/fb748a2b5f414f08932ced35ee1b03d8 : SUCCESS in 48m 51s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/88e8937977f8410cb5ac921ba83bc586 : SUCCESS in 34m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"057e8344fd4709b5018a66d3880a50b70e79d311","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2025-06-20 15:51:25.000000000","message":"Patch Set 1:\n\n(6 comments)","accounts_in_message":[],"_revision_number":1},{"id":"8c73560d70eee5075a3fbc8721a20996dfb6922d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"date":"2025-11-01 15:41:19.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"983451a87508d378809467ba11a1cc1b575a3fba","author":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"date":"2025-11-01 15:41:55.000000000","message":"Patch Set 2:\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"03e8c5e6ea678387ed584256864bce09b6ce8739","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"date":"2025-11-01 16:06:05.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"3671ba0d9e7f0d473251c66b1c254eec6eaae501","author":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"date":"2025-11-01 16:08:00.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"83ec2aed8544ab34dfbfcc287ea0987822e8e994","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-01 17:26:36.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ed25db6f15af40a0979b715063d30839\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/a4974853e8874592b2c81793aadfb155 : SUCCESS in 2m 58s\n- openstack-tox-linters-ansible-devel https://zuul.opendev.org/t/openstack/build/640ba0a812264eb59f579864c6ee7885 : FAILURE in 5m 54s (non-voting)\n- openstack-tox-linters-ansible-2.18 https://zuul.opendev.org/t/openstack/build/994feb5ba72e4f029e977b7dc03639f6 : FAILURE in 3m 39s\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/b7bdf46729d74e3ca262591e0d784a2b : FAILURE in 1h 12m 55s\n- ansible-collections-openstack-functional-devstack-releases https://zuul.opendev.org/t/openstack/build/77ca3e703c4242aebc9d2458aacf3e3a : SUCCESS in 1h 15m 27s\n- ansible-collections-openstack-functional-devstack-ansible-2.18 https://zuul.opendev.org/t/openstack/build/54e18b820b00480aa48d7d6f57f128a0 : SUCCESS in 56m 31s\n- ansible-collections-openstack-functional-devstack-ansible-devel https://zuul.opendev.org/t/openstack/build/7e60ae72f91c45b394413e9dcc16707d : FAILURE in 1h 14m 57s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/5189d1d6c3b34d1f880f62ffdb42fb2a : SUCCESS in 29m 18s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/7f7ecdcaa0fb44a0bbd59e59e02c195a : SUCCESS in 31m 07s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"70e7731e0201b3ed3a202e5a167aa1459964f8c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"date":"2025-11-03 09:44:09.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"e304fd6386386701d33d986dae089be8589da10e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-03 11:01:34.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c89e914e5ef24dffa8251ab547308f57\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/0495c401814942ef832090c3bf4ce169 : SUCCESS in 2m 56s\n- openstack-tox-linters-ansible-devel https://zuul.opendev.org/t/openstack/build/a6149cee7f6d43618b36957cc38be24b : SUCCESS in 5m 44s (non-voting)\n- openstack-tox-linters-ansible-2.18 https://zuul.opendev.org/t/openstack/build/34797809df914ab79c80b4b2448bfdf9 : SUCCESS in 5m 50s\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/25e7d61bf4f84d578bba0eb05436c930 : SUCCESS in 59m 01s\n- ansible-collections-openstack-functional-devstack-releases https://zuul.opendev.org/t/openstack/build/bbd79ca447354d93b6efe8a2471d9b02 : SUCCESS in 1h 12m 02s\n- ansible-collections-openstack-functional-devstack-ansible-2.18 https://zuul.opendev.org/t/openstack/build/01c4c5817fde40b090f7e418a71d2c15 : SUCCESS in 41m 40s\n- ansible-collections-openstack-functional-devstack-ansible-devel https://zuul.opendev.org/t/openstack/build/08398e9cbb3e4b4bb142aeee419eb771 : FAILURE in 1h 01m 27s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/5a9ffbf52efb4c59bb85477e070daa3f : SUCCESS in 40m 03s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/c0d0e7ba9c934bd1b3f6f3a54e7e4b07 : SUCCESS in 40m 36s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"847033b3162e99169f88abb9cd8606c15420a25b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"date":"2025-11-03 14:30:29.000000000","message":"Uploaded patch set 5: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"cb3ad781236b50d765947e25ef0c78a8096ad4e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-03 15:51:23.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b504b6f0ac6e4ed8859bbe115bdeaa35\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/d18778324d8b4cd591b8fcaa9ea2e777 : SUCCESS in 3m 10s\n- openstack-tox-linters-ansible-devel https://zuul.opendev.org/t/openstack/build/479af98f120c492db65511cf6567ea5a : SUCCESS in 3m 58s (non-voting)\n- openstack-tox-linters-ansible-2.18 https://zuul.opendev.org/t/openstack/build/36afaa68b8784d09af72ae1ad071c5de : SUCCESS in 6m 14s\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/3963d792beef4486b6f90f29e4cbc114 : SUCCESS in 42m 00s\n- ansible-collections-openstack-functional-devstack-releases https://zuul.opendev.org/t/openstack/build/0cee68bef92348e6b73274518bd6ae74 : SUCCESS in 1h 15m 38s\n- ansible-collections-openstack-functional-devstack-ansible-2.18 https://zuul.opendev.org/t/openstack/build/762ff1f689254649939b3d701e36fecc : SUCCESS in 36m 12s\n- ansible-collections-openstack-functional-devstack-ansible-devel https://zuul.opendev.org/t/openstack/build/70865cc5c2dd4e32bbbc2076c878d6cb : SUCCESS in 1h 08m 59s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/cc09046ea2c34cc29daed270aec6be90 : SUCCESS in 34m 35s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/59a1006dfdca41eaba969f66d177d778 : SUCCESS in 19m 24s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"5e0fdb3be5e119712ae878063be6e45397e5be3a","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2025-11-10 22:15:01.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"51c721af8df75bea658c599a0583df1c79840f05","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2025-11-12 06:18:12.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"17e883a199029426b4516cab48583e95547e82ab","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-07-17 15:36:42.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"107186cd2a5454b8811e96d2d8b61e8341fc39bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-07-17 16:25:12.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"b3f8b31ac00a6663215ad788286632bd668f32f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-17 17:59:56.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ff00b9d9de8d4155bfc23be43f51eb4a\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/3901c68b407c468fb84251bffa76da12 : SUCCESS in 3m 34s\n- openstack-tox-linters-ansible-devel https://zuul.opendev.org/t/openstack/build/6c51d3bf9be540b2bf5ff5f5a1d2c453 : FAILURE in 7m 12s (non-voting)\n- openstack-tox-linters-ansible-2.18 https://zuul.opendev.org/t/openstack/build/02c05561ff874660bcab1531997b55fe : FAILURE in 6m 25s\n- openstack-tox-linters-ansible-2.19 https://zuul.opendev.org/t/openstack/build/760854e24b1f445c8e6c0b0144df4a57 : FAILURE in 7m 39s\n- openstack-tox-linters-ansible-2.20 https://zuul.opendev.org/t/openstack/build/363e0bf8e25148efb5344dc84922f378 : FAILURE in 6m 28s\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/5d2c41b7b4594516aee720eee1b4ed22 : SUCCESS in 1h 20m 21s\n- ansible-collections-openstack-functional-devstack-releases https://zuul.opendev.org/t/openstack/build/6fc9567aa3464bb4bf6bcf2222bb3e72 : SUCCESS in 1h 14m 52s\n- ansible-collections-openstack-functional-devstack-ansible-2.18 https://zuul.opendev.org/t/openstack/build/c60c8c3008b34b1aa4008a77a01ae837 : SUCCESS in 1h 15m 20s\n- ansible-collections-openstack-functional-devstack-ansible-2.19 https://zuul.opendev.org/t/openstack/build/9fa45a67d79a44a2933b91fe341d12c3 : SUCCESS in 1h 12m 43s\n- ansible-collections-openstack-functional-devstack-ansible-2.20 https://zuul.opendev.org/t/openstack/build/e23fbaf6a84440df951257484e55830e : SUCCESS in 1h 01m 49s\n- ansible-collections-openstack-functional-devstack-ansible-devel https://zuul.opendev.org/t/openstack/build/0afe0f4c114a4afe8cec427364e1b328 : POST_FAILURE in 13m 15s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/abcf20a74943472db705d7d4d0ff57a4 : SUCCESS in 48m 31s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/f1eaf467c64840d28859d3167cda71ff : SUCCESS in 39m 20s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f5bbb8c93e7a55598197928c95821c761f509ced","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-07-23 15:39:30.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"8df45e32282e1264c33575b2ce3d37ebcfe1c875","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-07-23 15:40:32.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"d7f85ab39813b147eee5a59fdbbfb04a1e1552ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-23 17:01:20.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da5c94423bea46d48c0e0bdb49e33240\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/d9f60d660a0d4523942e582861610a9f : SUCCESS in 3m 15s\n- openstack-tox-linters-ansible-devel https://zuul.opendev.org/t/openstack/build/581894d1a7564d988a6164043afdc2af : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-linters-ansible-2.18 https://zuul.opendev.org/t/openstack/build/071a1dce286146198119fca1d8997e3a : SUCCESS in 6m 43s\n- openstack-tox-linters-ansible-2.19 https://zuul.opendev.org/t/openstack/build/ac0a6748fc634427a3b3f1c1a1e3ea34 : SUCCESS in 6m 35s\n- openstack-tox-linters-ansible-2.20 https://zuul.opendev.org/t/openstack/build/be160455d2d446659fdcdf3b588a063a : SUCCESS in 6m 36s\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/d76af0e6c5214f1e86aedf45415a54e6 : SUCCESS in 1h 15m 17s\n- ansible-collections-openstack-functional-devstack-releases https://zuul.opendev.org/t/openstack/build/9e7b9584dc1b4006b20b6c41c623b92c : SUCCESS in 1h 03m 17s\n- ansible-collections-openstack-functional-devstack-ansible-2.18 https://zuul.opendev.org/t/openstack/build/f6485793b457410184a229bea6774198 : SUCCESS in 1h 10m 56s\n- ansible-collections-openstack-functional-devstack-ansible-2.19 https://zuul.opendev.org/t/openstack/build/c9a0c7068d6f4f989b99f51637aaaf7a : SUCCESS in 1h 15m 28s\n- ansible-collections-openstack-functional-devstack-ansible-2.20 https://zuul.opendev.org/t/openstack/build/9e3c9ef45de34deb98a662f014990bb7 : SUCCESS in 57m 30s\n- ansible-collections-openstack-functional-devstack-ansible-devel https://zuul.opendev.org/t/openstack/build/5cf1850d047d4d409788a095e8751a57 : POST_FAILURE in 13m 48s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/fca1cd40449544f8876e6c12ede7020d : SUCCESS in 50m 29s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/4b746c14a8a74f7498be243b0c1c47a5 : SUCCESS in 33m 46s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"b78b87bea2473818bbc8291e62ff059b9d82bfb3","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2026-07-23 17:23:50.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"c7502672e20a60f47881567114875ce3fbf8c0d5","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-07-24 01:11:22.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"097d697eb32185344ef96663c4568b0f2df8926c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-24 07:24:34.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"04a585b80cb5dd25c4ce7f0da12a8b67e2fdf8b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-24 07:25:07.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"7184c5112dfea59064c331cab2b6a091014f9ab9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-24 08:20:19.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0dfb5cec15d343d38278a983758bc8d4\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/7a2eda2b3a434280829fdfce525c0e38 : SUCCESS in 3m 22s\n- openstack-tox-linters-ansible-2.18 https://zuul.opendev.org/t/openstack/build/ea2a283abf8d4cc9a5800f6e01f6516b : SUCCESS in 5m 09s\n- openstack-tox-linters-ansible-2.19 https://zuul.opendev.org/t/openstack/build/9e18bfe9815d48f5a1d3fdcdf1326d0b : SUCCESS in 5m 24s\n- openstack-tox-linters-ansible-2.20 https://zuul.opendev.org/t/openstack/build/9fe7d2acc08b4fdc9f668569be5984d4 : SUCCESS in 3m 23s\n- ansible-collections-openstack-functional-devstack-releases https://zuul.opendev.org/t/openstack/build/29b01e369bcb4825aa3c56ce5cf56573 : SUCCESS in 54m 50s","accounts_in_message":[],"_revision_number":7},{"id":"38500cf6f825e0485a0cd23dfcea99fc74343509","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-24 08:20:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"ee88cc15b87221ecabb474c7c3d0feaf6e01a7fa","revisions":{"fc3965745f4f6e9e6fd7e49d4dab7d149b9da9d2":{"kind":"REWORK","_number":1,"created":"2025-06-17 11:58:14.000000000","uploader":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"ref":"refs/changes/52/952752/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-collections-openstack","ref":"refs/changes/52/952752/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/52/952752/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/52/952752/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/52/952752/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/52/952752/1"}}},"commit":{"parents":[{"commit":"f584c54dfd03e81c1e9c30e2d22f19d2d17a4353","subject":"Merge \"Add volume_manage module\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/f584c54dfd03e81c1e9c30e2d22f19d2d17a4353"}]}],"author":{"name":"Jake Hutchinson","email":"jake@stackhpc.com","date":"2025-06-17 11:14:22.000000000","tz":60},"committer":{"name":"Jake Hutchinson","email":"jake@stackhpc.com","date":"2025-06-17 11:58:13.000000000","tz":60},"subject":"Remove MAC address requirement for node registration","message":"Remove MAC address requirement for node registration\n\nRemove NICs/MAC address requirement to allow node\nregistration from the node UUID or name and allow the\nsetting of the Ironic provision state.\n\nChange-Id: I9b1636d778e7477191bfbe6094bd6bc70a4d384e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/fc3965745f4f6e9e6fd7e49d4dab7d149b9da9d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/fc3965745f4f6e9e6fd7e49d4dab7d149b9da9d2"}]},"branch":"refs/heads/master"},"6bf261a2d97a2845180a87fe39ea488e0f0f732c":{"kind":"REWORK","_number":2,"created":"2025-11-01 15:41:19.000000000","uploader":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"ref":"refs/changes/52/952752/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-collections-openstack","ref":"refs/changes/52/952752/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/52/952752/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/52/952752/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/52/952752/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/52/952752/2"}}},"commit":{"parents":[{"commit":"f584c54dfd03e81c1e9c30e2d22f19d2d17a4353","subject":"Merge \"Add volume_manage module\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/f584c54dfd03e81c1e9c30e2d22f19d2d17a4353"}]}],"author":{"name":"Hollie Hutchinson","email":"hollie@stackhpc.com","date":"2025-11-01 15:19:37.000000000","tz":0},"committer":{"name":"Hollie Hutchinson","email":"hollie@stackhpc.com","date":"2025-11-01 15:41:17.000000000","tz":0},"subject":"Remove MAC address requirement for node registration","message":"Remove MAC address requirement for node registration\n\nRemove NICs/MAC address requirement to allow node\nregistration from the node UUID or name and allow the\nsetting of the Ironic provision state.\n\nChange-Id: I9b1636d778e7477191bfbe6094bd6bc70a4d384e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/6bf261a2d97a2845180a87fe39ea488e0f0f732c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/6bf261a2d97a2845180a87fe39ea488e0f0f732c"}]},"branch":"refs/heads/master"},"1c4b70ed7eb51128494eb390cd2651d0d8eb87b6":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-11-01 16:06:05.000000000","uploader":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"ref":"refs/changes/52/952752/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-collections-openstack","ref":"refs/changes/52/952752/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/52/952752/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/52/952752/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/52/952752/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/52/952752/3"}}},"commit":{"parents":[{"commit":"f584c54dfd03e81c1e9c30e2d22f19d2d17a4353","subject":"Merge \"Add volume_manage module\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/f584c54dfd03e81c1e9c30e2d22f19d2d17a4353"}]}],"author":{"name":"Hollie Hutchinson","email":"hollie@stackhpc.com","date":"2025-11-01 15:19:37.000000000","tz":0},"committer":{"name":"Hollie Hutchinson","email":"hollie@stackhpc.com","date":"2025-11-01 16:00:49.000000000","tz":0},"subject":"Remove MAC address requirement for node registration","message":"Remove MAC address requirement for node registration\n\nThis module currently does not allow for the creation of baremetal\nnodes without including the NICs. We\u0027re proposing using this\nmodule for enrolling nodes and then later running an out-of-band\nRedfish inspection to get the NICs after enrolling the node in\nIronic.\n\nThe default option of the module will still be to move the node\nto available state which will still require the NICs, however\nan option `make_available_on_node_create` is provided to allow\nleaving the node in enroll state.\n\nChange-Id: I9b1636d778e7477191bfbe6094bd6bc70a4d384e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/1c4b70ed7eb51128494eb390cd2651d0d8eb87b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/1c4b70ed7eb51128494eb390cd2651d0d8eb87b6"}]},"branch":"refs/heads/master"},"af6574ff8c712ae8bb4355f88b0eeb7cda8f60f1":{"kind":"REWORK","_number":4,"created":"2025-11-03 09:44:09.000000000","uploader":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"ref":"refs/changes/52/952752/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-collections-openstack","ref":"refs/changes/52/952752/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/52/952752/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/52/952752/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/52/952752/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/52/952752/4"}}},"commit":{"parents":[{"commit":"f584c54dfd03e81c1e9c30e2d22f19d2d17a4353","subject":"Merge \"Add volume_manage module\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/f584c54dfd03e81c1e9c30e2d22f19d2d17a4353"}]}],"author":{"name":"Hollie Hutchinson","email":"hollie@stackhpc.com","date":"2025-11-01 15:19:37.000000000","tz":0},"committer":{"name":"Hollie Hutchinson","email":"hollie@stackhpc.com","date":"2025-11-03 09:44:01.000000000","tz":0},"subject":"Remove MAC address requirement for node registration","message":"Remove MAC address requirement for node registration\n\nThis module currently does not allow for the creation of baremetal\nnodes without including the NICs. We\u0027re proposing using this\nmodule for enrolling nodes and then later running an out-of-band\nRedfish inspection to get the NICs after enrolling the node in\nIronic.\n\nThe default option of the module will still be to move the node\nto available state which will still require the NICs, however\nan option `make_available_on_node_create` is provided to allow\nleaving the node in enroll state.\n\nChange-Id: I9b1636d778e7477191bfbe6094bd6bc70a4d384e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/af6574ff8c712ae8bb4355f88b0eeb7cda8f60f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/af6574ff8c712ae8bb4355f88b0eeb7cda8f60f1"}]},"branch":"refs/heads/master"},"c1a25818d8cbad4f54e47ec9d023702e61e95c89":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2025-11-03 14:30:29.000000000","uploader":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"ref":"refs/changes/52/952752/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-collections-openstack","ref":"refs/changes/52/952752/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/52/952752/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/52/952752/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/52/952752/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/52/952752/5"}}},"commit":{"parents":[{"commit":"f584c54dfd03e81c1e9c30e2d22f19d2d17a4353","subject":"Merge \"Add volume_manage module\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/f584c54dfd03e81c1e9c30e2d22f19d2d17a4353"}]}],"author":{"name":"Hollie Hutchinson","email":"hollie@stackhpc.com","date":"2025-11-01 15:19:37.000000000","tz":0},"committer":{"name":"Hollie Hutchinson","email":"hollie@stackhpc.com","date":"2025-11-03 14:29:49.000000000","tz":0},"subject":"Remove MAC address requirement for node registration","message":"Remove MAC address requirement for node registration\n\nThis module currently does not allow for the creation of baremetal\nnodes without including the NICs. We\u0027re proposing using this\nmodule for enrolling nodes and then later running an out-of-band\nRedfish inspection to get the NICs after enrolling the node in\nIronic.\n\nThe default option of the module will still be to move the node\nto available state which will still require the NICs, however\nan option `make_available_on_node_create` is provided to allow\nleaving the node in enroll state.\n\nDepends-On: https://review.opendev.org/c/openstack/openstacksdk/+/965936\nChange-Id: I9b1636d778e7477191bfbe6094bd6bc70a4d384e\nSigned-off-by: Hollie Hutchinson \u003chollie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/c1a25818d8cbad4f54e47ec9d023702e61e95c89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/c1a25818d8cbad4f54e47ec9d023702e61e95c89"}]},"branch":"refs/heads/master"},"7c6b1be36093d209504983f7925c1b348005b3aa":{"kind":"REWORK","_number":6,"created":"2026-07-17 16:25:12.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/52/952752/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-collections-openstack","ref":"refs/changes/52/952752/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/52/952752/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/52/952752/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/52/952752/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/52/952752/6"}}},"commit":{"parents":[{"commit":"209cf2d60533d5094aaf0308765d2e538091f203","subject":"Merge \"Revive collection unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/209cf2d60533d5094aaf0308765d2e538091f203"}]}],"author":{"name":"Hollie Hutchinson","email":"hollie@stackhpc.com","date":"2025-11-01 15:19:37.000000000","tz":0},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2026-07-17 16:06:46.000000000","tz":60},"subject":"Remove MAC address requirement for node registration","message":"Remove MAC address requirement for node registration\n\nAllow the creation of baremetal nodes without specifying any NICs.\nIn this case switch the target provision state from available to\nenroll.\n\nWe\u0027re proposing this patch to support enrolling nodes as a first\nstep before running an out-of-band Redfish inspection to get the\nNICs.\n\nNo change is made to existing workflows.\n\nChange-Id: I9b1636d778e7477191bfbe6094bd6bc70a4d384e\nSigned-off-by: Hollie Hutchinson \u003chollie@stackhpc.com\u003e\nSigned-off-by: Doug Szumski \u003cdoug@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/7c6b1be36093d209504983f7925c1b348005b3aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/7c6b1be36093d209504983f7925c1b348005b3aa"}]},"branch":"refs/heads/master"},"ee88cc15b87221ecabb474c7c3d0feaf6e01a7fa":{"kind":"REWORK","_number":7,"created":"2026-07-23 15:39:30.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/52/952752/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-collections-openstack","ref":"refs/changes/52/952752/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/52/952752/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/52/952752/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/52/952752/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/52/952752/7"}}},"commit":{"parents":[{"commit":"209cf2d60533d5094aaf0308765d2e538091f203","subject":"Merge \"Revive collection unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/209cf2d60533d5094aaf0308765d2e538091f203"}]}],"author":{"name":"Hollie Hutchinson","email":"hollie@stackhpc.com","date":"2025-11-01 15:19:37.000000000","tz":0},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2026-07-23 15:36:27.000000000","tz":60},"subject":"Remove MAC address requirement for node registration","message":"Remove MAC address requirement for node registration\n\nAllow the creation of baremetal nodes without specifying any NICs.\nIn this case switch the target provision state from available to\nenroll.\n\nWe\u0027re proposing this patch to support enrolling nodes as a first\nstep before running an out-of-band Redfish inspection to get the\nNICs.\n\nNo change is made to existing workflows.\n\nChange-Id: I9b1636d778e7477191bfbe6094bd6bc70a4d384e\nSigned-off-by: Hollie Hutchinson \u003chollie@stackhpc.com\u003e\nSigned-off-by: Doug Szumski \u003cdoug@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/ee88cc15b87221ecabb474c7c3d0feaf6e01a7fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/ee88cc15b87221ecabb474c7c3d0feaf6e01a7fa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
