)]}'
{"id":"openstack%2Fansible-collections-openstack~964684","triplet_id":"openstack%2Fansible-collections-openstack~master~Ib3c63131f9ae09abd03fc7d619e3566267690216","project":"openstack/ansible-collections-openstack","branch":"master","attention_set":{"35345":{"account":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"last_update":"2026-09-23 08:33:20.000000000","reason":"Vote got outdated and was removed: Code-Review+1"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-09-23 08:33:20.000000000","reason":"Vote got outdated and was removed: Code-Review+1"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-09-23 08:33:20.000000000","reason":"Vote got outdated and was removed: Code-Review+2"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2026-09-23 08:33:20.000000000","reason":"Vote got outdated and was removed: Code-Review+1"},"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2025-11-18 13:22:22.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2026-09-23 10:06:41.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{"2":{"account":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"last_update":"2026-06-03 15:43:37.000000000","reason":"Reviewer/Cc was removed"}},"hashtags":[],"change_id":"Ib3c63131f9ae09abd03fc7d619e3566267690216","subject":"Fix: Include target_project_id in RBAC creation","status":"NEW","created":"2025-10-23 11:00:06.000000000","updated":"2026-09-23 10:06:41.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"76000067066b197ac00ca4e41d620be5bd2a82be","_number":964684,"virtual_id_number":964684,"owner":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-09-23 10:06:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-23 12:19:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-24 11:09:47.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2025-11-18 13:22:22.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2025-11-18 13:23:53.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-12-05 10:52:42.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2026-03-18 10:36:13.000000000","updated_by":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"reviewer":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"state":"REVIEWER"},{"updated":"2026-03-18 10:36:19.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2026-05-27 08:22:31.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"},{"updated":"2026-06-03 15:43:37.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REMOVED"},{"updated":"2026-09-08 12:11:41.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"896850f4f3533722a75ebc3260a431003c1cc9f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-10-23 11:00:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7ce67b6ce7a9bc8e6f9aea5c44d926c352bd5190","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-10-23 11:03:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ce732ee985cc8ee02c8d5d8773890f354f1b5f4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 12:19:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7894193b245a42e79bace129b02270bd\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/1a90a4de53764e92864271aa93115a3a : SUCCESS in 2m 57s\n- openstack-tox-linters-ansible-devel https://zuul.opendev.org/t/openstack/build/2b294dd2ae4f4f2596b57a4d7089ca52 : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-linters-ansible-2.18 https://zuul.opendev.org/t/openstack/build/e19e64c2780e461281cc91111f78bb4b : SUCCESS in 5m 44s\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/8f20338991494e1d8b36d78f707db9e6 : SUCCESS in 1h 14m 21s\n- ansible-collections-openstack-functional-devstack-releases https://zuul.opendev.org/t/openstack/build/1e509dc3ccdd44179d60884eaeb27584 : SUCCESS in 59m 29s\n- ansible-collections-openstack-functional-devstack-ansible-2.18 https://zuul.opendev.org/t/openstack/build/a10eb5d3c81541b29908e75d86f88318 : SUCCESS in 1h 09m 26s\n- ansible-collections-openstack-functional-devstack-ansible-devel https://zuul.opendev.org/t/openstack/build/f937aca0d05a40b9890156689fc2377a : FAILURE in 1h 10m 01s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/ef8265188229480fa2b5a0dd3461c06c : SUCCESS in 40m 35s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/5cfda4b4c6834c76aa891e4f79a54306 : SUCCESS in 20m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c65a009df5636ab6652d15936b5fa8c365e73442","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-11-18 13:23:53.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"85faebb87c9fc99d3fbd0e45a836985d39c81fcf","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2026-03-18 10:36:13.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"212b99e989a06da79ea829cc71885a8dfd550986","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-18 10:36:19.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"b2d89b7a10db5bd6589ab55f43c624cdf2f577e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"real_author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-03-19 09:37:26.000000000","message":"Patch Set 3: Patch Set 2 was rebased on behalf of \u003cGERRIT_ACCOUNT_35263\u003e\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"_revision_number":3},{"id":"20479dbf2acff70730e39897eb7ac0ce8aa2c961","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 10:58:51.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7a35bc4c64a34ed88407a9721612838e\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/9f4d8559b4ce4c229eb8eba0c5d55b05 : SUCCESS in 3m 18s\n- openstack-tox-linters-ansible-devel https://zuul.opendev.org/t/openstack/build/894f20b989b2430e826cb6278558f7c0 : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-linters-ansible-2.18 https://zuul.opendev.org/t/openstack/build/91962a021ba44961a2e1a9b0acc36a1f : SUCCESS in 7m 14s\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/438e8cae09534ad9b47ebff6218cf8f8 : FAILURE in 29m 10s\n- ansible-collections-openstack-functional-devstack-releases https://zuul.opendev.org/t/openstack/build/3c38d519ffb94072a8784a6be3d43412 : SUCCESS in 1h 18m 29s\n- ansible-collections-openstack-functional-devstack-ansible-2.18 https://zuul.opendev.org/t/openstack/build/1dd6617cd8fd4fb5b5ca7bf04df284f1 : SUCCESS in 37m 28s\n- ansible-collections-openstack-functional-devstack-ansible-devel https://zuul.opendev.org/t/openstack/build/1553334f58c541158d7a50c2ccd20802 : POST_FAILURE in 13m 36s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/26da69066ab945ac94c6ea1192a79eb4 : SUCCESS in 46m 59s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/48a2077107df4e22b1daadb87b6d29bc : SUCCESS in 49m 51s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ce8e8ee44801db89e4b9943ceddbbab5baf508e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-05-06 11:03:24.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"fc59969f779cc7e4937cc2faebb92336a6a9f5b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 12:27:09.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b8adbfe22514445b87b52008250e2b3d\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/e6477db49a0d46dca2319a1699237b04 : SUCCESS in 4m 24s\n- openstack-tox-linters-ansible-devel https://zuul.opendev.org/t/openstack/build/23e3182cec6748a78b80a4c38db91b0a : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-linters-ansible-2.18 https://zuul.opendev.org/t/openstack/build/62c78c29b33c4e51ba14726af5020384 : SUCCESS in 6m 39s\n- openstack-tox-linters-ansible-2.19 https://zuul.opendev.org/t/openstack/build/3f5cc7dd85054783a36f1fc1907f351b : SUCCESS in 6m 18s\n- openstack-tox-linters-ansible-2.20 https://zuul.opendev.org/t/openstack/build/3757f7454c5245119cb761425200f3d0 : SUCCESS in 6m 23s\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/3a3689729e4241488c82284e95fd0ad7 : SUCCESS in 40m 04s\n- ansible-collections-openstack-functional-devstack-releases https://zuul.opendev.org/t/openstack/build/877619a8fea04e52af97e511413e336f : FAILURE in 1h 08m 43s\n- ansible-collections-openstack-functional-devstack-ansible-2.18 https://zuul.opendev.org/t/openstack/build/f1c64087b0684585a9ce673655db8baf : SUCCESS in 40m 01s\n- ansible-collections-openstack-functional-devstack-ansible-2.19 https://zuul.opendev.org/t/openstack/build/78e5437166884253bbdd5d9dab93f5f8 : SUCCESS in 1h 15m 42s\n- ansible-collections-openstack-functional-devstack-ansible-2.20 https://zuul.opendev.org/t/openstack/build/fca90755b49f47078e7c85186f311436 : SUCCESS in 1h 03m 06s\n- ansible-collections-openstack-functional-devstack-ansible-devel https://zuul.opendev.org/t/openstack/build/7d602057837d41bb9add741105b4566d : POST_FAILURE in 13m 04s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/50bf4631eca64298a30e0ddf544a7bf1 : SUCCESS in 50m 18s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/5889bbfa3bcd495dacd3dfa9a1f25c0b : SUCCESS in 28m 59s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"51243e77b34c04896ddc0012c054340f8bc12374","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"real_author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-27 08:21:16.000000000","message":"Patch Set 5: Patch Set 4 was rebased on behalf of \u003cGERRIT_ACCOUNT_35263\u003e\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"_revision_number":5},{"id":"07bbe5c912dd3a23a7f43b6aa13d06ea861631c1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-27 08:22:31.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a0dea3412c15c396abeb152502355acf6f804234","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 09:48:30.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb6a6f73bcbe400290597fa4847d9967\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/b46427a7446e4cd3a0f89472fc0c22a6 : SUCCESS in 3m 05s\n- openstack-tox-linters-ansible-devel https://zuul.opendev.org/t/openstack/build/074858a2e1624a868eb2f0329a206c82 : SUCCESS in 6m 17s (non-voting)\n- openstack-tox-linters-ansible-2.18 https://zuul.opendev.org/t/openstack/build/e26ed383a5a142dfb5f450f21a837174 : SUCCESS in 3m 14s\n- openstack-tox-linters-ansible-2.19 https://zuul.opendev.org/t/openstack/build/644f3dc1633740b1a5dbb7fbf8f1999d : SUCCESS in 3m 55s\n- openstack-tox-linters-ansible-2.20 https://zuul.opendev.org/t/openstack/build/ed916e0d96b7403f90405ef23b0c51cc : SUCCESS in 5m 29s\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/1810a24b474543c183d38ab27ee2ffdd : SUCCESS in 1h 01m 59s\n- ansible-collections-openstack-functional-devstack-releases https://zuul.opendev.org/t/openstack/build/3cd068b0549d4f6c98bc5b404e56db51 : SUCCESS in 40m 33s\n- ansible-collections-openstack-functional-devstack-ansible-2.18 https://zuul.opendev.org/t/openstack/build/3531ff9b0a934dd9a76de3e369e50c21 : SUCCESS in 46m 11s\n- ansible-collections-openstack-functional-devstack-ansible-2.19 https://zuul.opendev.org/t/openstack/build/b432aa31448f4ee087bed83e8db9d543 : SUCCESS in 1h 24m 20s\n- ansible-collections-openstack-functional-devstack-ansible-2.20 https://zuul.opendev.org/t/openstack/build/c23d2acd1a08483c96103e518d1d98ad : SUCCESS in 1h 08m 11s\n- ansible-collections-openstack-functional-devstack-ansible-devel https://zuul.opendev.org/t/openstack/build/8a4139072d0b418c85d53902c1d4a683 : POST_FAILURE in 6m 30s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/43d2d585628c42a29eb1c6357748f874 : SUCCESS in 26m 18s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/aefe8e58c47343eabc82560e8511d320 : SUCCESS in 48m 58s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"db965b0a280edff5b345cf78e23265376c3712e6","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2026-06-03 15:43:37.000000000","message":"Removed reviewer \u003cGERRIT_ACCOUNT_2\u003e.","accounts_in_message":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"}],"_revision_number":5},{"id":"92c64cad6953ebade21ba94200415b4488a9259f","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-06-15 13:11:24.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"04990718101d00b1865d4effe901d95ac5151acd","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-08 12:11:41.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b78d116cc324d7597164c0b74c3762c4c54033fb","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-08 12:15:51.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"7d43498b56e67396522a55dd1ec756797c391bdd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-09-23 08:33:20.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"43347ab70f6ae4a2435d724d3fe0bb795673a4a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-09-23 08:35:35.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"5b309236971b11d0285a0dece4cf11401fb2abc3","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-09-23 08:36:11.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"76000067066b197ac00ca4e41d620be5bd2a82be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 10:06:41.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7dd62a347c024304b42df5ba3cd6ce0c\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/8247cb96de2846c28547d2f227bf1cfd : SUCCESS in 4m 29s\n- openstack-tox-linters-ansible-devel https://zuul.opendev.org/t/openstack/build/531aba6b2cbd46e497259d15af34e5da : RETRY_LIMIT in 3m 51s (non-voting)\n- openstack-tox-linters-ansible-2.18 https://zuul.opendev.org/t/openstack/build/70ce51d3d3dd4021abc8bc5cd95eabe1 : SUCCESS in 5m 29s\n- openstack-tox-linters-ansible-2.19 https://zuul.opendev.org/t/openstack/build/8f5d63e93f534f0ea51de2fc065ea16b : SUCCESS in 5m 57s\n- openstack-tox-linters-ansible-2.20 https://zuul.opendev.org/t/openstack/build/7ea4fd3a01b4486caa588711432b3a73 : SUCCESS in 4m 44s\n- openstack-tox-linters-ansible-2.21 https://zuul.opendev.org/t/openstack/build/dee9b001d3ec4bc5907941ff001096a0 : SUCCESS in 7m 48s\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/7871c375cb0d4da0975bc7af6870c377 : FAILURE in 44m 56s\n- ansible-collections-openstack-functional-devstack-releases https://zuul.opendev.org/t/openstack/build/78d0427bd2b64754b4bcf84395e22f11 : FAILURE in 38m 52s\n- ansible-collections-openstack-functional-devstack-ansible-2.18 https://zuul.opendev.org/t/openstack/build/091fbe3db17a49a5b11f8d569f8ce0fb : FAILURE in 43m 20s\n- ansible-collections-openstack-functional-devstack-ansible-2.19 https://zuul.opendev.org/t/openstack/build/04373ea6d90b46128780efdd029ebf4a : POST_FAILURE in 16m 09s\n- ansible-collections-openstack-functional-devstack-ansible-2.20 https://zuul.opendev.org/t/openstack/build/aab7609eefea4d769f1a01a5fd5e830a : FAILURE in 32m 05s\n- ansible-collections-openstack-functional-devstack-ansible-2.21 https://zuul.opendev.org/t/openstack/build/a9802ddb4b61403e885ea16fba395d99 : FAILURE in 43m 36s\n- ansible-collections-openstack-functional-devstack-ansible-devel https://zuul.opendev.org/t/openstack/build/bf65998778cf4b27867cec95e98690f0 : POST_FAILURE in 6m 39s (non-voting)\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/a8aec4002f6d4cb4931db9f3eabd7b7b : FAILURE in 10m 32s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/40e87cce85fa4cd58e261bdfd7c39b62 : FAILURE in 13m 38s (non-voting)","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"291d64adaa44617b19a66cbbb250c4079f12fe26","revisions":{"105632cf22e908c4b17b3534e6dac4b146a42f58":{"kind":"REWORK","_number":1,"created":"2025-10-23 11:00:06.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/84/964684/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-collections-openstack","ref":"refs/changes/84/964684/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/84/964684/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/84/964684/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/84/964684/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/84/964684/1"}}},"commit":{"parents":[{"commit":"b1ecd54a5db1263d272b944462bb79e0bc35a954","subject":"feat: introduce share_type modules","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/b1ecd54a5db1263d272b944462bb79e0bc35a954"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-10-23 10:59:14.000000000","tz":60},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-10-23 10:59:14.000000000","tz":60},"subject":"Fix: Include target_project_id in RBAC creation","message":"Fix: Include target_project_id in RBAC creation\n\nThis fixes an issue where we cannot create multiple Neutron RBAC\npolicies using the same source project and network.\n\nCloses-Bug: #2129677\nChange-Id: Ib3c63131f9ae09abd03fc7d619e3566267690216\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/105632cf22e908c4b17b3534e6dac4b146a42f58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/105632cf22e908c4b17b3534e6dac4b146a42f58"}]},"branch":"refs/heads/master"},"81f4be9f42ae363c2b9473198844f59a62fb0dfa":{"kind":"REWORK","_number":2,"created":"2025-10-23 11:03:28.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/84/964684/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-collections-openstack","ref":"refs/changes/84/964684/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/84/964684/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/84/964684/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/84/964684/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/84/964684/2"}}},"commit":{"parents":[{"commit":"b1ecd54a5db1263d272b944462bb79e0bc35a954","subject":"feat: introduce share_type modules","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/b1ecd54a5db1263d272b944462bb79e0bc35a954"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-10-23 10:59:14.000000000","tz":60},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-10-23 11:03:21.000000000","tz":60},"subject":"Fix: Include target_project_id in RBAC creation","message":"Fix: Include target_project_id in RBAC creation\n\nThis fixes an issue where we cannot create multiple Neutron RBAC\npolicies using the same source project and network.\n\nCloses-Bug: #2129677\nChange-Id: Ib3c63131f9ae09abd03fc7d619e3566267690216\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/81f4be9f42ae363c2b9473198844f59a62fb0dfa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/81f4be9f42ae363c2b9473198844f59a62fb0dfa"}]},"branch":"refs/heads/master"},"6a54ca043f9e4350a48a8c78a5a5df6d46bc5073":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-03-19 09:37:26.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"real_uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/84/964684/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-collections-openstack","ref":"refs/changes/84/964684/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/84/964684/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/84/964684/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/84/964684/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/84/964684/3"}}},"commit":{"parents":[{"commit":"ed4c4036af52a1c343e0f5142db85ab6adb37a34","subject":"Merge \"feat(images): Adds support for image ID reservation and queued images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/ed4c4036af52a1c343e0f5142db85ab6adb37a34"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-10-23 10:59:14.000000000","tz":60},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-03-19 09:37:26.000000000","tz":0},"subject":"Fix: Include target_project_id in RBAC creation","message":"Fix: Include target_project_id in RBAC creation\n\nThis fixes an issue where we cannot create multiple Neutron RBAC\npolicies using the same source project and network.\n\nCloses-Bug: #2129677\nChange-Id: Ib3c63131f9ae09abd03fc7d619e3566267690216\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/6a54ca043f9e4350a48a8c78a5a5df6d46bc5073"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/6a54ca043f9e4350a48a8c78a5a5df6d46bc5073"}]},"branch":"refs/heads/master","description":"Rebase"},"9d44f3d8119a88e9f067d234140ca0a14337f81a":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-05-06 11:03:24.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/84/964684/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-collections-openstack","ref":"refs/changes/84/964684/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/84/964684/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/84/964684/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/84/964684/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/84/964684/4"}}},"commit":{"parents":[{"commit":"0e6e281316f19a321da230f9797cfd4982afc315","subject":"Merge \"CI: Run jobs for all non-EOL Ansible versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/0e6e281316f19a321da230f9797cfd4982afc315"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-10-23 10:59:14.000000000","tz":60},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-05-06 11:03:24.000000000","tz":0},"subject":"Fix: Include target_project_id in RBAC creation","message":"Fix: Include target_project_id in RBAC creation\n\nThis fixes an issue where we cannot create multiple Neutron RBAC\npolicies using the same source project and network.\n\nCloses-Bug: #2129677\nChange-Id: Ib3c63131f9ae09abd03fc7d619e3566267690216\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/9d44f3d8119a88e9f067d234140ca0a14337f81a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/9d44f3d8119a88e9f067d234140ca0a14337f81a"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"6a54ca043f9e4350a48a8c78a5a5df6d46bc5073","theirs":"0e6e281316f19a321da230f9797cfd4982afc315","contains_conflicts":false}},"5ce28ec461628258991db2e98c5b357a2a2a5855":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-05-27 08:21:16.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"real_uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/84/964684/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-collections-openstack","ref":"refs/changes/84/964684/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/84/964684/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/84/964684/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/84/964684/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/84/964684/5"}}},"commit":{"parents":[{"commit":"c81422e3f6be86394284ce2db3e28ed87a530428","subject":"Merge \"[doc] Document usage of changelog\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/c81422e3f6be86394284ce2db3e28ed87a530428"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-10-23 10:59:14.000000000","tz":60},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-05-27 08:21:16.000000000","tz":0},"subject":"Fix: Include target_project_id in RBAC creation","message":"Fix: Include target_project_id in RBAC creation\n\nThis fixes an issue where we cannot create multiple Neutron RBAC\npolicies using the same source project and network.\n\nCloses-Bug: #2129677\nChange-Id: Ib3c63131f9ae09abd03fc7d619e3566267690216\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/5ce28ec461628258991db2e98c5b357a2a2a5855"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/5ce28ec461628258991db2e98c5b357a2a2a5855"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"9d44f3d8119a88e9f067d234140ca0a14337f81a","theirs":"c81422e3f6be86394284ce2db3e28ed87a530428","contains_conflicts":false}},"55cf97aa03d17d2fae730ab0da2df2e61e9925e7":{"kind":"REWORK","_number":6,"created":"2026-09-23 08:33:20.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/84/964684/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-collections-openstack","ref":"refs/changes/84/964684/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/84/964684/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/84/964684/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/84/964684/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/84/964684/6"}}},"commit":{"parents":[{"commit":"da4a211b2d1dd147dead0e3db63a373b43b3c09e","subject":"CI: Test with ansible-core 2.21","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/da4a211b2d1dd147dead0e3db63a373b43b3c09e"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-10-23 10:59:14.000000000","tz":60},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-09-23 08:33:11.000000000","tz":60},"subject":"Fix: Include target_project_id in RBAC creation","message":"Fix: Include target_project_id in RBAC creation\n\nThis fixes an issue where we cannot create multiple Neutron RBAC\npolicies using the same source project and network.\n\nCloses-Bug: #2129677\nChange-Id: Ib3c63131f9ae09abd03fc7d619e3566267690216\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/55cf97aa03d17d2fae730ab0da2df2e61e9925e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/55cf97aa03d17d2fae730ab0da2df2e61e9925e7"}]},"branch":"refs/heads/master"},"291d64adaa44617b19a66cbbb250c4079f12fe26":{"kind":"REWORK","_number":7,"created":"2026-09-23 08:35:35.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/84/964684/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-collections-openstack","ref":"refs/changes/84/964684/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/84/964684/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/84/964684/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/84/964684/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/84/964684/7"}}},"commit":{"parents":[{"commit":"da4a211b2d1dd147dead0e3db63a373b43b3c09e","subject":"CI: Test with ansible-core 2.21","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/da4a211b2d1dd147dead0e3db63a373b43b3c09e"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-10-23 10:59:14.000000000","tz":60},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-09-23 08:35:26.000000000","tz":60},"subject":"Fix: Include target_project_id in RBAC creation","message":"Fix: Include target_project_id in RBAC creation\n\nThis fixes an issue where we cannot create multiple Neutron RBAC\npolicies using the same source project and network.\n\nCloses-Bug: #2129677\nChange-Id: Ib3c63131f9ae09abd03fc7d619e3566267690216\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/291d64adaa44617b19a66cbbb250c4079f12fe26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/291d64adaa44617b19a66cbbb250c4079f12fe26"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
