)]}'
{"id":"openstack%2Fansible-collections-openstack~999179","triplet_id":"openstack%2Fansible-collections-openstack~master~I3276aa8613165a98bd5fafa1c7830039f95ad4bb","project":"openstack/ansible-collections-openstack","branch":"master","topic":"bug/2162084","attention_set":{},"removed_from_attention_set":{"39363":{"account":{"_account_id":39363,"name":"Daniel Wuensche","display_name":"Daniel Wuensche (B1-Systems)","email":"wuensche@b1-systems.de","username":"b1-wuensche"},"last_update":"2026-09-15 10:51:52.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-09-14 15:58:16.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"I3276aa8613165a98bd5fafa1c7830039f95ad4bb","subject":"fix using undefined variable","status":"MERGED","created":"2026-07-29 19:11:02.000000000","updated":"2026-09-15 10:51:52.000000000","submitted":"2026-09-15 10:51:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"999179-bug/2162084","meta_rev_id":"a69bed5fb99cd92998cc0016fbe087d162dd7987","_number":999179,"virtual_id_number":999179,"owner":{"_account_id":39363,"name":"Daniel Wuensche","display_name":"Daniel Wuensche (B1-Systems)","email":"wuensche@b1-systems.de","username":"b1-wuensche"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-15 10:51:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"all":[{"value":2,"date":"2026-09-14 13:46:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":2,"date":"2026-09-14 15:58:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":1,"date":"2026-09-14 15:58:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-29 20:35:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-08 12:12:34.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2026-09-14 13:46:05.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"state":"REVIEWER"}],"messages":[{"id":"a8dd7dc7265f04be196c4c52089f4cc131636846","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39363,"name":"Daniel Wuensche","display_name":"Daniel Wuensche (B1-Systems)","email":"wuensche@b1-systems.de","username":"b1-wuensche"},"date":"2026-07-29 19:11:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"daa004c4a66a9716aef31db25f21279551c05bde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-29 20:35:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c828e53647440888ce288b97a408a74\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/878980d42a9d4dea912f0f0c657844db : SUCCESS in 2m 21s\n- openstack-tox-linters-ansible-devel https://zuul.opendev.org/t/openstack/build/d8efb3edb84e4b79a4c0fae1f18eb642 : SUCCESS in 8m 14s (non-voting)\n- openstack-tox-linters-ansible-2.18 https://zuul.opendev.org/t/openstack/build/d2b351c8a6934eb890eda2257d98fa19 : SUCCESS in 7m 59s\n- openstack-tox-linters-ansible-2.19 https://zuul.opendev.org/t/openstack/build/1a1493603ef54692ad80ce623bdaf79e : SUCCESS in 5m 21s\n- openstack-tox-linters-ansible-2.20 https://zuul.opendev.org/t/openstack/build/e965a1f11d864e199e2a8faf6df5dafb : SUCCESS in 5m 59s\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/b2e088d1cba9439dac149e84676f4c2f : SUCCESS in 1h 17m 21s\n- ansible-collections-openstack-functional-devstack-releases https://zuul.opendev.org/t/openstack/build/722d085c1d1446d49e702a0ceb2d04d3 : SUCCESS in 35m 09s\n- ansible-collections-openstack-functional-devstack-ansible-2.18 https://zuul.opendev.org/t/openstack/build/252589a767ba4e9089ba1aa4f9744ba1 : SUCCESS in 54m 58s\n- ansible-collections-openstack-functional-devstack-ansible-2.19 https://zuul.opendev.org/t/openstack/build/4991f9b9fd1b4f2f91469886c57c73ca : SUCCESS in 1h 18m 33s\n- ansible-collections-openstack-functional-devstack-ansible-2.20 https://zuul.opendev.org/t/openstack/build/3b83a6a1c909490c80267f643e204a11 : SUCCESS in 1h 13m 59s\n- ansible-collections-openstack-functional-devstack-ansible-devel https://zuul.opendev.org/t/openstack/build/db0577f47ab84f4785fc2eb3703442ae : POST_FAILURE in 7m 41s (non-voting)\n- ansible-collections-openstack-functional-devstack-octavia https://zuul.opendev.org/t/openstack/build/a536171e9c9448aab58e87433b7b1aa6 : SUCCESS in 37m 52s\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/6e54f95f4a6241e68e4ac8ca000a3acb : SUCCESS in 55m 18s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/c271ed3d045948878859e02d54f318c7 : SUCCESS in 55m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3c2880068f884f43d18f264a70c68ca6f3559ed8","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-08 12:12:34.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1c72ec4cb47ce2ec7fb6c714673d1e8002af48b8","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-08 12:12:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e064b15d4b1c8051b54d6b3d22a9d3f934068502","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39363,"name":"Daniel Wuensche","display_name":"Daniel Wuensche (B1-Systems)","email":"wuensche@b1-systems.de","username":"b1-wuensche"},"date":"2026-09-14 09:56:04.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"1a062a436f75cd353b50739f35f5c1f89a7422cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39363,"name":"Daniel Wuensche","display_name":"Daniel Wuensche (B1-Systems)","email":"wuensche@b1-systems.de","username":"b1-wuensche"},"date":"2026-09-14 09:58:10.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"20b5a0caf03519b6a7811300bacabb016f710ff7","author":{"_account_id":39363,"name":"Daniel Wuensche","display_name":"Daniel Wuensche (B1-Systems)","email":"wuensche@b1-systems.de","username":"b1-wuensche"},"date":"2026-09-14 10:01:47.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"586b84fd326bc99f593654d0511c71bfce236bbf","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-09-14 13:46:05.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"81324e3032d37771e588dd4b02c7c4532dcdd6de","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-14 15:58:16.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"4b9359bf33baf5ed55afe58635bf568f0127b805","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-15 08:18:54.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"996a3fa16a36dedd0c321e77ff57897b7afb4873","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 09:45:09.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d09ee460821f4e00acf3a252b5cbe9ce\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/d057f72f43824378922bed5969b565c7 : SUCCESS in 2m 01s\n- openstack-tox-linters-ansible-devel https://zuul.opendev.org/t/openstack/build/4513d236e2c449069dbe99cdb5904b39 : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-linters-ansible-2.18 https://zuul.opendev.org/t/openstack/build/874618eccc834af38c9bb38e156c8d0a : SUCCESS in 6m 11s\n- openstack-tox-linters-ansible-2.19 https://zuul.opendev.org/t/openstack/build/3a34ab4d332d4a04bf5637363ad01aaa : SUCCESS in 8m 11s\n- openstack-tox-linters-ansible-2.20 https://zuul.opendev.org/t/openstack/build/bcf13773b35247e6adb2eac5fb1d7520 : SUCCESS in 6m 36s\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/3f4d7c0de15e4b029f212817723de154 : SUCCESS in 42m 18s\n- ansible-collections-openstack-functional-devstack-releases https://zuul.opendev.org/t/openstack/build/2ae5ec646c274091bee0b249ee5a5860 : SUCCESS in 36m 04s\n- ansible-collections-openstack-functional-devstack-ansible-2.18 https://zuul.opendev.org/t/openstack/build/b8111f522553424ba143f4da0b6650ab : SUCCESS in 1h 20m 12s\n- ansible-collections-openstack-functional-devstack-ansible-2.19 https://zuul.opendev.org/t/openstack/build/2a241cc377a84cdfb1034e1ba688c8eb : SUCCESS in 1h 05m 07s\n- ansible-collections-openstack-functional-devstack-ansible-2.20 https://zuul.opendev.org/t/openstack/build/6bbfbd97a4e14bcabddaf046c545e360 : SUCCESS in 1h 13m 02s\n- ansible-collections-openstack-functional-devstack-ansible-devel https://zuul.opendev.org/t/openstack/build/2cef2dd4128c4acfac7902c42f802f63 : POST_FAILURE in 5m 51s (non-voting)\n- ansible-collections-openstack-functional-devstack-octavia https://zuul.opendev.org/t/openstack/build/d48a1e3265a94887bb9bfb7a2b0b3193 : SUCCESS in 27m 35s\n- bifrost-collections-src https://zuul.opendev.org/t/openstack/build/3eb1ee46be004f6e8166a2bc93089e1b : FAILURE in 9m 39s (non-voting)\n- bifrost-keystone-collections-src https://zuul.opendev.org/t/openstack/build/22181dc584b34b3998d82e11dcba254e : FAILURE in 8m 23s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ded7148b18f978bc1e4867c1a45caafa4426b1e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 09:45:54.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"150f1e6b9b533c79009e7038913a9fc546cbf2be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 10:51:52.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/144ef95a80de49bda8e30cffc8aeb9ab\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/3681ae297b984b02898fe72d14542ac1 : SUCCESS in 3m 07s\n- openstack-tox-linters-ansible-2.18 https://zuul.opendev.org/t/openstack/build/1d0d225677cd4850a684450d0a04d86b : SUCCESS in 6m 01s\n- openstack-tox-linters-ansible-2.19 https://zuul.opendev.org/t/openstack/build/9e137a2927914f56870df1202f73e1a8 : SUCCESS in 9m 02s\n- openstack-tox-linters-ansible-2.20 https://zuul.opendev.org/t/openstack/build/3521cedf296c426d929dc78ba0d5c209 : SUCCESS in 5m 09s\n- ansible-collections-openstack-functional-devstack-releases https://zuul.opendev.org/t/openstack/build/ac76a20b9e354f248b9f65838488334e : SUCCESS in 1h 04m 34s\n- ansible-collections-openstack-functional-devstack-octavia https://zuul.opendev.org/t/openstack/build/604086fc007847828ab39004370e0d2b : SUCCESS in 50m 17s","accounts_in_message":[],"_revision_number":3},{"id":"a69bed5fb99cd92998cc0016fbe087d162dd7987","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 10:51:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9a437e49bcc0ef53cf910926249005bf478adcfa","revisions":{"b9b5535b3d8cf48dca4cd55c2bcceb3b373fcc68":{"kind":"REWORK","_number":1,"created":"2026-07-29 19:11:02.000000000","uploader":{"_account_id":39363,"name":"Daniel Wuensche","display_name":"Daniel Wuensche (B1-Systems)","email":"wuensche@b1-systems.de","username":"b1-wuensche"},"ref":"refs/changes/79/999179/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-collections-openstack","ref":"refs/changes/79/999179/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/79/999179/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/79/999179/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/79/999179/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/79/999179/1"}}},"commit":{"parents":[{"commit":"ee88cc15b87221ecabb474c7c3d0feaf6e01a7fa","subject":"Remove MAC address requirement for node registration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/ee88cc15b87221ecabb474c7c3d0feaf6e01a7fa"}]}],"author":{"name":"wuensche","email":"wuensche@b1-systems.de","date":"2026-07-29 19:02:17.000000000","tz":120},"committer":{"name":"wuensche","email":"wuensche@b1-systems.de","date":"2026-07-29 19:10:18.000000000","tz":120},"subject":"fix using undefined variable","message":"fix using undefined variable\n\nCloses-Bug: #2162084\nChange-Id: I3276aa8613165a98bd5fafa1c7830039f95ad4bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/b9b5535b3d8cf48dca4cd55c2bcceb3b373fcc68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/b9b5535b3d8cf48dca4cd55c2bcceb3b373fcc68"}]},"branch":"refs/heads/master"},"b1350281a9d80933f5492d61147b1dd063396af4":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-09-14 09:56:04.000000000","uploader":{"_account_id":39363,"name":"Daniel Wuensche","display_name":"Daniel Wuensche (B1-Systems)","email":"wuensche@b1-systems.de","username":"b1-wuensche"},"ref":"refs/changes/79/999179/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-collections-openstack","ref":"refs/changes/79/999179/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/79/999179/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/79/999179/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/79/999179/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/79/999179/2"}}},"commit":{"parents":[{"commit":"ee88cc15b87221ecabb474c7c3d0feaf6e01a7fa","subject":"Remove MAC address requirement for node registration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/ee88cc15b87221ecabb474c7c3d0feaf6e01a7fa"}]}],"author":{"name":"wuensche","email":"wuensche@b1-systems.de","date":"2026-07-29 19:02:17.000000000","tz":120},"committer":{"name":"Daniel Wuensche","email":"wuensche@b1-systems.de","date":"2026-09-14 09:55:27.000000000","tz":120},"subject":"fix using undefined variable","message":"fix using undefined variable\n\nCloses-Bug: #2162084\nChange-Id: I3276aa8613165a98bd5fafa1c7830039f95ad4bb\nSigned-off-by: Daniel Wuensche \u003cwuensche@b1-systems.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/b1350281a9d80933f5492d61147b1dd063396af4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/b1350281a9d80933f5492d61147b1dd063396af4"}]},"branch":"refs/heads/master"},"9a437e49bcc0ef53cf910926249005bf478adcfa":{"kind":"REWORK","_number":3,"created":"2026-09-14 09:58:10.000000000","uploader":{"_account_id":39363,"name":"Daniel Wuensche","display_name":"Daniel Wuensche (B1-Systems)","email":"wuensche@b1-systems.de","username":"b1-wuensche"},"ref":"refs/changes/79/999179/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-collections-openstack","ref":"refs/changes/79/999179/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/79/999179/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/79/999179/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/79/999179/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-collections-openstack refs/changes/79/999179/3"}}},"commit":{"parents":[{"commit":"ee88cc15b87221ecabb474c7c3d0feaf6e01a7fa","subject":"Remove MAC address requirement for node registration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/ee88cc15b87221ecabb474c7c3d0feaf6e01a7fa"}]}],"author":{"name":"wuensche","email":"wuensche@b1-systems.de","date":"2026-07-29 19:02:17.000000000","tz":120},"committer":{"name":"Daniel Wuensche","email":"wuensche@b1-systems.de","date":"2026-09-14 09:57:57.000000000","tz":120},"subject":"fix using undefined variable","message":"fix using undefined variable\n\nCloses-Bug: #2162084\nChange-Id: I3276aa8613165a98bd5fafa1c7830039f95ad4bb\nSigned-off-by: Daniel Wuensche \u003cwuensche@b1-systems.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/9a437e49bcc0ef53cf910926249005bf478adcfa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-collections-openstack/commit/9a437e49bcc0ef53cf910926249005bf478adcfa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
