)]}'
{"id":"openstack%2Fansible-role-python_venv_build~1006745","triplet_id":"openstack%2Fansible-role-python_venv_build~master~I749ef1156819e524bfeb204840053f46842d9542","project":"openstack/ansible-role-python_venv_build","branch":"master","topic":"osa/molecule","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-09-25 06:19:23.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"I749ef1156819e524bfeb204840053f46842d9542","subject":"Address ansible-galaxy fetching issues.","status":"NEW","created":"2026-09-22 14:55:30.000000000","updated":"2026-09-25 09:01:10.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f56437470eb06181acf1066406a8859b06353d70","_number":1006745,"virtual_id_number":1006745,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"all":[{"tag":"autogenerated:zuul:gate","value":0,"date":"2026-09-25 09:01:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-09-24 01:28:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2026-09-24 15:20:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":1,"date":"2026-09-24 15:20:52.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-22 17:18:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-24 01:28:51.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-09-24 15:20:52.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"c167e904aa66179acd904b64fe55783569224ec4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-22 14:55:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"66537a4dd71feab7fa90bf12e4b0688542a8d4c3","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-22 15:25:58.000000000","message":"Topic set to osa/molecule","accounts_in_message":[],"_revision_number":1},{"id":"2d86fe03bf4b852ce507fd0fb687c3b2ebf8f671","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 17:18:09.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/openstack-ansible/+/1006741 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"7478cb20bf584a3ece70db7b1e7ef6ffe557323f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-22 17:25:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4874c2fb69d9fdf012d70e20b4a4e512d4349f41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-22 18:08:49.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"d56dd1d9290948d21fd797d3ee2ed01a26e105b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 18:21:44.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/openstack-ansible/+/1006741 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"2c0a60eb0d170f5515a849233b106eeda0acb7e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-22 18:23:55.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"a915f1325d90b5de1659136f1b5f13376e55b044","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-22 18:36:38.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"a11b01031d6b3219a0f9cc4943613f03d1988343","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-22 18:54:32.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"c2aaba8181aaedb2818784e1a41929a1ba64e3ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 19:58:02.000000000","message":"Patch Set 5: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/openstack-ansible/+/1006741 is needed.","accounts_in_message":[],"_revision_number":5},{"id":"715f2cdb1be877227d23683579c008c13ffc294f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-22 19:59:59.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"026e0e77942d4ef793bb80da09e9c3ba2728bae8","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-22 20:28:10.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"1fbcb84e809c7562fb474f5e132475b25ddec65c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 23:36:32.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c162973856ac4caeb9b314daccabd1fc\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c13b9e5b4616440092af9e5f1d334e82 : SUCCESS in 11m 54s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/3b0cf7d3eee048c2bf00e65f07770d81 : FAILURE in 28m 49s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/ba5a20d20b36458881766997761ca174 : SUCCESS in 1h 30m 55s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/fff02358df6a458bbb90d6b7158682bc : SUCCESS in 2h 33m 51s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/727a70cee0374eaa884330324b866fff : SUCCESS in 2h 27m 29s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/328806d2a77e492989fd9d8d657893c1 : SUCCESS in 2h 34m 58s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/4921b0c3b97241e8875c93c95ba18ef1 : SUCCESS in 22m 03s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d0ee2429245f4708b93b2c86e42fcd1f : SUCCESS in 20m 09s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/f8f8de400dc4407182f0e868260f7ea1 : SUCCESS in 16m 59s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/13e4bdb8f35748048cac3315e4eba83e : SUCCESS in 23m 15s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/7d688c133eb14354baab2687dc2e19a6 : FAILURE in 5m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64ea97460d884344bddd4f3c13d089c7 : SUCCESS in 8m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cd6459cb3b774ce787dff8d9e631f2ad : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":6},{"id":"ff8544f4460d34eaeabe8e776d620d01704441a6","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-23 08:06:45.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"2284f774db1fb25f6cb78978cf9adc9adc1fd2e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 11:28:09.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1ab0dac939804409937278461e1c2ff3\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/42c2035c3a384761b0ab13efd283e281 : SUCCESS in 12m 25s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/096e0373cbf54b4a9e8530db338737fb : SUCCESS in 2h 04m 07s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/f45f01e72f014d0996d97f04bad8a02e : SUCCESS in 1h 39m 59s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/aadccc6154204901b23e00068d451a2b : SUCCESS in 1h 40m 40s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e8be98bca9df4afaa7316bcaa60532cd : FAILURE in 2h 24m 21s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/5af186c108e24201b1419f21ae9a6edb : SUCCESS in 2h 27m 05s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/f8cbce92b83d45a19f40a8be7debc710 : SUCCESS in 17m 18s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3ad796ba1b714d1988872160c179db55 : SUCCESS in 10m 32s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/70f60e7255354821b4371b2b613a8cb9 : SUCCESS in 15m 43s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/d271cbb9385e4f23affa678a3c813bb6 : SUCCESS in 22m 36s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b0dae2e226bd4e3296f280220cb14e94 : SUCCESS in 26m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/058ec5639bf64b2f8a70693396a3b9a4 : SUCCESS in 4m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/47a667ee03db403abf7bf99ca532422b : SUCCESS in 4m 50s","accounts_in_message":[],"_revision_number":6},{"id":"d4feb812b8fb6d9429f8c3090912599e14bb77ef","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-23 12:16:56.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"b02ba1313bd727aae680b1102e21d254a73291fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 14:59:24.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c96ac0d85434128a718bb5fd11a182f\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3081c72fd02240c18c4a90e88f43fb05 : SUCCESS in 13m 45s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/06d504d3b0c04822bde1f10b5b26c084 : SUCCESS in 2h 02m 30s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/72ac2c7ce0b14a7c987fe665b55ecf50 : SUCCESS in 2h 26m 50s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9353606a95204adb919e6896b3fe8fef : SUCCESS in 1h 21m 33s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/393c8ad59ed04653b3eb4075aeaaa568 : SUCCESS in 2h 18m 27s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/bc7c14d015c14ef397835df41e640f3f : SUCCESS in 2h 18m 58s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/08280e728d404b299185660c7002a7f5 : SUCCESS in 25m 34s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/666573b5c9f24f69842c0175f22426c9 : SUCCESS in 19m 29s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/13d7f08b37834d52ad1d26da2fd349a3 : SUCCESS in 20m 51s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/d8b2ef2cdd57424bb623a87c7ea60830 : SUCCESS in 31m 48s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4f41f7ff098146d7a6dda9629f6c44d6 : SUCCESS in 20m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65bca829a68146b98d63bfdff3969404 : SUCCESS in 7m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ab0ca01e6fcb40a6a7aba25bbedd1dbf : SUCCESS in 3m 28s","accounts_in_message":[],"_revision_number":6},{"id":"bf0b86a48e00d39d466e13ebdf9154c59ef5e817","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-24 01:28:51.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"be58bbba1a6feb8dfc93435008e2612554bc8f71","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2026-09-24 15:20:52.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"e83c9de2aa7ebd4307855df00ce4b938da50d2f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 15:29:00.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"912b9afa40223b306ccce43cda5d0be0c9991527","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 18:24:13.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d64a68a714c94f09a53e93678356f757\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1bfa65a385d641729e2e9772fe4d6126 : SUCCESS in 15m 55s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/33ec8082445045b7b0568cd0c40ba4e5 : SUCCESS in 1h 56m 47s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/2b7fe5915df942bdaedcaa34cabec75a : SUCCESS in 2h 22m 33s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0598b3f40059413f9e8b86dd896c18ad : SUCCESS in 2h 47m 09s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/439a1bec62354400b7904c9cb8761977 : SUCCESS in 2h 37m 40s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/62a32b0eb8ef47ca84e9d83c5d26c94b : SUCCESS in 2h 23m 03s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/616145354aeb4ccca47183a404892f17 : SUCCESS in 12m 59s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e426b5b718454445a1845858e7ee12d0 : SUCCESS in 23m 56s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/0a78e5b3a73643e699f2602641a39ee9 : TIMED_OUT in 31m 58s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/05e17341528f46e1b7f8b2ef9fdbb201 : FAILURE in 16m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1925ff962bac4a9b9b2a66ae6a9d8a9b : SUCCESS in 6m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b3bb02564a7e462ea8ea041d3419e128 : SUCCESS in 2m 52s","accounts_in_message":[],"_revision_number":6},{"id":"55b84c72b3d9e9467935fa659b420ae1a62362fd","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-24 20:30:03.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"9ee1e581cae84e49b1890dcb40bebd7d249658e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 22:43:39.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/49b9a4b794bf4546a98f83798b850596\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a007c16fa9b340a39112c99f54a5410b : SUCCESS in 10m 56s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/afe4fadeda0b4f569d3deb811490f020 : SUCCESS in 2h 01m 30s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/0c763555f10a4ac5ab869f4bd664f42e : SUCCESS in 2h 06m 31s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/bcc4d4d9d491484e9a3dc3979b06ae0b : SUCCESS in 2h 01m 26s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bf28e0fcc73747da967d52d3111cc386 : SUCCESS in 1h 53m 54s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/d9fa73ea1ca6430d84d344252fc7e95b : SUCCESS in 1h 49m 08s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/d9b00b968775413ea5f6f271fbc1fbc5 : SUCCESS in 18m 04s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/403269c946d34af78fd5c398c4d3241a : SUCCESS in 16m 58s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/16404a0cb9ff4f79bf886595c68dad29 : SUCCESS in 19m 01s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/2e92a24777144bb4b212a07e45df1af5 : FAILURE in 15m 06s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a7de814530fe4c49a07e4eb761dd5e2b : FAILURE in 4m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4e4a3ba65f140218282a3fdad693951 : SUCCESS in 3m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c666aa4c775149ff8082ea3a86b59697 : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":6},{"id":"5f01f207bdeb0368e613d7d34dacbd6175badb50","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-25 06:19:23.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"8c7d91c02313b51e8fafbd99c1c6bf39259e685d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 08:54:21.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e3496e34bd643a5a4b192f4357293fa\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/42af01cbfc874e539fed267deb637537 : SUCCESS in 15m 10s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/dca2a3451ffd40f8b21567b055d5fd88 : SUCCESS in 1h 50m 51s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/d9afb34d35b64a199e7daaa2152931b5 : SUCCESS in 2h 07m 13s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/60d7a513035a44d4aa8294c7e4c849bc : SUCCESS in 1h 18m 10s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a7653876a335419995558079970d28c7 : SUCCESS in 1h 13m 16s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/546c2dac906b48caa4d4dfae56a53abb : SUCCESS in 2h 23m 52s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/ea69f1e45948438d9eeca11c7fa83f13 : SUCCESS in 20m 45s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/df321abec4e14f4fa5e57d5521f8828b : SUCCESS in 17m 26s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/2175f87bd4894a308b5cedf42db44fc4 : SUCCESS in 19m 52s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/dd090b852ceb4b128f20a6f6cbe0d627 : SUCCESS in 21m 00s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/49979219ded8453299f440207088c597 : FAILURE in 5m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6fc3fc2b68a4b7bb096a955fb405f39 : SUCCESS in 6m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/347d1bb937324290ae74c6036a35a2ac : SUCCESS in 2m 59s","accounts_in_message":[],"_revision_number":6},{"id":"f56437470eb06181acf1066406a8859b06353d70","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 09:01:10.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"736d33ad85680ef27ebd00a60087febcde4e79cc","revisions":{"18b932a70dd99b1095feed8bb139fb404882f38f":{"kind":"REWORK","_number":1,"created":"2026-09-22 14:55:30.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/45/1006745/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-role-python_venv_build","ref":"refs/changes/45/1006745/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-role-python_venv_build refs/changes/45/1006745/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-role-python_venv_build refs/changes/45/1006745/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-role-python_venv_build refs/changes/45/1006745/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-role-python_venv_build refs/changes/45/1006745/1"}}},"commit":{"parents":[{"commit":"b331cf1da563791502aa1c17d2e81946340e5492","subject":"Replace outdated usage of egrep by grep","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-python_venv_build/commit/b331cf1da563791502aa1c17d2e81946340e5492"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-22 14:55:19.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-22 14:55:19.000000000","tz":120},"subject":"Address molecule intermittent issues","message":"Address molecule intermittent issues\n\nThis patch addresses two different issues that are present in our\nmolecule testing:\n- Rocky Linux can intermittently fail on facts gathering due to\nrestricted permissions to the /root, where ansible creates a tmp stroage\nof 0550 while systemd-userdbd takes a while to activate to fix it.\nThus, setting remote_tmp under /tmp fixes the issue.\n- Intermittent issues reaching Ansible\u0027s galaxy for molecule\ndependencies. This can be worked around by creating\\\ntests/requirements.yml, which is parsed by default with ansible-compat,\nand it does proceed with installation first.\nIf all dependencies are already present, molecule won\u0027t attempt\nreaching galaxy. We see a lot of failures recently on CentOS 10 Stream.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/1006741\nChange-Id: I749ef1156819e524bfeb204840053f46842d9542\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-python_venv_build/commit/18b932a70dd99b1095feed8bb139fb404882f38f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-python_venv_build/commit/18b932a70dd99b1095feed8bb139fb404882f38f"}]},"branch":"refs/heads/master"},"355be60678d506265caf7a7b97ebf6b7393106ce":{"kind":"REWORK","_number":2,"created":"2026-09-22 18:08:49.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/45/1006745/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-role-python_venv_build","ref":"refs/changes/45/1006745/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-role-python_venv_build refs/changes/45/1006745/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-role-python_venv_build refs/changes/45/1006745/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-role-python_venv_build refs/changes/45/1006745/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-role-python_venv_build refs/changes/45/1006745/2"}}},"commit":{"parents":[{"commit":"b331cf1da563791502aa1c17d2e81946340e5492","subject":"Replace outdated usage of egrep by grep","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-python_venv_build/commit/b331cf1da563791502aa1c17d2e81946340e5492"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-22 14:55:19.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-22 18:08:40.000000000","tz":120},"subject":"Address molecule intermittent issues","message":"Address molecule intermittent issues\n\nThis patch addresses two different issues that are present in our\nmolecule testing:\n- Rocky Linux can intermittently fail on facts gathering due to\nrestricted permissions to the /root, where ansible creates a tmp stroage\nof 0550 while systemd-userdbd takes a while to activate to fix it.\nThus, setting remote_tmp under /tmp fixes the issue.\n- Intermittent issues reaching Ansible\u0027s galaxy for molecule\ndependencies. This can be worked around by creating\\\ntests/requirements.yml, which is parsed by default with ansible-compat,\nand it does proceed with installation first.\nIf all dependencies are already present, molecule won\u0027t attempt\nreaching galaxy. We see a lot of failures recently on CentOS 10 Stream.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/1006741\nChange-Id: I749ef1156819e524bfeb204840053f46842d9542\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-python_venv_build/commit/355be60678d506265caf7a7b97ebf6b7393106ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-python_venv_build/commit/355be60678d506265caf7a7b97ebf6b7393106ce"}]},"branch":"refs/heads/master"},"a95d339570c10d9fa9742a5df18d863a61fc9a86":{"kind":"REWORK","_number":3,"created":"2026-09-22 18:23:55.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/45/1006745/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-role-python_venv_build","ref":"refs/changes/45/1006745/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-role-python_venv_build refs/changes/45/1006745/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-role-python_venv_build refs/changes/45/1006745/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-role-python_venv_build refs/changes/45/1006745/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-role-python_venv_build refs/changes/45/1006745/3"}}},"commit":{"parents":[{"commit":"b331cf1da563791502aa1c17d2e81946340e5492","subject":"Replace outdated usage of egrep by grep","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-python_venv_build/commit/b331cf1da563791502aa1c17d2e81946340e5492"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-22 14:55:19.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-22 18:23:55.000000000","tz":0},"subject":"Address molecule intermittent issues","message":"Address molecule intermittent issues\n\nThis patch addresses two different issues that are present in our\nmolecule testing:\n- Rocky Linux can intermittently fail on facts gathering due to\nrestricted permissions to the /root, where ansible creates a tmp stroage\nof 0550 while systemd-userdbd takes a while to activate to fix it.\nThus, setting remote_tmp under /tmp fixes the issue.\n- Intermittent issues reaching Ansible\u0027s galaxy for molecule\ndependencies. This can be worked around by creating\\\ntests/requirements.yml, which is parsed by default with ansible-compat,\nand it does proceed with installation first.\nIf all dependencies are already present, molecule won\u0027t attempt\nreaching galaxy. We see a lot of failures recently on CentOS 10 Stream.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/1006741\nChange-Id: I749ef1156819e524bfeb204840053f46842d9542\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-python_venv_build/commit/a95d339570c10d9fa9742a5df18d863a61fc9a86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-python_venv_build/commit/a95d339570c10d9fa9742a5df18d863a61fc9a86"}]},"branch":"refs/heads/master"},"4d519c20d03b92fa94517a5dcfee2eab11db8b8a":{"kind":"REWORK","_number":4,"created":"2026-09-22 18:36:38.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/45/1006745/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-role-python_venv_build","ref":"refs/changes/45/1006745/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-role-python_venv_build refs/changes/45/1006745/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-role-python_venv_build refs/changes/45/1006745/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-role-python_venv_build refs/changes/45/1006745/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-role-python_venv_build refs/changes/45/1006745/4"}}},"commit":{"parents":[{"commit":"b331cf1da563791502aa1c17d2e81946340e5492","subject":"Replace outdated usage of egrep by grep","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-python_venv_build/commit/b331cf1da563791502aa1c17d2e81946340e5492"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-22 14:55:19.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-22 18:36:38.000000000","tz":0},"subject":"Address molecule intermittent issues","message":"Address molecule intermittent issues\n\nThis patch addresses two different issues that are present in our\nmolecule testing:\n- Rocky Linux can intermittently fail on facts gathering due to\nrestricted permissions to the /root, where ansible creates a tmp stroage\nof 0550 while systemd-userdbd takes a while to activate to fix it.\nThus, setting remote_tmp under /tmp fixes the issue.\n- Intermittent issues reaching Ansible\u0027s galaxy for molecule\ndependencies. This can be worked around by creating\\\ntests/requirements.yml, which is parsed by default with ansible-compat,\nand it does proceed with installation first.\nIf all dependencies are already present, molecule won\u0027t attempt\nreaching galaxy. We see a lot of failures recently on CentOS 10 Stream.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/1006741\nChange-Id: I749ef1156819e524bfeb204840053f46842d9542\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-python_venv_build/commit/4d519c20d03b92fa94517a5dcfee2eab11db8b8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-python_venv_build/commit/4d519c20d03b92fa94517a5dcfee2eab11db8b8a"}]},"branch":"refs/heads/master"},"e2ab7f1fb8a85f590fac4a84056fabdd7c173dbb":{"kind":"REWORK","_number":5,"created":"2026-09-22 18:54:32.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/45/1006745/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-role-python_venv_build","ref":"refs/changes/45/1006745/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-role-python_venv_build refs/changes/45/1006745/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-role-python_venv_build refs/changes/45/1006745/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-role-python_venv_build refs/changes/45/1006745/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-role-python_venv_build refs/changes/45/1006745/5"}}},"commit":{"parents":[{"commit":"b331cf1da563791502aa1c17d2e81946340e5492","subject":"Replace outdated usage of egrep by grep","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-python_venv_build/commit/b331cf1da563791502aa1c17d2e81946340e5492"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-22 14:55:19.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-22 18:54:32.000000000","tz":0},"subject":"Address molecule intermittent issues","message":"Address molecule intermittent issues\n\nThis patch addresses two different issues that are present in our\nmolecule testing:\n- Rocky Linux can intermittently fail on facts gathering due to\nrestricted permissions to the /root, where ansible creates a tmp stroage\nof 0550 while systemd-userdbd takes a while to activate to fix it.\nThus, setting remote_tmp under /tmp fixes the issue.\n- Intermittent issues reaching Ansible\u0027s galaxy for molecule\ndependencies. This can be worked around by creating\\\ntests/requirements.yml, which is parsed by default with ansible-compat,\nand it does proceed with installation first.\nIf all dependencies are already present, molecule won\u0027t attempt\nreaching galaxy. We see a lot of failures recently on CentOS 10 Stream.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/1006741\nChange-Id: I749ef1156819e524bfeb204840053f46842d9542\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-python_venv_build/commit/e2ab7f1fb8a85f590fac4a84056fabdd7c173dbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-python_venv_build/commit/e2ab7f1fb8a85f590fac4a84056fabdd7c173dbb"}]},"branch":"refs/heads/master"},"736d33ad85680ef27ebd00a60087febcde4e79cc":{"kind":"REWORK","_number":6,"created":"2026-09-22 19:59:59.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/45/1006745/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-role-python_venv_build","ref":"refs/changes/45/1006745/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-role-python_venv_build refs/changes/45/1006745/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-role-python_venv_build refs/changes/45/1006745/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-role-python_venv_build refs/changes/45/1006745/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-role-python_venv_build refs/changes/45/1006745/6"}}},"commit":{"parents":[{"commit":"b331cf1da563791502aa1c17d2e81946340e5492","subject":"Replace outdated usage of egrep by grep","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-python_venv_build/commit/b331cf1da563791502aa1c17d2e81946340e5492"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-22 14:55:19.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-22 19:59:59.000000000","tz":0},"subject":"Address ansible-galaxy fetching issues.","message":"Address ansible-galaxy fetching issues.\n\nIntermittent issues reaching Ansible\u0027s galaxy for molecule\ndependencies. This can be worked around by creating\ntests/requirements.yml, which is parsed by default with \nansible-compat, and it does proceed with installation first.\nIf all dependencies are already present, molecule won\u0027t attempt\nreaching galaxy. We see a lot of failures recently on CentOS 10\nStream.\n\nChange-Id: I749ef1156819e524bfeb204840053f46842d9542\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-python_venv_build/commit/736d33ad85680ef27ebd00a60087febcde4e79cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-python_venv_build/commit/736d33ad85680ef27ebd00a60087febcde4e79cc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY"},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
