)]}'
{"id":"openstack%2Fansible-role-python_venv_build~814559","triplet_id":"openstack%2Fansible-role-python_venv_build~stable%2Fstein~I72fd009bf3084fd89545c228af4b1970ca3211c5","project":"openstack/ansible-role-python_venv_build","branch":"stable/stein","attention_set":{},"removed_from_attention_set":{"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2021-10-25 12:09:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I72fd009bf3084fd89545c228af4b1970ca3211c5","subject":"Workaround distro provided pip having old CA certs on centos-7","status":"MERGED","created":"2021-10-19 11:38:00.000000000","updated":"2021-10-25 12:16:43.000000000","submitted":"2021-10-25 12:09:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"814559","meta_rev_id":"5538d80ecebc60eb93e052502b1f81a256531efe","_number":814559,"virtual_id_number":814559,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-25 12:09:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2021-10-25 12:16:43.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2021-10-21 16:17:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2021-10-25 11:41:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":1,"date":"2021-10-21 16:17:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-19 11:40:17.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2021-10-19 12:02:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-21 16:17:02.000000000","updated_by":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"reviewer":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"state":"REVIEWER"}],"messages":[{"id":"11fcc6934b97822c5d55f6997512e5cb5b32ca14","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-10-19 11:38:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8b8be4aacdf8221b63a1f7b29c613900fe848a6d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-10-19 11:40:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"eee56d826212b5352208c7a8ba5c2d93d0eb7877","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-19 12:02:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d362f9a5408c49b4851735e7d13daa53 : SUCCESS in 7m 01s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/8c05b9ed74e04e50855a13b8634cc0b1 : FAILURE in 6m 38s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/611dd181bbe64e69830167a756623069 : SUCCESS in 15m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3942b158f5c54eca86f2978ed59ea67f : SUCCESS in 4m 19s\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/789f1bed9b3a4de2aedda53d683dd6d7 : FAILURE in 20m 33s","accounts_in_message":[],"_revision_number":1},{"id":"88928f7053318309811aa9470db5a5b0b7d84b2c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-10-20 11:27:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e2f5a1a3adf98c2c9a87c443433368fc719e8728","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-10-20 11:50:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8183b3a09b23b06fa0d43a69cf9dceaaa3e0f795","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 12:00:34.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/e397aad4c1074a88be8855b597c49745 : SUCCESS in 7m 02s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/21aef2f86faa4985b17706e6add75aa5 : FAILURE in 5m 50s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d777b090f7f84fdb856011927aa29969 : SUCCESS in 15m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/342ea584e2554810bfbc7a6198477376 : SUCCESS in 3m 45s\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/4684a80639e8428aac1b9f7f3cd39fbf : FAILURE in 28m 34s","accounts_in_message":[],"_revision_number":1},{"id":"570e13d39155dddd183ae63042e07d28fff03242","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-10-20 13:22:55.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"c2f8d6f54c596ab31e1f0268a9b027c3555ab2b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 13:51:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/80213b0e94154615b471ad696ca82405 : SUCCESS in 8m 48s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/86f2fc6069994419879f9e9b6dd577e1 : FAILURE in 9m 30s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3c95b03dd8734e21a15fadc9b09ba57d : SUCCESS in 15m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/117c9207ac7b40b7a34a72766c5371a1 : SUCCESS in 5m 08s\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/b1826e654e6448b287fa0eecb6c70b5a : FAILURE in 22m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4d1a1054a8f0895aba33eae78a613ec8faa20430","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2021-10-21 16:17:02.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e6187b1e5c1685359eccd02e0c197beac7df871f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-10-25 11:41:41.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2a3d7e8ca0e8d7227c494e86c020354377049a3b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 11:42:52.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"da10575886c9284f394cbc85d5a17b4ffdf39795","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 12:09:18.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/dd5bb8e099684768ab00827cbadce62a : SUCCESS in 10m 21s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/9537dc3217614a45b0926355590a4b4a : FAILURE in 8m 22s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bfe64d1282ae4694b1ad10e741870491 : SUCCESS in 17m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90b7bde0f56a429882b92744da304430 : SUCCESS in 5m 12s\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/2680567e8021461c8d740745a76d9255 : FAILURE in 23m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c360d06227f9a888c83af5a38f2618fc3205a260","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 12:09:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"d7ae1d327041154b7552b6bb56cb54b45547380c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 12:16:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99eb8fa1749445109760cb60d6ad9d1e : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0bc0da0d687d28516d8a0e2bf1a30eb2b615aeac","revisions":{"c7a0755d0c369ada0afbd8e24f5d9b6fc2615584":{"kind":"REWORK","_number":1,"created":"2021-10-19 11:38:00.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/59/814559/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-role-python_venv_build","ref":"refs/changes/59/814559/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-role-python_venv_build refs/changes/59/814559/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-role-python_venv_build refs/changes/59/814559/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-role-python_venv_build refs/changes/59/814559/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-role-python_venv_build refs/changes/59/814559/1"}}},"commit":{"parents":[{"commit":"08a00d666edf5b58fe04180672f06f0c1c8663ac","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-python_venv_build/commit/08a00d666edf5b58fe04180672f06f0c1c8663ac"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-10-19 11:37:03.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-10-19 11:37:03.000000000","tz":60},"subject":"Workaround distro provided pip having old CA certs on centos-7","message":"Workaround distro provided pip having old CA certs on centos-7\n\nUse the system trust store when upgrading versions of pip/setuptools/\nwheel on centos-7\n\nChange-Id: I72fd009bf3084fd89545c228af4b1970ca3211c5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-python_venv_build/commit/c7a0755d0c369ada0afbd8e24f5d9b6fc2615584"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-python_venv_build/commit/c7a0755d0c369ada0afbd8e24f5d9b6fc2615584"}]},"branch":"refs/heads/stable/stein"},"0bc0da0d687d28516d8a0e2bf1a30eb2b615aeac":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-10-20 13:22:55.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/59/814559/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-role-python_venv_build","ref":"refs/changes/59/814559/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-role-python_venv_build refs/changes/59/814559/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-role-python_venv_build refs/changes/59/814559/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-role-python_venv_build refs/changes/59/814559/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-role-python_venv_build refs/changes/59/814559/2"}}},"commit":{"parents":[{"commit":"738b1ee69b9f71eadb95a989bd08fc8a3a36e17f","subject":"Set centos-7 jobs to non voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-python_venv_build/commit/738b1ee69b9f71eadb95a989bd08fc8a3a36e17f"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-10-19 11:37:03.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-10-20 13:22:04.000000000","tz":60},"subject":"Workaround distro provided pip having old CA certs on centos-7","message":"Workaround distro provided pip having old CA certs on centos-7\n\nUse the system trust store when upgrading versions of pip/setuptools/\nwheel on centos-7\n\nChange-Id: I72fd009bf3084fd89545c228af4b1970ca3211c5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-python_venv_build/commit/0bc0da0d687d28516d8a0e2bf1a30eb2b615aeac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-python_venv_build/commit/0bc0da0d687d28516d8a0e2bf1a30eb2b615aeac"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
