)]}'
{"id":"openstack%2Fansible-role-python_venv_build~996974","triplet_id":"openstack%2Fansible-role-python_venv_build~master~I42b5e52585e4ac5a305a7e7459cada86cb2568f2","project":"openstack/ansible-role-python_venv_build","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-09-29 18:39:13.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-09-29 19:54:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I42b5e52585e4ac5a305a7e7459cada86cb2568f2","subject":"tox: remove debugging","status":"MERGED","created":"2026-07-13 06:24:47.000000000","updated":"2026-09-29 19:56:00.000000000","submitted":"2026-09-29 19:54:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"996974","meta_rev_id":"c8b344de0c75bfd3a838e5dfa646ef0cf3788ff2","_number":996974,"virtual_id_number":996974,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-29 19:54:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2026-09-29 18:39:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-09-24 15:48:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2026-09-29 18:39:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-13 06:25:48.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2026-07-13 06:55:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-27 13:14:01.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"f9c5f2a3cec8d435debf95166f8fd275355c2681","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-13 06:24:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"57606e39db1f501c8bd563d80eddd6fb5f640db5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-13 06:25:28.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"02be8c702d2d3c47664400a4b60d29ba55ab57ca","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-13 06:25:43.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"311fa0a9b1a26e406762a8ad3f98b694c2f550b0","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-13 06:25:48.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"462ae06aa5192de5a3019663f15bffe28e7774c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-13 06:55:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ac3e6ba9b02449d94d182f55f6227a1\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/72988742f68b4c5f9c69be0da2be68ce : SUCCESS in 11m 10s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/e00d6aaa989140c2b0eec92c28903687 : SUCCESS in 22m 04s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0d8b60e7eeb94e1abf75fb4f472e349c : SUCCESS in 10m 50s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/8885277098844f0b9316126c6921e419 : SUCCESS in 12m 16s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/f5aa2f8dd1ab4bdabed8ad3cfd925e01 : SUCCESS in 19m 42s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ab22001d4e8f4972bb054494ca63fc31 : FAILURE in 5m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c0b9518bb70449ebcc9b4759787f9c7 : SUCCESS in 6m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c4b7aeb810342a09227b6044a6eb9e7 : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":3},{"id":"431b3a61ff514a11959a35e8a27574a237cc7f07","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2026-08-27 13:14:01.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"a85e906b42343239e6ab6b5ffcc1caa6d1d07849","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-22 14:45:40.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"788bd1eb8ef156eab341c0e5d1ce0d5583c1ca0c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 14:49:19.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c081594bc222bbfa9d0464ad69825a152500b855","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 15:06:48.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4f564de5a6f64b53b60c8090bc0203c4\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c3f3162cbff34d26bca0c30ff2d06105 : SUCCESS in 12m 46s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/7cfa6f70b65d490a87edcc2b98d97e25 : SUCCESS in 14m 13s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/53bf3d0d63804858aff4ab54ede0252f : SUCCESS in 13m 06s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/32a47da353a4419298389a8fbf4b76ee : SUCCESS in 13m 36s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/1e2845f0d33d47d6a4ea100895928d12 : FAILURE in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64db21aba8244c4d80a5ce9994563c0f : SUCCESS in 5m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea6a684a706f4781824901a4725a5771 : SUCCESS in 2m 30s","accounts_in_message":[],"_revision_number":3},{"id":"5d89d7a1c9615f8222c3315985bc15e324b7196e","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-22 17:47:31.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"64c09475219e8bc0b771408ae36dafd8e7de119b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 17:47:36.000000000","message":"Patch Set 3: Verified-1\n\nUnable to freeze job graph: name \u0027LocalZKContext\u0027 is not defined","accounts_in_message":[],"_revision_number":3},{"id":"6db018970dea2d7c59bc691eb4950364484776d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"real_author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-24 15:48:55.000000000","message":"Patch Set 4: Patch Set 3 was rebased on behalf of \u003cGERRIT_ACCOUNT_37598\u003e\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"_revision_number":4},{"id":"1a7dc1ffae3cccc4c770a953b799adbfbd27f365","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 16:53:40.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c24ddcd60b83493ea46cb8e49f450a03\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c71ab7fee0354198b025d2ff99863217 : SUCCESS in 19m 28s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/099fde80af18410e902f855819717d2e : SUCCESS in 24m 41s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/53dfc73d365b407dabb3611934d4d2bf : SUCCESS in 19m 38s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/ac6d29c2e0ca4873b425b99ca2764f5e : SUCCESS in 24m 51s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/6f148895a31b4e99891cc9a1963faa79 : FAILURE in 17m 10s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1fdde6cb26c5449fa0c057dbe7744f5c : SUCCESS in 13m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b33d9dcfa636461499f2d3098987c220 : SUCCESS in 7m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7544ded5443341c380cb2bfb136c9131 : SUCCESS in 3m 11s","accounts_in_message":[],"_revision_number":4},{"id":"1d2cba9fc0d6b46fbc0e46aff6b31d5f41f8384e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-29 18:39:13.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1b3822a51ac5d524d67f23922422bee48d78a7f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 19:21:51.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a805f342e19d48158ae6dce4570b46d8\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e2cb4ec940ea4584bdd287270bd7e5e5 : SUCCESS in 14m 14s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/3be8e7a8cdcd4db784a6319850e0de95 : SUCCESS in 22m 21s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/96d53e20a95d46c9ae5a7ae101b5b7ce : SUCCESS in 12m 22s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/b50ea18a2c07496eb2d5e8dd8cfc415a : SUCCESS in 20m 16s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/5a25e988aeda40bfa6af3cf9439e159a : SUCCESS in 22m 20s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1175015551194f1fb6b1801ecc98ee6c : FAILURE in 5m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6474130a1e5e4a6aa3b64cda64db6a30 : SUCCESS in 6m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/46441c45f4304d7188a4ed1f9164cd17 : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":4},{"id":"9b56a366e1896b449db366649e33defd37edbbdb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 19:29:07.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"91220d944d1fa036211cdb33c28ffe2394ba260f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 19:54:21.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc5888211a8142d492cddc5267594aaa\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/92a80fb76a92450bad948db98a9d5cec : SUCCESS in 14m 50s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/bf1cf147ce424cc69807bf12b3e195ab : SUCCESS in 23m 42s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9bacdd0f9735414d8ad48f61534ad9e6 : SUCCESS in 19m 53s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/d483f9390a914b679b0b1c724cd09b3a : SUCCESS in 14m 08s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/2f46150d431a420180674dd4fab3a824 : SUCCESS in 22m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3647469d9e942abb215179c31a4c4a8 : SUCCESS in 6m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/869539f9a5ec481b90cafdd12fef87fe : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":4},{"id":"4998aec303131ddbf8c7bfc36f094e919f1939ad","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 19:54:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"c8b344de0c75bfd3a838e5dfa646ef0cf3788ff2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 19:56:00.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1e932a007db4cf2ba2caf1aebed506e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b19eddba57ce4c159838d6be77a1e11d : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"32fb388bcecc66f1fb7299add1da3bff9e473279","revisions":{"a3ee6ed323692f06cc61d3725445365e0dabab4d":{"kind":"REWORK","_number":1,"created":"2026-07-13 06:24:47.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/74/996974/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-role-python_venv_build","ref":"refs/changes/74/996974/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-role-python_venv_build refs/changes/74/996974/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-role-python_venv_build refs/changes/74/996974/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-role-python_venv_build refs/changes/74/996974/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-role-python_venv_build refs/changes/74/996974/1"}}},"commit":{"parents":[{"commit":"ff00f07fe0b002f97b5ebbfe703992608c842ecd","subject":"Check for resulting constraints file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-python_venv_build/commit/ff00f07fe0b002f97b5ebbfe703992608c842ecd"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-13 06:24:47.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-13 06:24:47.000000000","tz":0},"subject":"tox: remove debugging","message":"tox: remove debugging\n\nChange-Id: I42b5e52585e4ac5a305a7e7459cada86cb2568f2\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-python_venv_build/commit/a3ee6ed323692f06cc61d3725445365e0dabab4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-python_venv_build/commit/a3ee6ed323692f06cc61d3725445365e0dabab4d"}]},"branch":"refs/heads/master","conflicts":{"no_base_reason":"NO_MERGE_PERFORMED","contains_conflicts":false}},"382ea2fea13d6933998f2ed23bce992f70d73e8d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-07-13 06:25:28.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/74/996974/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-role-python_venv_build","ref":"refs/changes/74/996974/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-role-python_venv_build refs/changes/74/996974/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-role-python_venv_build refs/changes/74/996974/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-role-python_venv_build refs/changes/74/996974/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-role-python_venv_build refs/changes/74/996974/2"}}},"commit":{"parents":[{"commit":"6b3e103eeaeda0ed7750223fc6bc970681c0c86d","subject":"tox: Drop redundant injection of VIRTUAL_ENV variable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-python_venv_build/commit/6b3e103eeaeda0ed7750223fc6bc970681c0c86d"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-13 06:24:47.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-13 06:25:28.000000000","tz":0},"subject":"tox: remove debugging","message":"tox: remove debugging\n\nChange-Id: I42b5e52585e4ac5a305a7e7459cada86cb2568f2\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-python_venv_build/commit/382ea2fea13d6933998f2ed23bce992f70d73e8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-python_venv_build/commit/382ea2fea13d6933998f2ed23bce992f70d73e8d"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"ff00f07fe0b002f97b5ebbfe703992608c842ecd","ours":"a3ee6ed323692f06cc61d3725445365e0dabab4d","theirs":"6b3e103eeaeda0ed7750223fc6bc970681c0c86d","merge_strategy":"recursive","contains_conflicts":false}},"a54053add7df94cd84539b84bb64a3ffaf5a6079":{"kind":"REWORK","_number":3,"created":"2026-07-13 06:25:43.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/74/996974/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-role-python_venv_build","ref":"refs/changes/74/996974/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-role-python_venv_build refs/changes/74/996974/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-role-python_venv_build refs/changes/74/996974/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-role-python_venv_build refs/changes/74/996974/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-role-python_venv_build refs/changes/74/996974/3"}}},"commit":{"parents":[{"commit":"6b3e103eeaeda0ed7750223fc6bc970681c0c86d","subject":"tox: Drop redundant injection of VIRTUAL_ENV variable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-python_venv_build/commit/6b3e103eeaeda0ed7750223fc6bc970681c0c86d"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-13 06:24:47.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-13 06:25:42.000000000","tz":0},"subject":"tox: remove debugging","message":"tox: remove debugging\n\nChange-Id: I42b5e52585e4ac5a305a7e7459cada86cb2568f2\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-python_venv_build/commit/a54053add7df94cd84539b84bb64a3ffaf5a6079"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-python_venv_build/commit/a54053add7df94cd84539b84bb64a3ffaf5a6079"}]},"branch":"refs/heads/master"},"32fb388bcecc66f1fb7299add1da3bff9e473279":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-09-24 15:48:55.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"real_uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/74/996974/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-role-python_venv_build","ref":"refs/changes/74/996974/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-role-python_venv_build refs/changes/74/996974/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-role-python_venv_build refs/changes/74/996974/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-role-python_venv_build refs/changes/74/996974/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-role-python_venv_build refs/changes/74/996974/4"}}},"commit":{"parents":[{"commit":"736d33ad85680ef27ebd00a60087febcde4e79cc","subject":"Address ansible-galaxy fetching issues.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-python_venv_build/commit/736d33ad85680ef27ebd00a60087febcde4e79cc"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-13 06:24:47.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-24 15:48:55.000000000","tz":0},"subject":"tox: remove debugging","message":"tox: remove debugging\n\nChange-Id: I42b5e52585e4ac5a305a7e7459cada86cb2568f2\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-python_venv_build/commit/32fb388bcecc66f1fb7299add1da3bff9e473279"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-python_venv_build/commit/32fb388bcecc66f1fb7299add1da3bff9e473279"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"6b3e103eeaeda0ed7750223fc6bc970681c0c86d","ours":"a54053add7df94cd84539b84bb64a3ffaf5a6079","theirs":"736d33ad85680ef27ebd00a60087febcde4e79cc","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
