)]}'
{"id":"openstack%2Fansible-role-systemd_mount~654045","triplet_id":"openstack%2Fansible-role-systemd_mount~master~Ie49df5109e041f9d53eda9410f3c516d813f3284","project":"openstack/ansible-role-systemd_mount","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"Ie49df5109e041f9d53eda9410f3c516d813f3284","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-04-20 19:09:57.000000000","updated":"2019-04-23 11:58:50.000000000","submitted":"2019-04-23 11:58:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654045-1556020730135-354c625b","meta_rev_id":"be6fe7eb77b8c84a7ce14c9363c5892e0a589ea6","_number":654045,"virtual_id_number":654045,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":2,"date":"2019-04-23 11:58:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"date":"2019-04-23 09:31:42.000000000","_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"all":[{"value":2,"date":"2019-04-23 09:12:54.000000000","_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"all":[{"value":1,"date":"2019-04-23 09:12:54.000000000","_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-22 23:03:54.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-04-23 09:12:54.000000000","updated_by":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"reviewer":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"state":"REVIEWER"},{"updated":"2019-04-23 09:31:42.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-04-23 11:58:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a6473bb3c81cb1973decbd6c6bf53f212ec31f59","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-04-20 19:09:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5871db4d2ae6ea9b0b9e6a91c49c4716134fda9f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 19:23:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/45/654045/1/check/openstack-ansible-linters/81d640d/ : SUCCESS in 8m 27s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/45/654045/1/check/openstack-ansible-functional-debian-stable/2f959e3/ : FAILURE in 4m 41s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/45/654045/1/check/openstack-ansible-functional-centos-7/a18f52b/ : SUCCESS in 6m 33s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/45/654045/1/check/openstack-ansible-functional-ubuntu-bionic/f74b0e0/ : SUCCESS in 5m 45s\n- openstack-tox-docs http://logs.openstack.org/45/654045/1/check/openstack-tox-docs/0ad680d/html/ : SUCCESS in 2m 25s","accounts_in_message":[],"_revision_number":1},{"id":"84a53e0b6517ad8947603c95af685234d1388963","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-04-21 10:07:06.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"584dd5b558d2d498ec22d63791f5e8d553cef807","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-21 10:43:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/45/654045/1/check/openstack-ansible-linters/fc3eb1b/ : SUCCESS in 4m 26s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/45/654045/1/check/openstack-ansible-functional-debian-stable/248291e/ : FAILURE in 4m 52s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/45/654045/1/check/openstack-ansible-functional-centos-7/cebf400/ : SUCCESS in 7m 36s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/45/654045/1/check/openstack-ansible-functional-ubuntu-bionic/6c35b89/ : SUCCESS in 9m 25s\n- openstack-tox-docs http://logs.openstack.org/45/654045/1/check/openstack-tox-docs/b01f657/html/ : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":1},{"id":"8572229fbe3dde9541750165fa99fdc659f51fb9","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-04-22 23:03:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b1499c9ddd9d027e9aaddd34a528ecb3d3916920","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 23:20:45.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/45/654045/1/check/openstack-ansible-linters/d2c0b0a/ : SUCCESS in 6m 00s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/45/654045/1/check/openstack-ansible-functional-debian-stable/b883654/ : FAILURE in 5m 54s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/45/654045/1/check/openstack-ansible-functional-centos-7/b500f4c/ : SUCCESS in 7m 45s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/45/654045/1/check/openstack-ansible-functional-ubuntu-bionic/e4b178b/ : SUCCESS in 5m 09s\n- openstack-tox-docs http://logs.openstack.org/45/654045/1/check/openstack-tox-docs/1a04c03/html/ : SUCCESS in 2m 45s","accounts_in_message":[],"_revision_number":1},{"id":"3ed12d3e6dd30c54313c6a28002c94345425e938","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-04-23 08:08:48.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"5e1415b5cd44c0f0bdb59e1e2b53006942ce84e2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 08:51:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/45/654045/2/check/openstack-ansible-linters/cf86b2f/ : SUCCESS in 5m 51s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/45/654045/2/check/openstack-ansible-functional-debian-stable/292915f/ : SUCCESS in 5m 32s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/45/654045/2/check/openstack-ansible-functional-centos-7/e41db0c/ : SUCCESS in 7m 07s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/45/654045/2/check/openstack-ansible-functional-ubuntu-bionic/ac5a903/ : SUCCESS in 6m 46s\n- openstack-tox-docs http://logs.openstack.org/45/654045/2/check/openstack-tox-docs/9e819e9/html/ : SUCCESS in 2m 20s","accounts_in_message":[],"_revision_number":2},{"id":"dc30377a9fa4a4069fc7ab344f205c2fdfbfa946","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-04-23 09:12:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0da1b21ec17ec04211cdcd1cbfbd9274bea5c822","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 09:13:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d346ee547bea4e995a2e4bb63a31f6b02c4d6929","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 09:28:05.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"64cd5c1171f3723add92608010cc6ff1b2570479","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-04-23 09:31:42.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d9ee3fec31245c1e810d24ec5e502bb335f48857","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 09:59:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/45/654045/2/check/openstack-ansible-linters/367c8e2/ : SUCCESS in 5m 09s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/45/654045/2/check/openstack-ansible-functional-debian-stable/f77e3ce/ : SUCCESS in 8m 04s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/45/654045/2/check/openstack-ansible-functional-centos-7/704adef/ : SUCCESS in 7m 57s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/45/654045/2/check/openstack-ansible-functional-ubuntu-bionic/c8d7cd4/ : SUCCESS in 6m 24s\n- openstack-tox-docs http://logs.openstack.org/45/654045/2/check/openstack-tox-docs/75bd43c/html/ : SUCCESS in 3m 16s","accounts_in_message":[],"_revision_number":2},{"id":"8ad4470fd6399d7ed67244c5f0f09a0bd62f1ecc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 11:39:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"345694d56f3baf524a8f08d29a60df30225c95a7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 11:58:49.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/45/654045/2/gate/openstack-ansible-linters/93c0a5e/ : SUCCESS in 5m 49s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/45/654045/2/gate/openstack-ansible-functional-debian-stable/d83c190/ : SUCCESS in 6m 22s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/45/654045/2/gate/openstack-ansible-functional-centos-7/57f29f1/ : SUCCESS in 11m 34s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/45/654045/2/gate/openstack-ansible-functional-ubuntu-bionic/8c256eb/ : SUCCESS in 11m 27s\n- openstack-tox-docs http://logs.openstack.org/45/654045/2/gate/openstack-tox-docs/e3ee337/html/ : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":2},{"id":"8787ad6d3d11663d60872798cd0adb84128ba20b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 11:58:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"08fb0faf1600d4bd8e2e4d0b2acb45234039dc5a","revisions":{"1beabf867ca93d0db278602beafbb1ced4674dcc":{"kind":"REWORK","_number":1,"created":"2019-04-20 19:09:57.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/45/654045/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-role-systemd_mount","ref":"refs/changes/45/654045/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_mount refs/changes/45/654045/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_mount refs/changes/45/654045/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_mount refs/changes/45/654045/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-role-systemd_mount refs/changes/45/654045/1"}}},"commit":{"parents":[{"commit":"e79bebfa34f5fe11f34aa1bee79966e64e174ee2","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_mount/commit/e79bebfa34f5fe11f34aa1bee79966e64e174ee2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-20 19:09:56.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-20 19:09:56.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: Ie49df5109e041f9d53eda9410f3c516d813f3284\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_mount/commit/1beabf867ca93d0db278602beafbb1ced4674dcc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_mount/commit/1beabf867ca93d0db278602beafbb1ced4674dcc"}]},"branch":"refs/heads/master"},"08fb0faf1600d4bd8e2e4d0b2acb45234039dc5a":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-04-23 08:08:48.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/45/654045/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-role-systemd_mount","ref":"refs/changes/45/654045/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_mount refs/changes/45/654045/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_mount refs/changes/45/654045/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_mount refs/changes/45/654045/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-role-systemd_mount refs/changes/45/654045/2"}}},"commit":{"parents":[{"commit":"e79bebfa34f5fe11f34aa1bee79966e64e174ee2","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_mount/commit/e79bebfa34f5fe11f34aa1bee79966e64e174ee2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-20 19:09:56.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-04-23 08:08:46.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nDepends-On: Ie0c0b306c7f0cb05350b4e4cea950b2bb3857814\nChange-Id: Ie49df5109e041f9d53eda9410f3c516d813f3284\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_mount/commit/08fb0faf1600d4bd8e2e4d0b2acb45234039dc5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_mount/commit/08fb0faf1600d4bd8e2e4d0b2acb45234039dc5a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
