)]}'
{"id":"openstack%2Fansible-role-systemd_mount~689193","triplet_id":"openstack%2Fansible-role-systemd_mount~master~Ia9fba86a15acdc050e9014e4d2c44324d8eae692","project":"openstack/ansible-role-systemd_mount","branch":"master","topic":"build-pdf-docs","hashtags":[],"change_id":"Ia9fba86a15acdc050e9014e4d2c44324d8eae692","subject":"PDF Documentation Build tox target","status":"MERGED","created":"2019-10-17 15:01:18.000000000","updated":"2019-10-18 08:17:14.000000000","submitted":"2019-10-18 08:15:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"689193-1571386534332-ed58b23b","meta_rev_id":"cda06c491ebcba004108e0826c22d37967a3882c","_number":689193,"virtual_id_number":689193,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-18 08:15:34.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":2,"date":"2019-10-17 18:02:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-18 08:01:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-18 08:01:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-17 18:02:57.000000000","updated_by":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"reviewer":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"state":"REVIEWER"},{"updated":"2019-10-18 08:01:49.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-10-18 08:15:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0771e450eb6acdf1ddd8eb5a53b22960831f6595","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-10-17 15:01:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"30095bc386bca0cccbe4325e543495a0d8859d26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 15:19:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/14b23897d4a14cf0b57a98954f9eea5e : SUCCESS in 6m 26s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/c1ab0791e7bd44379f971dda49fc7524 : SUCCESS in 7m 43s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/ab09e7e6bc044148befb45ed6425e94e : SUCCESS in 12m 53s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bf03ef567b1d46c0a6f607501e2c924f : SUCCESS in 6m 41s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5f4d4b43a2814ecc985e9b857022c0ab : SUCCESS in 1m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f23065be40e418998a755068f9366f2 : SUCCESS in 4m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2a99e3c44b034259848550cfbc8a9b63 : SUCCESS in 2m 47s","accounts_in_message":[],"_revision_number":1},{"id":"39b7bed056632ecc5016f6727b90f3ef2d574628","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-10-17 15:36:56.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"b34cb8b5ceec8a6c9b5eaf71426fd755d7495ddf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 15:50:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/de8cd59c9f7c44b498d6909f60b76ff5 : SUCCESS in 5m 20s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/ca428dbaf189473daded42e755ef2641 : SUCCESS in 5m 55s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/1bf5d34d1cd04497aa5065fc2420aa17 : SUCCESS in 8m 42s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/08a048ceed83412f94ed169563cd3067 : SUCCESS in 9m 10s\n- requirements-check https://zuul.opendev.org/t/openstack/build/bf4c349bd9dc40f690f3d9f8714e203e : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/850c22c918874a358080ad21b34e72e9 : SUCCESS in 7m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a86000954a524344b58497064821c10b : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":2},{"id":"9fac15ea170243b8dcbb9e1ed6c0c17619add74d","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-10-17 18:02:57.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"736779bd19ca0e0ad520fdc0c50b80d2f90d288c","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-10-18 08:01:49.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"592d30d5bc005febceb951e1e2db29e86a71cb13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 08:02:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e660eb4b556b25fb6ef7d1e19e531577a7522842","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 08:15:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"2a28eb53fd02ddae8a14df6221de598c4df709a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 08:15:34.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/b05e455a4af446b2a21b09a27b63fa0a : SUCCESS in 6m 37s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/dcdb3cec756242d6bb7d2a8a52609441 : SUCCESS in 7m 03s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/34e9a79e1ad84c8a8e88fcc7fd9ebebe : SUCCESS in 7m 21s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d4c54f40ddb842bc8830e34a3a3e4f92 : SUCCESS in 5m 59s\n- requirements-check https://zuul.opendev.org/t/openstack/build/abddf1bc134e4eabb995b4a7fc0e7c98 : SUCCESS in 2m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4287ff945894bc382e772c8bc0f7567 : SUCCESS in 4m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a617d396725b4c8d9a0c02a410f28e23 : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":2},{"id":"abfb0a44854710a10aab6ae62d8dfd27630ccd72","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 08:17:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d12f0962ba7a4c79b604e838f9312821 : SUCCESS in 1m 09s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/693d20db0f324be2bcd9269d9d167e77 : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6acf8cb6589ce3e04511c3e8d2484d118c12e8f0","revisions":{"725993037a3564590a420003fc1fb04bf71eb727":{"kind":"REWORK","_number":1,"created":"2019-10-17 15:01:18.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/93/689193/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-role-systemd_mount","ref":"refs/changes/93/689193/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_mount refs/changes/93/689193/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_mount refs/changes/93/689193/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_mount refs/changes/93/689193/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-role-systemd_mount refs/changes/93/689193/1"}}},"commit":{"parents":[{"commit":"8d662a04a22bf679baf7f8d3e3e82b52d3211638","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_mount/commit/8d662a04a22bf679baf7f8d3e3e82b52d3211638"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-10-17 15:00:58.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-10-17 15:01:15.000000000","tz":180},"subject":"PDF Documentation Build tox target","message":"PDF Documentation Build tox target\n\nThis patch adds a `pdf-docs` tox target that will build\nPDF versions of our docs. As per the Train community goal:\n\n  https://governance.openstack.org/tc/goals/selected/train/pdf-doc-generation.html\n\nAdd sphinxcontrib-svg2pdfconverter to doc/requirements.txt\nto convert our SVGs.\n\nChange-Id: Ia9fba86a15acdc050e9014e4d2c44324d8eae692\nStory: 2006105\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_mount/commit/725993037a3564590a420003fc1fb04bf71eb727"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_mount/commit/725993037a3564590a420003fc1fb04bf71eb727"}]},"branch":"refs/heads/master"},"6acf8cb6589ce3e04511c3e8d2484d118c12e8f0":{"kind":"REWORK","_number":2,"created":"2019-10-17 15:36:56.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/93/689193/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-role-systemd_mount","ref":"refs/changes/93/689193/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_mount refs/changes/93/689193/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_mount refs/changes/93/689193/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_mount refs/changes/93/689193/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-role-systemd_mount refs/changes/93/689193/2"}}},"commit":{"parents":[{"commit":"8d662a04a22bf679baf7f8d3e3e82b52d3211638","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_mount/commit/8d662a04a22bf679baf7f8d3e3e82b52d3211638"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-10-17 15:00:58.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2019-10-17 15:36:53.000000000","tz":0},"subject":"PDF Documentation Build tox target","message":"PDF Documentation Build tox target\n\nThis patch adds a `pdf-docs` tox target that will build\nPDF versions of our docs. As per the Train community goal:\n\n  https://governance.openstack.org/tc/goals/selected/train/pdf-doc-generation.html\n\nAdd sphinxcontrib-svg2pdfconverter to doc/requirements.txt\nto convert our SVGs.\n\nChange-Id: Ia9fba86a15acdc050e9014e4d2c44324d8eae692\nStory: 2006105\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_mount/commit/6acf8cb6589ce3e04511c3e8d2484d118c12e8f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_mount/commit/6acf8cb6589ce3e04511c3e8d2484d118c12e8f0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
