)]}'
{"id":"openstack%2Fansible-role-systemd_mount~712398","triplet_id":"openstack%2Fansible-role-systemd_mount~master~I65989f4d4f44a86e370fd8e86e460cf61c4cb30c","project":"openstack/ansible-role-systemd_mount","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I65989f4d4f44a86e370fd8e86e460cf61c4cb30c","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2020-03-11 12:48:06.000000000","updated":"2020-03-11 17:57:50.000000000","submitted":"2020-03-11 17:54:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"712398-1583949264764-46c24852","meta_rev_id":"598d082d1635baf018a7561eaac8ad6a1decc512","_number":712398,"virtual_id_number":712398,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-11 17:54:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-11 16:23:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-03-11 16:23:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-11 16:23:56.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-03-11 17:54:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d05bd36b5c53e55f722c187338a4913200f79c29","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-03-11 12:48:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0838eae75c22329de83199b1c79529efa3c4858b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 16:16:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/b71061c7b87e4907beba010a4c0ec62c : SUCCESS in 7m 02s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/bbfaf6d17e09436f80bab40e11e3b56a : SUCCESS in 11m 15s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/d3277b8fedaf410894ac3a9a0a15755f : SUCCESS in 10m 11s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/83430fe312e94c31aee825ce9ae11b87 : SUCCESS in 8m 08s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/3ca48f4d22fb456288e85dc43a7fb655 : SUCCESS in 1h 23m 38s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/6ca6be1bd7204818aac6eb281af09471 : SUCCESS in 1h 53m 41s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/d4f67ae3823444de8c842cb65e100208 : SUCCESS in 1h 35m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1be6ae1e5736415497cfef45fa89a5b5 : SUCCESS in 1h 10m 01s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0ec1c1dd96d146b6ad2d99a975971e28 : SUCCESS in 2h 33m 06s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e365b74067e944c09db08865b287b61a : SUCCESS in 1h 55m 34s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9f23e4f29f544fe0b29a7c9b83dd525b : SUCCESS in 1h 19m 20s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/89802db99fb045ae990c069c3f360cec : FAILURE in 34m 38s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b5ee48cf4a5b4285ac44c38b7a6053ea : SUCCESS in 50m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b55900bf004d40c08acfa69124e361a1 : SUCCESS in 7m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6f58bdd804004e83a18276045a2d2c8d : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":1},{"id":"e12e19db519b63e94fbc4b89f8451c214042c47e","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-03-11 16:23:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c6b33ca53847acdffc3ef21de4682ade68c4ab88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 16:24:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"123699b8da1005c841730b3fc6df667c72e82060","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 17:54:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d8ab1ed827794591a5110e82a74089f6 : SUCCESS in 9m 25s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/cffeccb7eebc46dbaaa96cb5209b4ccf : SUCCESS in 8m 50s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/903a31917ae34e36913f7fe168e2456a : SUCCESS in 11m 04s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/32e4eddeab5c41fdaa273c2acf848a7c : SUCCESS in 7m 31s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/3264b19ff77b48c29665ce23f4418c90 : SUCCESS in 1h 05m 31s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d669e087d4c94cd2919ae018a6f5679a : SUCCESS in 1h 24m 46s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/80fc82f24d9e43da97e502179fcdd2d0 : SUCCESS in 1h 25m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e0ebbf44c820447785b25b9a2afe2985 : SUCCESS in 1h 13m 14s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0ddaf3a52e354e10846fac8c9373c07f : SUCCESS in 1h 25m 42s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a1bae6f57b494153a3cd9bef76954c6f : SUCCESS in 55m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de66e3fe37b2447ebd4a2c03f8e47089 : SUCCESS in 5m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fbd7a3d7cdf34f41b7983010b1df973c : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":1},{"id":"173da409f07dc96e14efafdcb45bbfdc729f0542","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 17:54:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1b47a44b6b5f0e57c0811f3bca0b039f6c3266f4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 17:57:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa99e8389d964664a54932019e76b27f : SUCCESS in 2m 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/45077c3e78244c138c242b0e1e95d94d : SUCCESS in 2m 16s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"69698ec24baafe1c0c8545b5bf02b9c5da353dac","revisions":{"69698ec24baafe1c0c8545b5bf02b9c5da353dac":{"kind":"REWORK","_number":1,"created":"2020-03-11 12:48:06.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/98/712398/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-role-systemd_mount","ref":"refs/changes/98/712398/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_mount refs/changes/98/712398/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_mount refs/changes/98/712398/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_mount refs/changes/98/712398/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-role-systemd_mount refs/changes/98/712398/1"}}},"commit":{"parents":[{"commit":"f102605cb97b5f0a423c61cdd2fb8edf0dd3dbfe","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_mount/commit/f102605cb97b5f0a423c61cdd2fb8edf0dd3dbfe"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-11 12:48:05.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-11 12:48:05.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I65989f4d4f44a86e370fd8e86e460cf61c4cb30c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_mount/commit/69698ec24baafe1c0c8545b5bf02b9c5da353dac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_mount/commit/69698ec24baafe1c0c8545b5bf02b9c5da353dac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
