)]}'
{"id":"openstack%2Fansible-role-systemd_mount~730253","triplet_id":"openstack%2Fansible-role-systemd_mount~master~I2c2e7cc5dde6107e143641baa884b53ba6b1bb4b","project":"openstack/ansible-role-systemd_mount","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I2c2e7cc5dde6107e143641baa884b53ba6b1bb4b","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2020-05-22 07:12:29.000000000","updated":"2020-05-22 15:36:22.000000000","submitted":"2020-05-22 15:34:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"730253-1590161647399-8638809e","meta_rev_id":"59e01a950e0d43cc31ff69231c99118128e8b6af","_number":730253,"virtual_id_number":730253,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-22 15:34:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-22 13:53:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-05-22 13:53:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-22 13:53:11.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-05-22 15:34:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4810e916d7b72a4be6eca475c8ddb6358ec0bcd6","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-05-22 07:12:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"07bab4669b8fa950ba3a3c99e2baf48abd96accd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 10:04:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/00b56951527c466d86baa51667b237a2 : SUCCESS in 9m 31s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/351863b5622e4e1ea55fa436e7c585a1 : SUCCESS in 10m 48s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/02fa73902bc342e692d6bc6060a2b249 : SUCCESS in 9m 58s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/16555750409d4d61bf9c0d9bff2da2af : FAILURE in 9m 30s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/67410ee021474516bc3b9cd7eb40dd57 : SUCCESS in 9m 12s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/c8c676911df741c7ac3776ab6d6eedff : SUCCESS in 1h 12m 29s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b9e98ca0f677423fb7a3e3390b180a38 : SUCCESS in 1h 35m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9c317cdfd0f6417294e151943eadc5e9 : SUCCESS in 1h 14m 09s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/865b8e20b0d14bea878f24b985dce71c : SUCCESS in 2h 24m 55s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bd5abe4d69024aef9cefa56d67878336 : SUCCESS in 1h 55m 53s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/6f2f99fa195545fabf80c1d9af241128 : SUCCESS in 1h 23m 21s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ede755ed48694bd28063851afae8a009 : SUCCESS in 1h 00m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/854827c6f4e6478dac13d6e18f5c9bc4 : SUCCESS in 6m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c31a46c32704e3a9fddd1f73c7a9169 : SUCCESS in 6m 47s","accounts_in_message":[],"_revision_number":1},{"id":"401f0fa24649b016d0ef965f0a50a948f4e93991","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-22 13:53:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ad8fd0e3d88103fd6db2736575676b2a5962ce15","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 13:53:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bbc934cbc0402e37db42bcc8515e82576f53dc31","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 15:34:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f4d6cc8a884a4f15937b44a1c2c55a70 : SUCCESS in 8m 23s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/fdc98cbf311943f5b3bd6aaebd7864a5 : SUCCESS in 8m 04s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/6727166daebc4a0db0d2381461f87c20 : SUCCESS in 10m 55s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/86bdcbeac695472a9fcc1c0df92449a8 : SUCCESS in 11m 08s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/c5b597ef167545f9a225839e4e2e031a : SUCCESS in 1h 12m 52s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/32c02f1a13c2404cac2a2f0300ee8cbf : SUCCESS in 1h 32m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/89e4c67d577849288a3f36728cf2ada5 : SUCCESS in 1h 22m 33s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/616cc012f143403f91236b1e5f892b08 : SUCCESS in 1h 21m 14s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8784567241b94d93b902abbde6cfd420 : SUCCESS in 1h 01m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ac3efe8f3cd40008f50f0fac9c98637 : SUCCESS in 9m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0a7e06cb85594bc3b876752b0123dfe2 : SUCCESS in 5m 20s","accounts_in_message":[],"_revision_number":1},{"id":"8410f0ffa5db46d9fcdfac372f90e29054b8669c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 15:34:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ee0f83bf1cde710fd9c9ae308a391165effc7ce5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 15:36:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0711ab27179b4e40b5e5d6c6128caa52 : SUCCESS in 1m 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9aadc2906c47440cad87ebcad26a5c13 : SUCCESS in 1m 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f3393cd3d6f4846f0bed4c6c56c808bad3f5e293","revisions":{"f3393cd3d6f4846f0bed4c6c56c808bad3f5e293":{"kind":"REWORK","_number":1,"created":"2020-05-22 07:12:29.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/53/730253/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-role-systemd_mount","ref":"refs/changes/53/730253/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_mount refs/changes/53/730253/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_mount refs/changes/53/730253/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_mount refs/changes/53/730253/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-role-systemd_mount refs/changes/53/730253/1"}}},"commit":{"parents":[{"commit":"dda77e1e2828b1f491b08997ba46876a4d8c7c2d","subject":"Merge \"[ussuri][goal] Update contributor documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_mount/commit/dda77e1e2828b1f491b08997ba46876a4d8c7c2d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-05-22 07:12:26.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-05-22 07:12:26.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I2c2e7cc5dde6107e143641baa884b53ba6b1bb4b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_mount/commit/f3393cd3d6f4846f0bed4c6c56c808bad3f5e293"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_mount/commit/f3393cd3d6f4846f0bed4c6c56c808bad3f5e293"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
